)]}'
{"id":"openstack%2Fkayobe~835191","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I70f8ddf984996d0821917ba5e5b8977c5a3b6614","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-25 20:41:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I70f8ddf984996d0821917ba5e5b8977c5a3b6614","subject":"Pin Jinja2\u003c3.1.0 to avoid contextfilter removal","status":"ABANDONED","created":"2022-03-25 09:19:34.000000000","updated":"2022-03-25 20:41:26.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"067f90cedc0ba9a0156956e360b8ac170a5f8932","_number":835191,"virtual_id_number":835191,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-25 11:29:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-25 09:19:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-03-25 09:38:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-03-25 11:29:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81ec1b2afaaa08adb4503d526838afb1cd3aac65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 09:19:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dc4538acc4c16079b7c9b275cd170c500c2e343","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-25 09:38:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74c9bfa27ab2f402b479d83acdd32d339779e35d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 10:10:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbe2e9279cd9a16cb421e81cb0b77162e8c2aeed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 11:29:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4a0b043e0074a1c8c1c5ad497846196 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cfbe7bdb0cb438f888a7d4b7af99356 : SUCCESS in 9m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ee1639ac4f84caeb372b13a8e3a31df : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8d3f397d1ca48fba83ccc1e739dbf6a : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17102ffc40554b52b73e95b1fecf6b0f : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6edd5f613a4ab1a8c2543a05f2b803 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06a2196dc5ce4a389942cf7cc4123bb3 : SUCCESS in 5m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e3e7ece94734f3c8b53d8433bf39bd7 : SUCCESS in 7m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/43340a38b4c942a18774e79a5a54ef51 : SUCCESS in 14m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a35cfe1673de4b59a70f72ef690b5a75 : SUCCESS in 32m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/00cde26b544f42c3b12fea5a978b1638 : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7d6146e33dd943368a05befb83987a96 : SUCCESS in 1h 06m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8b81f8f363b414b88adf6955362f85b : SUCCESS in 59m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5ca61cf529eb463983ddf7bbb16c83d9 : FAILURE in 1h 20m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ab43aa9417da4366a9193392b2815a70 : SUCCESS in 24m 03s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f1685519c3df40619f50ee540a7f7532 : SUCCESS in 23m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e25d80a2fd04135923870409d571a9b : SUCCESS in 17m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8dbd6e07aa0440e6a5409d4872e5e04c : SUCCESS in 2h 07m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6699e453994548479bed995d613a16eb : FAILURE in 12m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b4a98ccf5fb0461895a145ec076075b3 : SUCCESS in 46m 45s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2d1acd48d1a24150a0c0299c1d9e52a0 : SUCCESS in 32m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/862b3fc8d4154a8c9f4951632e9ba379 : SUCCESS in 32m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/99e769b52816464790ded9857bc8a755 : SUCCESS in 50m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76999e2744924c19aad85d5c7741395f : FAILURE in 9m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9b21ceb2f8ee41b1ad834e1f292bad95 : SUCCESS in 22m 19s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a437f790f43848d1b1523a3ab50c7f27 : SUCCESS in 16m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0c904d44d774ae08564212839962386 : SUCCESS in 13m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ba72b2228047460cb5f42f540158a14f : SUCCESS in 20m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/329dbe874ccd486d81f18ea142395b48 : SUCCESS in 17m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc03e95b8634437ca0c855ac7c3abf2c : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ecab9ccf894c59291f7c6f2ea96e9b235ecb64ad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 17:33:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf9e4dfbd1cbda5a67321b4709edceb7dcad9bdc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-25 20:38:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"067f90cedc0ba9a0156956e360b8ac170a5f8932","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-25 20:41:26.000000000","message":"Abandoned\n\nAbandoned in favour of https://review.opendev.org/c/openstack/kayobe/+/835279","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69090469e3559a2a2cf3513f7fe3589174d6cb6a","revisions":{"69090469e3559a2a2cf3513f7fe3589174d6cb6a":{"kind":"REWORK","_number":1,"created":"2022-03-25 09:19:34.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/91/835191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/835191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/835191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/835191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/835191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/835191/1"}}},"commit":{"parents":[{"commit":"36756307cf563ef55592614e0f18d7628a3db0bd","subject":"Merge \"Ubuntu: add support for Apt repository configuration\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36756307cf563ef55592614e0f18d7628a3db0bd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 15:52:36.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-25 09:18:30.000000000","tz":180},"subject":"Pin Jinja2\u003c3.1.0 to avoid contextfilter removal","message":"Pin Jinja2\u003c3.1.0 to avoid contextfilter removal\n\nJinja2 3.1.0 removes the contextfilter function which is used heavily in\nKayobe. The following errors are seen:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\n    template error while templating string: No filter named \u0027net_ip\u0027..\n    String: {{ admin_oc_net_name | net_ip }}\n\nThis change fixes the issue by pinning Jinja2 to \u003c3.1.0.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/835126\nChange-Id: I70f8ddf984996d0821917ba5e5b8977c5a3b6614\n(cherry picked from commit 9257806b27075c3db0886a474247884fd75c08a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69090469e3559a2a2cf3513f7fe3589174d6cb6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69090469e3559a2a2cf3513f7fe3589174d6cb6a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
