)]}'
{"id":"openstack%2Fkayobe~835610","triplet_id":"openstack%2Fkayobe~master~Idecda342f3ab86733e8d59061458d44af834dbb0","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-04-06 12:49:57.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-04-06 10:58:13.000000000","reason":"Michal Nasiadka replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-06 12:49:57.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-31 09:19:50.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Idecda342f3ab86733e8d59061458d44af834dbb0","subject":"CI: separate image builds into a non-voting job","status":"MERGED","created":"2022-03-29 10:51:25.000000000","updated":"2022-04-06 12:52:35.000000000","submitted":"2022-04-06 12:49:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835610","meta_rev_id":"78b077e5a356d767eef9d557709d1b0947c47019","_number":835610,"virtual_id_number":835610,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-06 12:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-06 10:58:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-31 15:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-06 10:58:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 11:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 07:17:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-03-31 09:19:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-04-06 07:16:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2022-04-06 07:16:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9e1b50b02fe9a4a889173ada7b77b764a2fbf78d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 10:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b51020c275bf68d5dd1a714369c0dc8e1c4d84f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 11:06:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0428cc17e446ab9fa9fae7fa8eaadc68a40788bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 11:36:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c253a26fe0fd43169c1cb2f60230d3ba : SUCCESS in 23m 01s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c234bde51657441cbf1f520994e384cf : SUCCESS in 25m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a65bfb15011443309faeb0d7ee9764f5 : FAILURE in 9m 38s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/bff116af93ce41de9faf85a7d6996660 : SUCCESS in 20m 59s (non-voting)\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/75c2a21831124956abc3542d352ccc65 : FAILURE in 22m 58s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/200a0b00cbaa487e800beacc19dd7c3f : FAILURE in 9m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cc954ff156e387fce34c7158df8c0fc781c62796","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 11:47:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"721328fb92bfd0969a14e429781a2224b5d2fe93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 12:22:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0b53fdfa4f79483b9a5afbbdca1e4ad2 : SUCCESS in 28m 42s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f548f36d3a3c4355ad74ec52727a80f6 : SUCCESS in 29m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5af2436d805e4aa6b94d44307d95973e : FAILURE in 9m 51s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/d1587ed91edf438aa3d8ba923221f6f6 : SUCCESS in 27m 00s (non-voting)\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/0274a1d012b2467f81166848c89424a5 : SUCCESS in 23m 56s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33cb11f46eb546b1bf9649926b0d3032 : FAILURE in 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8c719e5d3be6451c65f552a6bea76c30a2466a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 12:28:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"33cf410d48733aad1bb3b79d54b83bddb5c7d472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 13:07:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2d87c49df7a44cf7bc2e3f2ab82e960b : SUCCESS in 22m 14s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e5f3589378934647b5928e9f251bcddb : SUCCESS in 25m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a8e2ee16c724ffabaa451e5e572320e : FAILURE in 9m 57s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/d7feaf53c58942b1b0e22fd77aba01e4 : FAILURE in 31m 16s (non-voting)\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/406c14ab436b44a7998b937ff40824a2 : FAILURE in 30m 49s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc9ce0275b9f41bbaa1f0fccf3ecc31e : FAILURE in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f7d43851a064fd3823258d97d827d93538bf50f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 13:11:16.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"3ea6b2ea52e395962a9dca2a8ce470e42c98e449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 13:46:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8801db40d8ae4767acc881290ed36646 : SUCCESS in 20m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5e70226b3c794b01923bb2647f66ea49 : SUCCESS in 28m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/687a2304869c459287ce1dd718215b04 : FAILURE in 10m 39s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/7dc30987e3d34884801dc62c6f9b8c63 : FAILURE in 28m 09s (non-voting)\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/15ada4f927a84e49b252243d04cb207d : FAILURE in 29m 50s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fc182c8a7d745b4922dca7b6915634e : FAILURE in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e586610dab641de0213c29b39a22646b3918ee26","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 14:36:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0759fcd339c0c523ca643b335e70002ce56b439c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 15:47:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ec5b18e91186434d96804e5ebf8af9c5 : SUCCESS in 28m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/986410fd8b4549ca9b2c1609ec6c1be4 : SUCCESS in 20m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ac38f03fa08421cab3e4a0cac463b74 : FAILURE in 12m 48s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/36120965ae944633989504cd5aa009c6 : SUCCESS in 1h 03m 42s (non-voting)\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/8ce61f4688194c09a1730a58df8f3451 : SUCCESS in 40m 50s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59b50e1a39de4354801976980af0533b : FAILURE in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"162eedda41fa703dadf86fa956c868244cbb00f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 09:38:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"adb6cb9610926c3981b2eae36b8f012dfd9a2e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 10:37:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8509a22415e24518a5c71aa62bc5d472 : SUCCESS in 25m 31s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/93275dddadae4de3a5d27b0257907113 : SUCCESS in 33m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42425b56be244c9fbc8af82d602d2378 : SUCCESS in 18m 01s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/d56c42baa98e473b9c0245cc220abf96 : SUCCESS in 52m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ba55fc0195c4e8617d57dd7856671d19dc257a29","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:10:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9008733f604f726b6f9ad50e3a94af3ead4b8acb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:34:17.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"245edfda2b01a0d09ad0c3de60dc8b74329cb218","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:34:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"80fbb6d0c1c65327cbd6bb6db6210a03a4f6dc0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 12:25:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ce8c7a9f8aa041c093c1d76e0e45f24d : SUCCESS in 20m 47s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1ac1449980e84d7889e785d2b25cb626 : SUCCESS in 19m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49d143a77b8847888b8132baa7cff91d : SUCCESS in 21m 10s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/b773bab2da1b4976a718a65d41bbace5 : SUCCESS in 46m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d4c579af1d3c064379194b38cc490d2de109a836","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 12:29:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/bb5da002e8384edba69f416ccf81a90a : SUCCESS in 49m 52s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/517cbc17c748455f98fab70b6cf4cb38 : SUCCESS in 36m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7d970f9c4fca6c34f3b9063eba4f072158ea27c3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 13:18:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3c3a7b5bb87306b83b8976f9caa49229abdda06c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 13:18:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"f64c5692bd8b7d2c954977afe9628f803550d92e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 15:11:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c8bee4da35a476789673c482a889c52 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b8ddb7b31064a2ab3246b765500c216 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc957494293d437e94cfb24db70229b0 : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfd790ce29a2463891b5cdc879c9a15c : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9390a073d45e4926bb99f25c86b9fab5 : SUCCESS in 10m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7cbfb2d7cb0844a28ec6b103da2d66c9 : SUCCESS in 7m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3be983a568374d978fd7d3f3e416661f : SUCCESS in 13m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb948425f898447e959120c31e94e562 : SUCCESS in 32m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2bf0e2b05706439ba2a72de917701d02 : SUCCESS in 1h 03m 44s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7732f83609624b4e8f6d91acf34fe8e2 : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/233898e75c434404bf6d551245aae82c : SUCCESS in 53m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a8693e61ded447e5836587c1d6edd042 : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d2c833c3e710423aa29c01bdd0a61c7c : SUCCESS in 22m 48s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/020e25544c054c8db8e2fa431b2a2202 : SUCCESS in 20m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51d00c4910ed46ed9a70d5215eaf9d2b : SUCCESS in 14m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/788c5a3b214c4d8da3ff1d76395e1537 : SUCCESS in 1h 47m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/909183fdeeae45b1b9cace7956a0a75c : SUCCESS in 1h 12m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8cc9e2e4bb284245a6e80a920da1a0e7 : SUCCESS in 21m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/09b163cfd6ec475cbf23243368e85204 : SUCCESS in 20m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8efa86737d543a7bc1301aefdfcebe5 : SUCCESS in 17m 31s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/7de97052bc034ae6bda0a30e2b326b08 : SUCCESS in 46m 14s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f5497e073651403dbe1805d0c9ec49f3 : SUCCESS in 40m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d32db749b4f4eacbe63dfca4022a2ec : SUCCESS in 36m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/33f8068e3cfa4b07b76743e8a661650a : SUCCESS in 17m 52s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/48da3ff817cd4f669911dadea48bb141 : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a7c5bdf08544900ae4bd35f2a4fa21c : SUCCESS in 13m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/dbd9d366e3bc44ac997484f38dfaf0a2 : SUCCESS in 17m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d21a4cd21e4f4b8bb50a049bf6106a0d : SUCCESS in 16m 54s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc1e8c4ebd384130bb7b0300941fb7bc : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":9},{"id":"97972efd32b70ac6e491993ab5fe94d8b0cb88e0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 07:17:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"829a2a6a098641f64642ecbe11fc51b07e1986b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 08:52:17.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"fcd7130cce024e1e1652cd277f76739661e741f6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 08:52:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bce417a4ba9fc3ad39d1bdd415effa4160bfa908","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 09:19:50.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"41354219a78496d1ef9c7c99fb29fdc090a4eb0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 11:09:17.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4352b519cfb4875be573d0eb2834515 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0825d6f002f740fa87a5287a70e849ab : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75cbdaa546544feebabbddc2ff626975 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aaacd83b92941c4a172df10ec5a1e29 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddae6cad111441f0bb9fc0c7e3f3f48d : SUCCESS in 10m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a780bfbb0038406e94b22d8d90a68671 : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f12f02e4712c4d15963d4db4d35f8c68 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91825976c91e48189e8304d39833c8c8 : SUCCESS in 33m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/aff8a0da65564c058fead415964ed4d5 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/83b47a8d2fde434e88192267d277c82f : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf5cec8d8fab4985a7961acd61650fba : SUCCESS in 55m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/676c864e392d4e5c9d706b06c4d86659 : POST_FAILURE in 1h 15m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/737d06864dfd430db8c15b417531c62c : SUCCESS in 24m 28s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b618c01a7b9d4414abc96c3b98634434 : SUCCESS in 19m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/033af0daedd1407db09b2a1ac3aac9c7 : SUCCESS in 20m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2c19517a7b234a32a6e56b8416267929 : SUCCESS in 1h 56m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eb5d36be14442a5a10efb73691b5363 : SUCCESS in 1h 10m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c5f8c4e010614ddfaae5227248ed7a40 : SUCCESS in 20m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/35198232ec28412a8d5d37abb513ec83 : SUCCESS in 20m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b052fbe6a29a49c3918ef4c53f95ae8e : SUCCESS in 16m 34s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/d3f9dc03cbba49b39f5e1c57d12a08b9 : SUCCESS in 43m 25s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a6013effa17840dca51f8a4412fda705 : SUCCESS in 35m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40ae71851dc54b16acd1625aea02fdc1 : SUCCESS in 31m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/11e0db611fce49a5b8ade8a827961343 : SUCCESS in 17m 14s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4af9ca74555c43e6b5f4944c33c46da1 : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34f9c6b7fd3c4d6b9ab161e3fc4b68ec : SUCCESS in 12m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/dce4e62599d643f380dff3b788a54138 : SUCCESS in 18m 17s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/018503b6c4bf44019a5ff9d65ae97213 : SUCCESS in 17m 15s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31623fc3228340428682f4d5101bf968 : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":10},{"id":"b5bc475f7ca43e8388592014ef12cfd6c686344e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 13:01:55.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"dcc26ea915fbe01a90618b607187524c9bf2c61d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 15:09:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baf36403742f48b7b604df9156140c48 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06f7d8f27ed24650b7f687ca781a76e0 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/727f491fc895494b909708af8412c618 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f30dff0d99b64697beb9bc1cda969f6f : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4853685078ab478eb4e5c4f9fe6df103 : SUCCESS in 9m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14222d5a6b32479993bcdf71166b7aee : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73c5ed9612184240b76d7dea51a5585f : SUCCESS in 13m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d127a828b96f4a3287572aff80c951dc : SUCCESS in 30m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8a80e47c42b946c187c536b86d8aa873 : SUCCESS in 1h 07m 30s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/35c2b22d05ca479c97e7b5e8e4029519 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e5f2aec46684f4c98eddec9605aeca3 : SUCCESS in 50m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/628cc34c3efa4747a1b2c6bc65ac5698 : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4f0b575d27804d0da759f3555c3822ab : SUCCESS in 28m 05s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9b8fc2d9025c43dba8e269795a264873 : SUCCESS in 19m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4417f454123e48c08a8597fab278d8e3 : SUCCESS in 15m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9b5df29e561f45e0bffcf88a3b316cf2 : SUCCESS in 1h 59m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39238c571da440a2aec24452af45bd23 : SUCCESS in 1h 11m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d2d2cf57d00f4776b2138bd54f2425ca : SUCCESS in 20m 38s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d3377adfb5154add8ff3d3888e41f238 : SUCCESS in 19m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cfd66f7c2864b41996446c83ba9ab50 : SUCCESS in 16m 16s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1cc4f043c103447a83d0f1cd6fb198f2 : SUCCESS in 48m 27s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/838e993a0d5e492ba049d62b5c4ed1c0 : SUCCESS in 36m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82d457a7604d47aba6fe8a117d2bcb36 : SUCCESS in 34m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/70e4daf0d4e543ed9304b7b3ec896166 : SUCCESS in 15m 36s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/889b994cb75d4bcfaa80639acb9d47c7 : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9a64aba72d941d0bb8eccba0b71fcb7 : SUCCESS in 12m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae023891d3a44e13b9b0495c741016fc : SUCCESS in 17m 44s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6f46bccbf19e4a8a9e51c2fb9ac391fa : SUCCESS in 15m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca203d3b94004b119cca7ead8a286172 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":10},{"id":"b2e372dc24fe526fb725b627988376fcc6370e1e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 15:49:42.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3e4ea64c2da634d0e96689cc686977cb6f949dcc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-06 10:58:13.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"ace14ace8eb51fa2391cb0fe46c1ab8ff682ed73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:59:00.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"d266cdc1f59ab235f5e20c8ac38a930aab9ec7f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 12:49:56.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d86f2167b747659504e5d5f210ccab : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/910a7b2380624ad8a92daa81cf599cb5 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54bb7b2d3c5843eda4b58f0c575191af : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9ab040179144639ec23c3354525ece : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21cdb3c60b714a2fb7a92a07b8c29676 : SUCCESS in 6m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36474d6a9173494cae235c91b01529c9 : SUCCESS in 13m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d22a52406b654f3985728c5db2c236eb : SUCCESS in 27m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/13909ab1690843458dc5fc8af0457db9 : SUCCESS in 1h 08m 44s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/646ecda931c045cdb73490134bb72cac : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/833b5cd5d80e420cb2f72ec83da32af5 : SUCCESS in 55m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/02fc8b33ad7c46a49d4b19a993a4c5fa : SUCCESS in 1h 10m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/587b215ced934e99ba5f716cc801c255 : SUCCESS in 22m 06s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/1db95d2bbd1f4e13b571122453bb53d2 : SUCCESS in 21m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb9d6ea4ee2b4cb484d8add427484aae : SUCCESS in 16m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3f5502a391224f299ae1563f10b39096 : SUCCESS in 1h 49m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d163bfe07154e0c90cecbf4071dcc7d : SUCCESS in 1h 14m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/baf0c2798a0b44609c3e92c078c54556 : SUCCESS in 20m 17s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b9cf18d544cd4ab994bcc4b9d547cc37 : SUCCESS in 20m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c397b55255942ab8d375e12d7b52877 : SUCCESS in 18m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fdc63be6325f4c1e8d355234ca1caedf : SUCCESS in 40m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ef0df142c5844d9ac14ef0ae6567e62 : SUCCESS in 29m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5a098ba6f21e40c4beae29386d0e3103 : SUCCESS in 16m 52s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a215d2ae75284ba99f8003337de2c40b : SUCCESS in 16m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c02119cd9a514ce1988f8e76d0017871 : SUCCESS in 13m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9c01fd6694d14a388d7c2c2249665318 : SUCCESS in 18m 56s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7760f525b3114640b134b979af38a0f3 : SUCCESS in 16m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c76186f57a0d4cfaa6aa2f78dd8e668a : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":10},{"id":"b0f9b7225bf3707912730b4e6324823aff5654e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 12:49:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"78b077e5a356d767eef9d557709d1b0947c47019","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 12:52:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec0f110c637b460e9a0ba71a7fdedb9d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a7ee3ac5c8739d7547ae430574a5ac8abc79c41d","revisions":{"6d11391b9d907c9867e935e6a5bdbaeda1f86dd8":{"kind":"REWORK","_number":1,"created":"2022-03-29 10:51:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/1"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:49:56.000000000","tz":60},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nAdds a non-voting kayobe-seed-images-centos8s job, which does the following:\n\n* Build IPA images\n* Build overcloud host images\n* Build a base container image\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d11391b9d907c9867e935e6a5bdbaeda1f86dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d11391b9d907c9867e935e6a5bdbaeda1f86dd8"}]},"branch":"refs/heads/master"},"737c2e74b260f9300642969eea3ea527d903ea5d":{"kind":"REWORK","_number":2,"created":"2022-03-29 11:06:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/2"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 11:05:21.000000000","tz":60},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nAdds a non-voting kayobe-seed-images-centos8s job, which does the following:\n\n* Build IPA images\n* Build overcloud host images\n* Build a base container image\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/737c2e74b260f9300642969eea3ea527d903ea5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/737c2e74b260f9300642969eea3ea527d903ea5d"}]},"branch":"refs/heads/master"},"62b0424a26dc9ac434ca9d3b46340f9786b6013b":{"kind":"REWORK","_number":3,"created":"2022-03-29 11:47:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/3"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 11:47:01.000000000","tz":60},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nAdds a non-voting kayobe-seed-images-centos8s job, which does the following:\n\n* Build IPA images\n* Build overcloud host images\n* Build a base container image\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62b0424a26dc9ac434ca9d3b46340f9786b6013b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62b0424a26dc9ac434ca9d3b46340f9786b6013b"}]},"branch":"refs/heads/master"},"76a36aea2055ab6bca8aeb895abd23459276ecd6":{"kind":"REWORK","_number":4,"created":"2022-03-29 12:28:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/4"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 12:28:50.000000000","tz":60},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nAdds a non-voting kayobe-seed-images-centos8s job, which does the following:\n\n* Build IPA images\n* Build overcloud host images\n* Build a base container image\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76a36aea2055ab6bca8aeb895abd23459276ecd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76a36aea2055ab6bca8aeb895abd23459276ecd6"}]},"branch":"refs/heads/master"},"62d92f6fba7636586a4167987c53e5b9d264f675":{"kind":"REWORK","_number":5,"created":"2022-03-29 13:11:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/5"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 13:11:16.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nAdds a non-voting kayobe-seed-images-centos8s job, which does the following:\n\n* Build IPA images\n* Build overcloud host images\n* Build a base container image\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62d92f6fba7636586a4167987c53e5b9d264f675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62d92f6fba7636586a4167987c53e5b9d264f675"}]},"branch":"refs/heads/master"},"68b3d088aa2a37d2f734700ffb10077f43c1f7e0":{"kind":"REWORK","_number":6,"created":"2022-03-29 14:36:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/6"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 14:35:45.000000000","tz":60},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nAdds a non-voting kayobe-seed-images-centos8s job, which does the following:\n\n* Build IPA images\n* Build overcloud host images\n* Build a base container image\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68b3d088aa2a37d2f734700ffb10077f43c1f7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68b3d088aa2a37d2f734700ffb10077f43c1f7e0"}]},"branch":"refs/heads/master"},"579fcc9df87d6884d40b9b8a5044193bf68ad293":{"kind":"REWORK","_number":7,"created":"2022-03-30 09:38:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/7"}}},"commit":{"parents":[{"commit":"0cd0f05781cdc1915792fe1aecbde18b0aee4063","subject":"Merge \"libvirt: support SASL authentication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cd0f05781cdc1915792fe1aecbde18b0aee4063"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 09:38:30.000000000","tz":60},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/579fcc9df87d6884d40b9b8a5044193bf68ad293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/579fcc9df87d6884d40b9b8a5044193bf68ad293"}]},"branch":"refs/heads/master"},"7065111b4dc836c3c690dca405eab1cb1d9d63f2":{"kind":"REWORK","_number":8,"created":"2022-03-30 11:34:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/8"}}},"commit":{"parents":[{"commit":"0cd0f05781cdc1915792fe1aecbde18b0aee4063","subject":"Merge \"libvirt: support SASL authentication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cd0f05781cdc1915792fe1aecbde18b0aee4063"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 11:34:17.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7065111b4dc836c3c690dca405eab1cb1d9d63f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7065111b4dc836c3c690dca405eab1cb1d9d63f2"}]},"branch":"refs/heads/master"},"d946c4209798a256df375f657513a2ade769b39c":{"kind":"REWORK","_number":9,"created":"2022-03-30 13:18:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/9"}}},"commit":{"parents":[{"commit":"0cd0f05781cdc1915792fe1aecbde18b0aee4063","subject":"Merge \"libvirt: support SASL authentication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cd0f05781cdc1915792fe1aecbde18b0aee4063"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 13:17:53.000000000","tz":60},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d946c4209798a256df375f657513a2ade769b39c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d946c4209798a256df375f657513a2ade769b39c"}]},"branch":"refs/heads/master"},"a7ee3ac5c8739d7547ae430574a5ac8abc79c41d":{"kind":"REWORK","_number":10,"created":"2022-03-31 08:52:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/835610/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/835610/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/835610/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/835610/10"}}},"commit":{"parents":[{"commit":"0cd0f05781cdc1915792fe1aecbde18b0aee4063","subject":"Merge \"libvirt: support SASL authentication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cd0f05781cdc1915792fe1aecbde18b0aee4063"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 08:52:17.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7ee3ac5c8739d7547ae430574a5ac8abc79c41d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7ee3ac5c8739d7547ae430574a5ac8abc79c41d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
