)]}'
{"id":"openstack%2Fkayobe~835801","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I3c59fa140a89d7d761846cc712e65aa0efda1363","project":"openstack/kayobe","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-04 13:44:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c59fa140a89d7d761846cc712e65aa0efda1363","subject":"CI: Avoid image downloads and builds in seed jobs","status":"MERGED","created":"2022-03-30 11:15:39.000000000","updated":"2022-04-04 13:45:44.000000000","submitted":"2022-04-04 13:44:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835801","cherry_pick_of_change":835811,"cherry_pick_of_patch_set":1,"meta_rev_id":"bcb1047c623197dd60392ce70ebdafb87c9a0d68","_number":835801,"virtual_id_number":835801,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-04 13:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-04 12:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-04 12:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-04 12:28:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-30 11:17:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 12:09:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-04 12:28:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"5acd9e3954e615da56713491ab206307db034fff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:15:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* playbooks/kayobe-seed-base/overrides.yml.j2","accounts_in_message":[],"_revision_number":1},{"id":"b257cf55e5ba34125aacb6fe34047d36b3f271c3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:16:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ecef3f6c36bb1ff8592819e609cbeae20d825df9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:16:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"edd8f10b5eee02fba25e269971935e40f778d420","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:16:41.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"a79e2c3b5a4b0a9b94275ecef2716218c0f368d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 11:17:05.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kayobe for 835801,3","accounts_in_message":[],"_revision_number":3},{"id":"7467ffbd863868f749cba16ba335cce551b051c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:24:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1c449ef0e9b6a16a01d540d212a2fd992ce44573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 12:33:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dff24a10fc447d38e6ce73ef0294ed4 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790427e3f44643ba860f8a3c8a572942 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0373c05d82414a6486e0f62913cc6cb4 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c20d3a77d6f94789a58b0f40e3d1c7b9 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b0ab0dddc6b4311bd07ca09a167f942 : SUCCESS in 8m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a18f25930ba40de8773fa027d4dd3a4 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1d462fe8c5c4e59abcb222b1af638d7 : SUCCESS in 10m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dfa6db722eb547a8bf7f9a83c19e87d8 : SUCCESS in 20m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d09cdc250c6f4b1b899ed207bf6804de : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3463f3879cc145e29f029879d55d954d : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a07e39b645874d9e8b1b15df42d973a7 : SUCCESS in 16m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c8a80e9b663447199fcaddc346f0e4eb : FAILURE in 18m 46s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/08a8736814a24fc9bed8815aa238a8e0 : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/651cb4bd41154b77a2fb75c6f75b9a9a : FAILURE in 20m 06s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/779f35541e52463ebcc54035885b4043 : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":4},{"id":"933ff6e84fbd48d2813c9d7e7a37b10b0b750028","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-04 12:09:12.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d52437e5f0efb24411a33dbf092e6d2afc809730","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-04 12:28:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5d32b96f10895acefb96aa126140a9d9889ee457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 12:28:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"52715edb72e667e4602f37f88e0150bc35f648a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:44:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ccb7f470a841cbbd8aef03eb7df159 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a92b1ebbe888424297bc9e95fb64e85f : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bedc2553d814a4ca856762fe0c5aa45 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd41641e64894e8ca509122952146d10 : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a6426cc1fc4b4de7afe8445800b06ad0 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a5bc8b10fc943d38e29870100340e54 : SUCCESS in 8m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e20b2875db5a451994d75bd9c92aeb23 : SUCCESS in 15m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b3ed73735d6a4cbc8caa65ab0c986b3a : SUCCESS in 51m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ed89ae38cba743a8aa17bd16500223ac : SUCCESS in 1h 02m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/593516962d864dd99d86dbcdce6a440c : SUCCESS in 15m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9ed7d8716b5b44788e6bfef708919fd8 : SUCCESS in 14m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a2bd3093974148198ddebf0594fad5be : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":4},{"id":"81867cbb09ec8763eefdf322c2789991ae1adb42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:44:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bcb1047c623197dd60392ce70ebdafb87c9a0d68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:45:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c1a6ec0f9104d1495ff2696469448c1 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b3f6e79b3d6c87a331b79c1869d4b4dffed34667","revisions":{"d30b0ef7a7a90fb52d5931b467a71dea8793ee94":{"kind":"REWORK","_number":1,"created":"2022-03-30 11:15:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/835801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/835801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/835801/1"}}},"commit":{"parents":[{"commit":"0ea814e0b9f85dce90079191fb284b91d4aaf774","subject":"Merge \"Use naming convention to infer VLAN tagging\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ea814e0b9f85dce90079191fb284b91d4aaf774"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 11:15:39.000000000","tz":0},"subject":"CI: Avoid image downloads and builds in seed jobs","message":"CI: Avoid image downloads and builds in seed jobs\n\nThis change is a combination of previous changes, with some adaptations\nfor changes in Wallaby:\n\n* Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n* If657bf5b0117812d3c53942464cc41cf86cc8ad5\n\nThe overall effect is to disable IPA and overcloud host image builds in\nseed jobs, since these can be unreliable.\n\nChange-Id: I3c59fa140a89d7d761846cc712e65aa0efda1363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d30b0ef7a7a90fb52d5931b467a71dea8793ee94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d30b0ef7a7a90fb52d5931b467a71dea8793ee94"}]},"branch":"refs/heads/stable/victoria"},"610ddc9c141dff75c3e4704fe995c28bd58b230e":{"kind":"REWORK","_number":2,"created":"2022-03-30 11:16:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/835801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/835801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/835801/2"}}},"commit":{"parents":[{"commit":"0ea814e0b9f85dce90079191fb284b91d4aaf774","subject":"Merge \"Use naming convention to infer VLAN tagging\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ea814e0b9f85dce90079191fb284b91d4aaf774"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 11:16:23.000000000","tz":0},"subject":"CI: Avoid image downloads and builds in seed jobs","message":"CI: Avoid image downloads and builds in seed jobs\n\nThis change is a combination of previous changes, with some adaptations\nfor changes in Wallaby:\n\n* Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n* If657bf5b0117812d3c53942464cc41cf86cc8ad5\n\nThe overall effect is to disable IPA and overcloud host image builds in\nseed jobs, since these can be unreliable.\n\nChange-Id: I3c59fa140a89d7d761846cc712e65aa0efda1363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/610ddc9c141dff75c3e4704fe995c28bd58b230e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/610ddc9c141dff75c3e4704fe995c28bd58b230e"}]},"branch":"refs/heads/stable/victoria"},"c773099f4692b679ed0e06c8303ce73ae5be2ad5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-03-30 11:16:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/835801/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/835801/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/835801/3"}}},"commit":{"parents":[{"commit":"0ea814e0b9f85dce90079191fb284b91d4aaf774","subject":"Merge \"Use naming convention to infer VLAN tagging\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ea814e0b9f85dce90079191fb284b91d4aaf774"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 11:16:37.000000000","tz":0},"subject":"CI: Avoid image downloads and builds in seed jobs","message":"CI: Avoid image downloads and builds in seed jobs\n\nThis change is a combination of previous changes, with some adaptations\nfor changes in Victoria:\n\n* Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n* If657bf5b0117812d3c53942464cc41cf86cc8ad5\n\nThe overall effect is to disable IPA and overcloud host image builds in\nseed jobs, since these can be unreliable.\n\nChange-Id: I3c59fa140a89d7d761846cc712e65aa0efda1363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c773099f4692b679ed0e06c8303ce73ae5be2ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c773099f4692b679ed0e06c8303ce73ae5be2ad5"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"},"b3f6e79b3d6c87a331b79c1869d4b4dffed34667":{"kind":"REWORK","_number":4,"created":"2022-03-30 11:24:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/835801/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/835801/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/835801/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/835801/4"}}},"commit":{"parents":[{"commit":"57651f454891dbee03f33939fc88293bc07e6898","subject":"Merge \"Build CentOS stream images\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57651f454891dbee03f33939fc88293bc07e6898"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 11:24:33.000000000","tz":60},"subject":"CI: Avoid image downloads and builds in seed jobs","message":"CI: Avoid image downloads and builds in seed jobs\n\nThis change is a combination of previous changes, with some adaptations\nfor changes in Victoria:\n\n* Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n* If657bf5b0117812d3c53942464cc41cf86cc8ad5\n\nThe overall effect is to disable IPA and overcloud host image builds in\nseed jobs, since these can be unreliable.\n\nChange-Id: I3c59fa140a89d7d761846cc712e65aa0efda1363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3f6e79b3d6c87a331b79c1869d4b4dffed34667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3f6e79b3d6c87a331b79c1869d4b4dffed34667"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
