)]}'
{"id":"openstack%2Fkayobe~835802","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I8cc9531570b76a8282a95c4036324e9d1025d7cb","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-31 09:09:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8cc9531570b76a8282a95c4036324e9d1025d7cb","subject":"Fix custom config idempotence","status":"MERGED","created":"2022-03-30 11:23:06.000000000","updated":"2022-03-31 09:10:59.000000000","submitted":"2022-03-31 09:09:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835802","cherry_pick_of_change":829301,"cherry_pick_of_patch_set":2,"meta_rev_id":"695a569d01676be5b1e4ff0397ba9abb9cbf8f1d","_number":835802,"virtual_id_number":835802,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-31 09:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-30 13:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-31 07:12:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-31 07:12:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-30 13:18:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 07:12:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"809e6af1979e1e5dbf49c95904fa290521fb8178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:23:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f8107eec1013f5d2fc8df2a47263ab2d5b0895f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 13:18:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cd9e7f229964e5b8c398a61ad76748d : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7243c3aa025743a78a385e6c2c17f014 : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/591adf3109db4f2ca705f27f96944e20 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a2e076ed03c443c8fef4a064cfefb7c : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0047de49eb34a55899010a3b70ea96e : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d8930aeb78b4f488ae4b78d6e0b28ff : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1f3cfa2a1d849cca3707eaf74c66e1f : SUCCESS in 5m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b922849f0dcd4d2d936c71a0764f882f : SUCCESS in 7m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b43d514c97934e35bf70103c29e05783 : SUCCESS in 15m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e461fdc22084b10b96ff7b188729523 : SUCCESS in 27m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9a700935915e4746bd360bcddb0af7c2 : SUCCESS in 1h 13m 16s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/87a2a4b23ad1475c8efc553f3e4ba605 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2496e22655114f2781f27798c26add8f : SUCCESS in 58m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/56b665d2262e46799dd08f818ea05c60 : SUCCESS in 1h 28m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cd6a49f8a5e446e482d65eb513512ba4 : SUCCESS in 27m 25s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5948fc8d3d9441b39d11fa76c4d9fad0 : SUCCESS in 23m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0f5859d72b04370b8412efb7777f944 : SUCCESS in 18m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2e2f6b26cf1c46c8a23720c9591f60ee : SUCCESS in 1h 53m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c799bef6ade41f5be75b8392444bf15 : SUCCESS in 1h 25m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/61b0679822c7406fb4d75a4a12f92f55 : SUCCESS in 29m 04s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/aee0ea2fcda44acc9282242f0de75114 : SUCCESS in 21m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92e6c816ddf24beba6f3647deb582d20 : SUCCESS in 18m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6e13d90257ba434f80d1b99761ed098c : SUCCESS in 33m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42951b1bcedf4ca78fa49bc85f1e6098 : SUCCESS in 25m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/37045eae7b2342cc8df00ccc29d55a40 : SUCCESS in 19m 02s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3c33f444f73e4ae585e558e3fe9bd5c5 : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/babbaa4654ff4a0fa6e8803df23c5322 : SUCCESS in 14m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e4fbfe9fad26473b832dc4ae016004bf : SUCCESS in 24m 19s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bbfc3a7cbc1749f897e4803a69e33c45 : SUCCESS in 18m 46s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c14831e3216d44459d8ba24b37175005 : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f886cee7642f2737ab05aaa4f748616896b19af8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 13:41:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"081497d25667d79eb8226d64055be22302b2cf04","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 07:12:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2164ac5686d06491943d5a5da4c679faedcf9e05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 07:13:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e70b08059fb3cae6b371c897ba323c28871f913c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 09:09:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7690f473876a43058effaf1e527f94b7 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76766cf86c9344c89df239a250dbc791 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b38b0da30b6048508bf79cc7d13941dd : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b369b5a590a645b589c1e898bab638b1 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/402db4cd75f04c17a06d3c0a46c41c89 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f718f5ab871455584b21fdd461ea114 : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/028c28680ad44f7a96544c5bddb33e31 : SUCCESS in 13m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a5a09d269914c93987d68dae6e274af : SUCCESS in 31m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e970a93465de4cf1803af5ec588c38b2 : SUCCESS in 1h 12m 39s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/43889d508cd140c999675c19c6953777 : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8b0bd5c968b4ca2b538fa6b84a1b639 : SUCCESS in 54m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/87e3c482afe146778f7941d8c8cc8f83 : SUCCESS in 1h 10m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a17f9a21a95d4935991ef49b848a677c : SUCCESS in 24m 46s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/224cafd9fb7f4546a7bcefa318423816 : SUCCESS in 22m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4b67108da654c35a964f34b6be4f7c1 : SUCCESS in 14m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/844dc5afc86b4c5487fbcf42a172475b : SUCCESS in 1h 52m 36s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4327ec97b0554d2db560a2884ce500ab : SUCCESS in 1h 24m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/527cf5dfd5994127b1219b83b9a1057e : SUCCESS in 26m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/aab7261abdbd43f29a86c1cb627631f5 : SUCCESS in 18m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06595b485024459a924e2e7fefd7a0f7 : SUCCESS in 17m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3c12f8b3b393489285b20e91bd28ac5a : SUCCESS in 30m 04s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03e89521ed9f4cf9af1996d5bb406422 : SUCCESS in 25m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/24f3b4da12514a99a840890b0425dd56 : SUCCESS in 17m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d7bb418c371d4f95ab3cc71c770d479b : SUCCESS in 16m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f7c46e7b36741bdb55e06bf060eaad7 : SUCCESS in 11m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c7aa939ba13c4d2d80238ccd632d8848 : SUCCESS in 22m 32s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3074a09b33254dcc9311ab8f3798c759 : SUCCESS in 15m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94a289ae12704e8694ad9940441c844f : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a9d2b1036ebe63d3aaa103cb38b2216ffa60c7c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 09:09:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"695a569d01676be5b1e4ff0397ba9abb9cbf8f1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 09:10:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203de2011098496fa24611a174525b95 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b33ea61ae86047f296bdd31c6ccebfc3 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a451ff7af762574870a89ad193063f6358001ed5","revisions":{"a451ff7af762574870a89ad193063f6358001ed5":{"kind":"REWORK","_number":1,"created":"2022-03-30 11:23:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/02/835802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/835802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/835802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/835802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/835802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/835802/1"}}},"commit":{"parents":[{"commit":"4efb80a10d60268e42e64741da09317e42d5e67e","subject":"Ubuntu: add support for Apt configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4efb80a10d60268e42e64741da09317e42d5e67e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-15 15:11:23.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 11:23:06.000000000","tz":0},"subject":"Fix custom config idempotence","message":"Fix custom config idempotence\n\nWhen some custom config files are used, generation of local\nconfiguration for Kolla Ansible is not idempotent. This happens because\nan executable permission is applied recursively in a directory, then\nreverted on regular files.\n\nChange-Id: I8cc9531570b76a8282a95c4036324e9d1025d7cb\n(cherry picked from commit 8de9d5368180f733d5ea55b36e152050137095d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a451ff7af762574870a89ad193063f6358001ed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a451ff7af762574870a89ad193063f6358001ed5"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
