)]}'
{"id":"openstack%2Fkayobe~835804","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I8cc9531570b76a8282a95c4036324e9d1025d7cb","project":"openstack/kayobe","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-31 09:09:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8cc9531570b76a8282a95c4036324e9d1025d7cb","subject":"Fix custom config idempotence","status":"MERGED","created":"2022-03-30 11:23:23.000000000","updated":"2022-03-31 09:11:00.000000000","submitted":"2022-03-31 09:09:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835804","cherry_pick_of_change":829301,"cherry_pick_of_patch_set":2,"meta_rev_id":"9e39cfb6deced1c66399a77fca5304c166592440","_number":835804,"virtual_id_number":835804,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-31 09:09:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-30 13:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-31 07:12:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-31 07:12:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-30 12:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 07:12:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"574efe1ab1de2c33ba3f4fd4508f5b830820de40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 11:23:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dc6d4b681950312e4d18279c0007605682117a4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 12:33:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9e3a93bda6a49c1888eb16eca0f9a94 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc25fc752de74a77b3bccd112f70012b : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/651ad7d56c8541f99599e54d92faebef : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa67bfe838ae44b1b7e4c6509c2123d8 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7cb23b704bf415294a67ed8a723acd5 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3beb7c027acc4d559304c2fb1c1c0b51 : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ac55f45ef7d640e98a165a97d26c2c26 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a43532034ddb4d04b17b16691906bea5 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2e89841cd7714a578201b68a94fac5d7 : SUCCESS in 18m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8c4d0723fcf64b7aabe40292421f360c : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0854d0dbbbbb470cb4d5066afb42512a : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0d77cea62b10403e8131f92910dd6bdd : SUCCESS in 20m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/16af9080b7144a919791caba9bf3a621 : FAILURE in 17m 47s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e5f5b6b3104048be9f827dd545e2da86 : SUCCESS in 30m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2a16b09eeb0e42c1ad11f2f46abdb6ae : FAILURE in 20m 32s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3f67927d83df4b7d88374c3b48239c3d : SUCCESS in 16m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3055bfadfd7ec1138bc74ae26afb1a0d37211780","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 13:41:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eee9f34a165624c701149355a48e0e12d2b7931d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 07:12:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"969c0715b684fe99219688d55b645c56d8ecf47a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 07:13:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97bd95faadb5aecea5a59e6710670e7f57c1ebfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 09:09:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07b62eb6e71b4e84baaa07c4bd39ee0f : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7193304624a34590b49a908865381cca : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/668d640a24a948658b788ee2b420fa77 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94f1ddde55e4953894c75320af6ee97 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef8f5a1407754842abf1a9b51444004b : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab5d1031007840098987a55e607c914b : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd925d11f7b6406d825bbd1eb38d3131 : SUCCESS in 8m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/def8448f0d3a42c1bf6b3e454a5e8c76 : SUCCESS in 16m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/877eb15288304c36b2a5196086c05150 : SUCCESS in 1h 02m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/460a07f8f4724df9af318e40cdc20610 : SUCCESS in 1h 12m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4ba9c25456d54469961e664424a95821 : SUCCESS in 15m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7eb74f0ffb9c42fc950ae5b298dd7b61 : SUCCESS in 32m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6db11fa8466e402d81b269f54d3fabc0 : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1181fb7e4a3e967ecad9a0947522323afd0559cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 09:09:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9e39cfb6deced1c66399a77fca5304c166592440","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 09:11:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55671018ead94c4f81af9f62011b0be4 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e873a2b31508445c9475e311a6b608af : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6cb821c43e49d672a1755f8568790aa05c5a593","revisions":{"f6cb821c43e49d672a1755f8568790aa05c5a593":{"kind":"REWORK","_number":1,"created":"2022-03-30 11:23:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/835804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/835804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/835804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/835804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/835804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/835804/1"}}},"commit":{"parents":[{"commit":"57651f454891dbee03f33939fc88293bc07e6898","subject":"Merge \"Build CentOS stream images\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57651f454891dbee03f33939fc88293bc07e6898"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-15 15:11:23.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 11:23:23.000000000","tz":0},"subject":"Fix custom config idempotence","message":"Fix custom config idempotence\n\nWhen some custom config files are used, generation of local\nconfiguration for Kolla Ansible is not idempotent. This happens because\nan executable permission is applied recursively in a directory, then\nreverted on regular files.\n\nChange-Id: I8cc9531570b76a8282a95c4036324e9d1025d7cb\n(cherry picked from commit 8de9d5368180f733d5ea55b36e152050137095d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6cb821c43e49d672a1755f8568790aa05c5a593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6cb821c43e49d672a1755f8568790aa05c5a593"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
