)]}'
{"id":"openstack%2Fkayobe~835933","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I532f43d5fcf67de73f1be30d83157311b474de93","project":"openstack/kayobe","branch":"stable/xena","topic":"xena-feature-flags","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-04 13:38:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I532f43d5fcf67de73f1be30d83157311b474de93","subject":"Sync Kolla Ansible feature flags and inventory","status":"MERGED","created":"2022-03-31 05:33:42.000000000","updated":"2022-04-04 13:40:06.000000000","submitted":"2022-04-04 13:38:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835933-xena-feature-flags","meta_rev_id":"7a38be030f4fb48ac7cb2845429e9086ef6828b0","_number":835933,"virtual_id_number":835933,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-04 13:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-04 12:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-04 13:40:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-04 12:08:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-31 07:13:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 08:49:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-04 12:08:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"fc3e1358f44b878e9aa828d5b09d16f1a10e17cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 05:33:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87e926ff426d8b28ff918734f726ec14fb18416c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 05:37:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3c82302bab23e19ca6a4cb96ec89fb30032f0cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 07:13:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ad33573541f46adad1d1c94ce8249e0 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69c9b9d0c3e4058a45b19cc6eb414ac : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8ba2c4019a94947975e5041b2c72047 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51f1469ba1204f1fa82503ebcb3c7a4b : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be27dfe8b41046078780ae178324ff1d : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d6da3d9c784f03903b22bcb9a75ff7 : SUCCESS in 8m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cf8aba45d4b34318ae74acbef760bac8 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c05650dbf7c34cc88cdc1bdc2ed376e3 : SUCCESS in 10m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bac756744ef04db1a8c7f3358580831c : SUCCESS in 31m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3280afcd0e7b4d4ab869aeb463cb93b3 : SUCCESS in 1h 17m 41s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b65e1288bb2f49158fe96885a89a0149 : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3f2a5291b2e4f9c8e80f41a545983d6 : SUCCESS in 59m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a2e28b7e0ef489095bd2b502c8d206a : FAILURE in 1h 11m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ea41f287aefb481497403189ed4cf97b : SUCCESS in 21m 00s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/094b9f3bd7184658987f546ca12e2f84 : SUCCESS in 21m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57bd5a8f395a4f37a94adf97b1def39f : SUCCESS in 15m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7ffcd12add214df39e20a54691dd0e86 : SUCCESS in 1h 34m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b3d1c7b662a4f5784d8e35b2dc5923b : SUCCESS in 1h 02m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/96454122f01b4d0c8836123d22e7ce93 : SUCCESS in 25m 18s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a9469d511d2b4530ba370e2e8714e2c7 : SUCCESS in 20m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd14a39d677d4f68af39ed3ad271a4e0 : SUCCESS in 15m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6ced83cd67154c0b94dab0f268f0ddd9 : SUCCESS in 30m 14s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/503305f727d041c498b38dec3e77c333 : SUCCESS in 27m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f18d7da9b9784f708b4cce148ec3f5a2 : SUCCESS in 15m 55s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9c030bc2a9e34fc595e57419a665be4b : SUCCESS in 14m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdbf3c6af17345b88ccfff86b52b87a4 : SUCCESS in 11m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c17ca25ebdf744eeb783c75cca92ecc6 : SUCCESS in 18m 19s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1d41ace9f8c54bdca445b6298de15068 : SUCCESS in 16m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a241b110ae4d47eb829c3fd7a8935a71 : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":2},{"id":"043778f407ebb11fb724fb9bb3c48585d2ae4cc9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 07:16:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3b96dab59330b951474cbcb481e20ffef2133e1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 08:49:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25b88782aff4b8b3de30c30a10d923d2496dea43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 09:39:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f6bcb5045e841ecbb5513b0d0e8a7f1 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81954cc52d764ec88f196765b7de9981 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ef447b4725d4c24aa2382562dc807f1 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e243762da02646f2a3e581cb463ed263 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6b210fc75c8418d9997ca100070ba05 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7820d6eb6e884a6c928cb5091610204a : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7030ffb44d44e398b699f7ee09b39b4 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6acf2e0e325b4acfa55ba9d69f14ad01 : SUCCESS in 13m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94ec9c07ce5442c293c1ec139054d3fe : SUCCESS in 33m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6ab0f444a3a645d28338160ec3465d53 : SUCCESS in 1h 20m 37s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7be1c6ffab2b4f00b481f91bf947930c : SUCCESS in 1h 10m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dc4984266da426d88f4a4834b15ec7a : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7b749f6db3764d18a35dfb90096211da : FAILURE in 1h 25m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8e5988ba6de54c0db37ce502b6e67695 : SUCCESS in 27m 47s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/77d465a2c4f14b38a9e3e52276cb6dee : SUCCESS in 22m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3223d4b4006c439180ffdddb2de34887 : SUCCESS in 15m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/76b764b531804d7a980e6f830d643011 : SUCCESS in 2h 20m 17s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83b31033fdb54f4683f641001190de5b : SUCCESS in 1h 12m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7a2d72504d1d4e7a9f40d5b299e82e3c : SUCCESS in 38m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/562a20ef5b35489e98d57bce48030d73 : SUCCESS in 18m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b159b941c994c0a9bb4a5700f3d78bb : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4f877b2136474c258e65f4b07c40b049 : SUCCESS in 32m 42s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38ba6d9bea054ef1b0a277d4a51b4742 : SUCCESS in 30m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/76760b37e782400e92018534bdc91ba7 : SUCCESS in 26m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a326054bd1cc4fc8b9ad41da7f93f2de : SUCCESS in 14m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33b0e20beb2f4021b8de6130df3c7e20 : SUCCESS in 13m 33s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b7a7f2dc31c54f13a9217b7ffb3a9846 : SUCCESS in 17m 17s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cef57339629f4d7d8370d06604768155 : SUCCESS in 19m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73b807fcc9c145869e0687dd27d76cbf : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":2},{"id":"03fa01c1c308a4de7d02e4bca58704d91989e974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 13:44:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"53f87b69ad62f05f3da58b05554dc3c9c7fbacbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 15:43:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a57ec880b434b3ab71b79e1f976a0e5 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a433107d5c042c2843aa5bd330ba8a4 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9dfe1c8a073f4cf7a0ea46a6d5d2b5e3 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/033fa21062a8488baacea53faaebf01c : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d440fba84b03439d97c4a15b358e26ad : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/078ae9dfda8e4078b51ad294c72adc79 : SUCCESS in 9m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19f3f57c80e34e968d663c0cd2387c06 : SUCCESS in 7m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72fc0f23bc344b7cbcb045db7fcfed19 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b450576bda2d4454bdfe227453445f9a : SUCCESS in 32m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/34995a18990741b2a7cc4a5ac267038a : SUCCESS in 1h 07m 09s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/635c3a31f4f4434981407120342bf651 : SUCCESS in 1h 20m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf01c2445fab4d959cd189ec9868046d : SUCCESS in 52m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bc4850b30ce84173a942f1a9bc4c7996 : SUCCESS in 1h 09m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/941d987bed6e4a42b64ffa8edb5d1826 : SUCCESS in 23m 22s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/32f75dd63e684c19a44f28b8d187addf : SUCCESS in 26m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ee8f089348848f1ac75ba4419a0c139 : SUCCESS in 16m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/491ef3b867774eba827905fef3130fc2 : SUCCESS in 1h 52m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fd338232a8a49438106be3d554539b1 : SUCCESS in 1h 16m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f8ddc786209d403bab8d5eacc35a21f9 : SUCCESS in 26m 19s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/33c780d7e8fd4c8898bdd14ca9c14ba4 : SUCCESS in 18m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22003ca9c90e4853b29fb8dfe6c0594a : SUCCESS in 15m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6ea68c872cd24fdaa83db4a123b99f1a : SUCCESS in 30m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e02da4ad986e4807815c2c967e22b61d : SUCCESS in 27m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/eb7357832bf44467a7d1747dd2bf554a : SUCCESS in 18m 26s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f12f2499d9464d95ae341400a896d4e6 : SUCCESS in 14m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31bbeced60054d0eab284600e18782c8 : SUCCESS in 12m 50s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f76b4f665711491f8ba30e72a6ec9f8e : SUCCESS in 19m 16s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/82eac3f366ba4dc3be8f7d72d72fe174 : SUCCESS in 16m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d51ea43db9644cd1923c96495916737c : SUCCESS in 14m 15s","accounts_in_message":[],"_revision_number":3},{"id":"187a74cfa8d85e811412e905746107787aac9f92","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-04 12:08:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e58db8480197850aa9a5151b4f4b1a2340774990","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 12:09:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9705da6f51349d4b0c08cb99dc32ebc2f30376df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:38:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7f3b6f5eab4e0183d67a96085e7a7e : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0615b2674b8b477aae941328f145d7c4 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afbfc124767e40b59469ebc1645227d7 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab464001851747f8ad8517bd879fbbd7 : SUCCESS in 8m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bccc9be614664d49b5bf906c7fae7ae3 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d804dace93fe46fe82c64d8e016558b5 : SUCCESS in 12m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a71dc7beac54af4b8a283043b343b93 : SUCCESS in 31m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/339818a12d904acea3443e453ec1d5fb : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/76725877ab154f90a130881c54b8c341 : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27544dc985dd419c9e12eda766f2d573 : SUCCESS in 55m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/44bdb3f9b40a42f78c7ceb8fe2f47e17 : SUCCESS in 1h 15m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d62878ca26ac4feaae0f878ceb06e2c9 : SUCCESS in 24m 04s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b6176c854ebc43c8b075f42a58752123 : SUCCESS in 23m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c559d845e6b404895361816a054c590 : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/387c7f2dd811470f8ead631a5d5c3062 : SUCCESS in 1h 28m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/576dee0aa4b74ac8842d106bbcb8203d : SUCCESS in 1h 15m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b6b7fb86383846939ba0e2ba2afde8f4 : SUCCESS in 21m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d54a5b3ead2947609e27fe92f48124ed : SUCCESS in 18m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4afd8d307ffc4abe9f961fdc5d32fd61 : SUCCESS in 13m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fe5ee57e7c034628b64b1d446a9843b0 : SUCCESS in 29m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0cce62deaa04a1c82fba3dbefcb3e61 : SUCCESS in 22m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7ae429540eee46bfb3b0d556cab01423 : SUCCESS in 14m 20s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5dec368b07024842b26eb9b483c043a0 : SUCCESS in 15m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a424f15de6a3413ba71edf8fac763e24 : SUCCESS in 11m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8edb673dbfa4459fa2929dc169b10f8b : SUCCESS in 16m 48s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4b4f54e47e9949a9a3c6269def9c2aff : SUCCESS in 16m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/264acd70ddb247e5a91acbf79ac6f44a : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":3},{"id":"394ca45130033b10e2bb89f4621d32dbd0c4b515","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:38:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ec8c90661ac47f33c32d684ae45df3d61ef84b1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:40:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d36b2ca5c44b99905f864a237af5c3 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ef3bb4075db61f97c2dc9e89655fb4c155d73439","revisions":{"c163114f1258de40c5e8c051a4af9ff65d5ad78e":{"kind":"REWORK","_number":1,"created":"2022-03-31 05:33:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/33/835933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/835933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/835933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/835933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/835933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/835933/1"}}},"commit":{"parents":[{"commit":"4efb80a10d60268e42e64741da09317e42d5e67e","subject":"Ubuntu: add support for Apt configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4efb80a10d60268e42e64741da09317e42d5e67e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 05:32:40.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 05:32:40.000000000","tz":120},"subject":"Sync Kolla Ansible feature flags for Xena backports","message":"Sync Kolla Ansible feature flags for Xena backports\n\nChange-Id: I532f43d5fcf67de73f1be30d83157311b474de93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c163114f1258de40c5e8c051a4af9ff65d5ad78e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c163114f1258de40c5e8c051a4af9ff65d5ad78e"}]},"branch":"refs/heads/stable/xena"},"07da34935c8d696cc87efebadc55a39654683dcc":{"kind":"REWORK","_number":2,"created":"2022-03-31 05:37:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/33/835933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/835933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/835933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/835933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/835933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/835933/2"}}},"commit":{"parents":[{"commit":"4efb80a10d60268e42e64741da09317e42d5e67e","subject":"Ubuntu: add support for Apt configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4efb80a10d60268e42e64741da09317e42d5e67e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 05:37:21.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 05:37:21.000000000","tz":120},"subject":"Sync Kolla Ansible feature flags and inventory","message":"Sync Kolla Ansible feature flags and inventory\n\nChange-Id: I532f43d5fcf67de73f1be30d83157311b474de93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07da34935c8d696cc87efebadc55a39654683dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07da34935c8d696cc87efebadc55a39654683dcc"}]},"branch":"refs/heads/stable/xena"},"ef3bb4075db61f97c2dc9e89655fb4c155d73439":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-31 13:44:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/33/835933/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/835933/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/835933/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/835933/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/835933/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/835933/3"}}},"commit":{"parents":[{"commit":"a7791250b908f4ab95fe4a123eaa4847d928b66b","subject":"CI: fix TLS job by freeing up memory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7791250b908f4ab95fe4a123eaa4847d928b66b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 05:37:21.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 13:44:02.000000000","tz":0},"subject":"Sync Kolla Ansible feature flags and inventory","message":"Sync Kolla Ansible feature flags and inventory\n\nChange-Id: I532f43d5fcf67de73f1be30d83157311b474de93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef3bb4075db61f97c2dc9e89655fb4c155d73439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef3bb4075db61f97c2dc9e89655fb4c155d73439"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
