)]}'
{"id":"openstack%2Fkayobe~835951","triplet_id":"openstack%2Fkayobe~stable%2Fxena~If483a6a6fb6d5b2c9b6b7dbd22939b0b46599538","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-31 13:29:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If483a6a6fb6d5b2c9b6b7dbd22939b0b46599538","subject":"CI: fix TLS job by freeing up memory","status":"MERGED","created":"2022-03-31 09:16:03.000000000","updated":"2022-03-31 13:31:15.000000000","submitted":"2022-03-31 13:29:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835951","meta_rev_id":"15a440529d801e063eb34c5860b9207f4afdbcdd","_number":835951,"virtual_id_number":835951,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-31 13:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-31 09:33:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-03-31 09:33:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-31 09:33:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-31 11:18:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a0644abb330f943926afc91adb570e511c25f54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 09:16:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8400c1cf8e1f113524fdb39dd5647f13f0c1424","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 09:33:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a252239418437ead8069b8e8fbdf98e2ac51150b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 11:18:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/153dce1b5c974c63b087a5108eb39065 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbafa3435958483c866d9dbc39857448 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/211acfb8fa74449ba35de4fcb83c852a : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdd68caf608e47fb888c117917863da6 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1135e83fb8044554accada9b4a170ecb : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48499fe71ae44b83a05e0dfd32f8f50e : SUCCESS in 7m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8bd1f2d482ee4e269d37a2007009f549 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc747c6801c14695aeea228443b35098 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8c2cd284dc204beaadd6678ed137ab01 : SUCCESS in 32m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e86f54aa766a41a2ba9167b861d6ca9a : SUCCESS in 1h 03m 41s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/aa7c189b94be4d1aa83e10e895063357 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/163c94b59e644a608779d8912da05ddf : SUCCESS in 52m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/877e5bed72644d3cb34e61f4a96d9242 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ab56a3b73e7246c5af04b9392cf179cb : SUCCESS in 23m 32s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7c7dafef81bf4c949c5019cc05d0f938 : SUCCESS in 23m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8eb2a795c5364c6c8b807834c5f8efcc : SUCCESS in 14m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/767101819aa4414692dabf5d17c01d4f : SUCCESS in 1h 48m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c132c11292774de1849f45d54bae4f43 : SUCCESS in 1h 21m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0f7de0e425ec49839b339e43c51acf13 : SUCCESS in 26m 09s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/50e0dae9a87441599760a115e8dbeebe : SUCCESS in 20m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a6f205c2887476991cd1a328c4f1b34 : SUCCESS in 17m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/787ad1dd86bd4124bf1038099c640dc0 : SUCCESS in 28m 59s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7234e11b94e646288ddc9339e73c0788 : SUCCESS in 26m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/55182ec895c3421ea3e69f1534dcd559 : SUCCESS in 19m 28s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f88974e895384504a4135f87362cb927 : SUCCESS in 14m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c682bd20236d470c8d38cd6bab25f817 : SUCCESS in 12m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e54f0e85cdc24e1ab05ba868da88aba7 : SUCCESS in 19m 05s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3a05e53015104cb293417815b7ba15cd : SUCCESS in 17m 28s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/368cb8bdb67e40bc82b366df27c35d45 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bf4690a7b287d0519c67ba01b97fa306da041dac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 11:18:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7463d8b46fbd603136a85b570510cfcec95bea23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 13:29:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b115685aa6a14bdc941d8e74b40ae0fe : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f214fb881a514ff4b465d154590d345b : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e91ae93b03546c38ddaa5bff5cc18d3 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c399b6493eb5455c86b13548e7a1a6ba : SUCCESS in 9m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ce76bbf986c449db0ca222bfb043fc7 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/929fd5206b304fc298ae1b2a5ef14f42 : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f8b55777a7dd4827840d0565720a6740 : SUCCESS in 31m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2fba89a597f04443bf85b120cf5b9294 : SUCCESS in 1h 12m 08s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/cd6e441a93dd49ac87a96d9f5329003d : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ffa32051dfb43cdb7c695a65d8583e4 : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/50993be59f8747ffaa2ff28dd2963c0d : SUCCESS in 1h 13m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5cc7bae39ff84053b399db7aca5f291e : SUCCESS in 24m 42s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c166a75cd4d3438aac763f566477048c : SUCCESS in 22m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a725e5a9ce2445baa6c8b2dd354ffa4e : SUCCESS in 14m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cbc6c1de8c8d4b9690055b539b1c462c : SUCCESS in 2h 08m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/834e8cb0484145448fe1c331a8600f1e : SUCCESS in 1h 24m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6860d4f1787b42379d816e66daa0caba : SUCCESS in 23m 43s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/034992272b134727931185316b94382e : SUCCESS in 22m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5928e2e86e494e5694a0d8ea94f70eaa : SUCCESS in 15m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0f2548663cc4446fbae5cde908b2c8f6 : SUCCESS in 31m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06bc243d2ae945dcbd8140b7ed86312c : SUCCESS in 24m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae4ec9efa1ad4828baad2edc0a2c9031 : SUCCESS in 16m 46s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/be012fb9b6b34b9dab6f0c3e7a093aef : SUCCESS in 16m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ea805eb3e384c498d9d024ac5f63f60 : SUCCESS in 11m 41s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a92a09f56f6b4df99a12be596c51ea2b : SUCCESS in 16m 55s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4b674a838eb342c5bac0da36726ec27f : SUCCESS in 16m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc7b70d52178444e9ae5d359c2251f81 : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"452baf3a1562c7e48219317203d6514a3a704432","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 13:29:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"15a440529d801e063eb34c5860b9207f4afdbcdd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 13:31:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76e1abbcdd842bb8b92ca1fd0bcc077 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7791250b908f4ab95fe4a123eaa4847d928b66b","revisions":{"a7791250b908f4ab95fe4a123eaa4847d928b66b":{"kind":"REWORK","_number":1,"created":"2022-03-31 09:16:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/51/835951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/51/835951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/835951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/835951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/835951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/51/835951/1"}}},"commit":{"parents":[{"commit":"a451ff7af762574870a89ad193063f6358001ed5","subject":"Fix custom config idempotence","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a451ff7af762574870a89ad193063f6358001ed5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-11 16:38:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 09:15:56.000000000","tz":120},"subject":"CI: fix TLS job by freeing up memory","message":"CI: fix TLS job by freeing up memory\n\nWhen TLS is enabled, extra RAM usage is causing the OOM killer to\nterminate Tenks VMs, which are using large RSS amounts (around 1.5 GB).\n\nDisable Heat and Horizon to free up enough memory to make the job pass.\n\nChange-Id: If483a6a6fb6d5b2c9b6b7dbd22939b0b46599538\n(cherry picked from commit b0c1d4c53a4b22711adba81fb1e282075a5f502b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7791250b908f4ab95fe4a123eaa4847d928b66b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7791250b908f4ab95fe4a123eaa4847d928b66b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
