)]}'
{"id":"openstack%2Fkayobe~836040","triplet_id":"openstack%2Fkayobe~master~Ib7495c1bf79de8b1ea67e4a8652345c22e9d1a3e","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-11 10:29:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7495c1bf79de8b1ea67e4a8652345c22e9d1a3e","subject":"[release] Use OpenStack Yoga release","status":"MERGED","created":"2022-03-31 16:21:45.000000000","updated":"2022-04-14 08:09:18.000000000","submitted":"2022-04-11 10:29:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"836040","meta_rev_id":"e2f8dcc9cc4b097246912411c7c350597e52d3a4","_number":836040,"virtual_id_number":836040,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 10:29:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-11 08:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-08 19:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-11 08:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-04-08 19:31:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-31 17:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 19:39:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-04-11 08:49:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"28550ffbf1268423da5930f01941f119ab0d04dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 16:21:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4d50d5325a69b42d0cb491afbb30b7c8bd24506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 17:45:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ae7b613168b4927806cff1513ee893c : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc56870c4a6e4ba69588ba2c94ac3f65 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48877066e68c4a40870f6247e669ebca : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bbfc0e5f4f94084b734d31b44a1af4d : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8402141ac31944b291068ee2f0325ceb : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dc4188f37bf47b8a18a5aac6e2f994c : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9bdb65e70f884747a7f4956082e81a57 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23d79a3b63fb47ee8e42a3ee4ba1cd0e : FAILURE in 7m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/39bababa78b94b24b43e790a028388fe : SUCCESS in 29m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cb4ebaadc12741f29c9893e444e8d0fe : FAILURE in 19m 11s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/16b3fbdacd45467894edfc6c75525693 : FAILURE in 16m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec589d6088e2495bb9d90c4bdd6a833e : FAILURE in 16m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3762e31465a84451bd3c476b33b827ee : FAILURE in 21m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/411ce1a4355a47ccb51b223d31d36b5d : SUCCESS in 24m 27s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/32cb0c26b88d4c8ca83df2c4a46e2d18 : SUCCESS in 23m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70b2a56eb3064c28af5af4e586769a1f : SUCCESS in 15m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4e18cf42e63d4b1b9193af751c539cb3 : FAILURE in 1h 22m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f9b059fdb734efaa765c38f5b3212e7 : FAILURE in 1h 02m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d2572b0d2c1c41ebbd110bb9966a0b46 : FAILURE in 18m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8ecfaa26a55b49f69cd07b2b9f408bae : FAILURE in 19m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc151fefee4c49ce93d082314d292175 : FAILURE in 13m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/591f0ae4ac104dcca98378a926cbf9e8 : FAILURE in 36m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/534c9d0061824ebda1873cd80bafdc93 : FAILURE in 26m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ed73c4eb35a74085b2aa3e70334544fd : SUCCESS in 21m 19s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fe9181664ac04be39f646cc0fe9eb0f4 : SUCCESS in 16m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/244d3cf45aca468485d36b4c75338789 : SUCCESS in 15m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a9d05ec18dad421195637ab59689c244 : SUCCESS in 18m 10s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ff0a56841f2148448b4fe289cf2ef5a6 : SUCCESS in 17m 54s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/000e5067a20e4785b85433ff07faf8cd : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0c8c4f7ed52429eb60abd20d79a54e5a0ba0dc58","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-01 08:45:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d800d81f752116f11446f27d12888760cfbe66f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-06 16:10:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e9a041ff8e1a5c644e7cff01c252ac209dadaf0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 12:50:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"04d274312648902b56f967a83533fbdbca258d2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 12:50:57.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kayobe for 836040,2","accounts_in_message":[],"_revision_number":2},{"id":"1926261fa93bb6bf4d4bef4f86063b2c9aa9ca2b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 12:51:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"89b6af2b8036686e824dc8d5d0ce46aaaead3c16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 14:08:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3a2c6cd2efc40549ebb0215d14f0282 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed44a83077e48cbb0f2066453da9d32 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a16f388652294029b3ae17e1ca09873d : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c57581c04ddf4f31b1eae065d4ae8483 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3af971128444d3ad64c368163a0147 : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5be0aae31fd745c4b1a137d72349f349 : SUCCESS in 5m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7571efdbd1834b8c9420ecd07ee15a02 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07455ba8de50448591a871a845e319d3 : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2906e9704b2147939dbc56377d1b7eba : SUCCESS in 34m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/921020a5f5444895b536b38102e0b68c : FAILURE in 15m 58s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e9f54fc071d54aee878df448b7ba8f9b : FAILURE in 17m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50b187f0abd14c4cbaebfbc16e28d25f : FAILURE in 12m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c71cd0d054904c8ab7af2522c26bc199 : FAILURE in 21m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0a022519db0a426e85efce9b62d8b360 : SUCCESS in 19m 36s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b3c0e7879e80482c953803e832b8066a : SUCCESS in 22m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d43cd8bcca747859375c127bd33f01f : SUCCESS in 13m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4ec94fcc88bd4d0bafec1dc2984ff5f8 : FAILURE in 1h 14m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c035a9a782b453cb822c32e2906e983 : FAILURE in 45m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7e64196f51fe4916b9e450c46a73fd3b : FAILURE in 17m 21s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/caebaa24df784cdfaa48b22998abbc48 : FAILURE in 15m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c6505191c074e7a95ee30d16e727691 : FAILURE in 10m 33s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/36b764dbf0d946dda86aae8c8326c051 : FAILURE in 17m 10s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/825088e2ae714b0d9c5fe79d83a69af2 : FAILURE in 36m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb79bbd6813e4aee8d821798868f2e47 : FAILURE in 24m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/eef19d78acf641ffb8872db38a74991b : SUCCESS in 17m 56s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4bdd5ed0c97d4e6d885614e3aeba7870 : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c627d8f51964b53a3782b043378be34 : SUCCESS in 13m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/584453a1315343c287109b953e07a27b : SUCCESS in 17m 03s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4e76d9cb4e5040b98112b1f9c289581c : SUCCESS in 18m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ade172786a5147808345c7dd12e68c1c : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":3},{"id":"13bab91011d1dcc5b7ae3df75b19069bbc26d9d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 15:36:21.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"677506e6c63765d1ed7e419bc04a66484d5c6338","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 15:36:25.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"e29af38d0b36f2a56c463d317d3a35f54773f816","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-08 15:17:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d14789c66a27fca7739381a11b5041c6080e7e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 17:18:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98123f4f4b1244068b410102483f3e19 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3653b61d164452ac1cf67cd914ee7e : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/306ed8b78f854dbeba1c8e76b1ed47a3 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49011b8209c749c6a77636453bef82a9 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b32e38897a48b1b2c3ec5ead622f8c : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e1db761bf2d4d77a50a5a9493c1171d : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af56dce16597475db2768897fc1b583d : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdd7c15a24484891812ec9573d387a38 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/68c6c24ccdb8494f8c73b216b3e4de7e : SUCCESS in 25m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/192b1d504a8b473a9f6d7e235829bc8a : SUCCESS in 1h 11m 39s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a58b7d6dd016440596e114c810eaedcf : SUCCESS in 1h 13m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fc1299589ae439c82dea40bde9b12ac : SUCCESS in 53m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cc9ff7a2973c4b96aa9b7c48699e02e4 : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c42050da53664ce298974f33785a54a7 : SUCCESS in 24m 18s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/bb18882cf48d4192aeaa8c4adfb0ec4d : SUCCESS in 18m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0c99ad4d337426cafab0a54290df8c6 : SUCCESS in 14m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/abd591cb29df4482aa26c27676dacb07 : SUCCESS in 1h 54m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd9cd11cd63147e7b48268b11d38b24e : SUCCESS in 1h 11m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/196d03810934410b99d46c935275b7c1 : SUCCESS in 19m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/eed886e059e7491184266f14f62788d6 : SUCCESS in 21m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12db14a96b414c7c91f1d797c00c1e0b : SUCCESS in 16m 28s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/cb1ebcf5a39f481fb18d934859cf4520 : SUCCESS in 46m 38s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2c7becefc7724c2287e620f671bf6498 : SUCCESS in 41m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44ef27d5fba04e68973d26cbf49ecd8b : SUCCESS in 28m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/32120915a9e34ac3b7f661dd22e768bd : SUCCESS in 15m 28s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ea774e90114045f6bbe717692642b020 : SUCCESS in 15m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe868ea84d154fd6ad987fdbdebd4e1e : SUCCESS in 12m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a7cb53f4b7274997a2a7ad0534e8c5ad : SUCCESS in 17m 50s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4b0e5d897c93422d89a67f901c4e9cf7 : SUCCESS in 17m 54s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22986df8073748d0afac55bb21c7cfa3 : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":3},{"id":"b239fa9fc1fb25ca61c6b2f5e94ae0fa51b6c110","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-08 19:30:30.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"b1606c1e7e88d64681f8f8354ceeedf9033a1312","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-08 19:31:47.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"e6516039c4f1667a5b913f4faa25f710123575e4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-08 19:39:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"74c8c250707bd35ec2d16316ebc04f60854e9e30","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-11 08:49:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a0f62ad2a9167101cc041be4100c64b567108e9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 08:49:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6abd952d87a9c07af4fcd75b96007aa39d01e8e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:29:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5e7d3efb2e24f0b8297533749f05f4b : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c303bb3712d4b30a01e55ad3eba3d13 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c209804643142f5b6dce74fb5bba934 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd62fea29004da1bbbc719742034e84 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7339c51db20348b997f59d0a93103520 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4fb08195b9774265a889c70460605c0f : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f2e920cf80341a68a2aade0630b73b0 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6874445bb7e34fd58be1aad3f65e152b : SUCCESS in 32m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f7d216e45b8d44eaa4bd40478655fc32 : SUCCESS in 57m 49s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f07b1c35175c4e82bb73de9b0cf81936 : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b228f8107f2442880e6211f00f12f68 : SUCCESS in 58m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dd96b692b1344058b8d4a2ac4d052ba0 : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3671096e61e34503975ca24a4a685787 : SUCCESS in 21m 39s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4a2000502fd845499970a2f148e8b1dc : SUCCESS in 21m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f07c4b89c0a049b6a9a84ee67e9e39c1 : SUCCESS in 12m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e60dc31edc7745f7b2ad926698a00077 : SUCCESS in 1h 38m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1678d039655549eba282b7c16207ad8e : SUCCESS in 1h 15m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9539e8dbd0ed47b7bcb7ca5a6daac72b : SUCCESS in 16m 38s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b9cb66e72aac43b3841eccd531867ffc : SUCCESS in 20m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7601d23e5c1440b7878434ddad5e7141 : SUCCESS in 16m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7babd4fde027462092aae395e9becfd2 : SUCCESS in 30m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78727afc182d440a9a71b18ff686cd63 : SUCCESS in 25m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/549abfc36f7f4a25b75449da053c844a : SUCCESS in 17m 15s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/56ac3f2ab8cd48c5a1aaaf5cae05bd0a : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b38f92d2ac2e4d61a1609a29542881a7 : SUCCESS in 10m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/073bb76b5cee4ac2a1978dd26e0f9d60 : SUCCESS in 15m 53s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/acd037c3ff114b2e80ce7f99258b7614 : SUCCESS in 20m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/702ae0e8b27343629e4c410d136a25fa : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":3},{"id":"aff634f4288075b9dbac2c5e0305ac2a90d10cfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:29:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a9e18536e855d881228b206a466da9b0931174cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:32:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0008eb447ead4e68ae21adb6ae7ef40e : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f82ac0703f844ed8862badbc1dad85b : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e2f8dcc9cc4b097246912411c7c350597e52d3a4","tag":"autogenerated:gerrit:revert","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-14 08:09:18.000000000","message":"Created a revert of this change as I9a18c41dbdc369a7a719632d6e41e2373784f29e","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"88dd02dc92ca4b544aec18bfbb1d1d93954c55fe","revisions":{"25a16600800c0925ac4b4b8d910902008dbd2a04":{"kind":"REWORK","_number":1,"created":"2022-03-31 16:21:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/836040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/836040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/836040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/836040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/836040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/836040/1"}}},"commit":{"parents":[{"commit":"d18321e03de78ebbe7cf1d87c1d6f665512ad4b1","subject":"[release] Sync Kolla Ansible inventory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d18321e03de78ebbe7cf1d87c1d6f665512ad4b1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 16:20:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 16:20:51.000000000","tz":60},"subject":"[release] Use OpenStack Yoga release","message":"[release] Use OpenStack Yoga release\n\nSwitch all dependencies to use the OpenStack Yoga release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/yoga branch has been cut and RC1 released.\n\nChange-Id: Ib7495c1bf79de8b1ea67e4a8652345c22e9d1a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25a16600800c0925ac4b4b8d910902008dbd2a04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25a16600800c0925ac4b4b8d910902008dbd2a04"}]},"branch":"refs/heads/master"},"220bbd6fce9193664e67647901e99d4325ac3f91":{"kind":"REWORK","_number":2,"created":"2022-04-07 12:50:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/836040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/836040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/836040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/836040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/836040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/836040/2"}}},"commit":{"parents":[{"commit":"d18321e03de78ebbe7cf1d87c1d6f665512ad4b1","subject":"[release] Sync Kolla Ansible inventory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d18321e03de78ebbe7cf1d87c1d6f665512ad4b1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 16:20:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-07 12:49:53.000000000","tz":0},"subject":"[release] Use OpenStack Yoga release","message":"[release] Use OpenStack Yoga release\n\nSwitch all dependencies to use the OpenStack Yoga release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/yoga branch has been cut and RC1 released.\n\nChange-Id: Ib7495c1bf79de8b1ea67e4a8652345c22e9d1a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/220bbd6fce9193664e67647901e99d4325ac3f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/220bbd6fce9193664e67647901e99d4325ac3f91"}]},"branch":"refs/heads/master"},"88dd02dc92ca4b544aec18bfbb1d1d93954c55fe":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-07 12:51:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/836040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/836040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/836040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/836040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/836040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/836040/3"}}},"commit":{"parents":[{"commit":"7b8c2fa6a634e2ac61c3e27e56736e82a1a84366","subject":"Merge \"Fix Ansible inventory generation when reusing group names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b8c2fa6a634e2ac61c3e27e56736e82a1a84366"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 16:20:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-07 12:51:18.000000000","tz":60},"subject":"[release] Use OpenStack Yoga release","message":"[release] Use OpenStack Yoga release\n\nSwitch all dependencies to use the OpenStack Yoga release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/yoga branch has been cut and RC1 released.\n\nChange-Id: Ib7495c1bf79de8b1ea67e4a8652345c22e9d1a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88dd02dc92ca4b544aec18bfbb1d1d93954c55fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88dd02dc92ca4b544aec18bfbb1d1d93954c55fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
