)]}'
{"id":"openstack%2Fkayobe~836778","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Idecda342f3ab86733e8d59061458d44af834dbb0","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-12 16:42:55.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-12 14:00:23.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Idecda342f3ab86733e8d59061458d44af834dbb0","subject":"CI: separate image builds into a non-voting job","status":"MERGED","created":"2022-04-06 15:46:12.000000000","updated":"2022-04-12 16:46:41.000000000","submitted":"2022-04-12 16:42:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836778","cherry_pick_of_change":835610,"cherry_pick_of_patch_set":10,"meta_rev_id":"bd96fe383587f7fa56d680b918a6fe462118cf00","_number":836778,"virtual_id_number":836778,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 16:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-12 14:47:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-12 14:00:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-04-12 14:47:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 17:37:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-07 08:05:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-04-07 12:27:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"cae4bfd3abdfe2388718e4823ac6359b8abf8227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-06 15:46:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"efdf51f3e25b248f5068d9633ded4949381953d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 17:37:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca1802d71a8a44f787b1c7870a1cf93b : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd875ca616f46649a5a7037853c2c50 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6c8a26c6b1e441e99d2be943e75a704 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80398e210a0843c98969ceef1dcdaacb : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e48db0a24258411aba5f33490fcf9253 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9d66e2a64940ce8f8ce592dbd86754 : SUCCESS in 8m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/853be8f1cae844aa95c2376987d7cbee : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c03900aaf27845cbba693ccf31a8ef1e : SUCCESS in 15m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba573f216fc1478c990985a8790e3f8d : FAILURE in 10m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1cee6271fe2f4c4ca7a15a371ef922c8 : SUCCESS in 1h 00m 28s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/318c97fd7ca24045a2962e11bad6d8cf : SUCCESS in 1h 15m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73bf028761404027868f5e2fb89d8562 : SUCCESS in 44m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a94e7bbd320e45adad47d2aa7df9c0e8 : SUCCESS in 58m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3b5c3a7319704b8cae102e839dd71092 : SUCCESS in 22m 35s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/91373ce420484965b5c47f8ddfc7b862 : SUCCESS in 24m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5a108c7015d49278a37309c68679cf3 : SUCCESS in 16m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/40b46fbbae874663bf4e8f4ff2ee3200 : SUCCESS in 1h 47m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7eb770e10cf4742920dc2402f711d6d : SUCCESS in 1h 24m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7572de2991124a6fb083ea96f06c4f77 : SUCCESS in 16m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1e642387f0434b5b80a9eadb26996adc : SUCCESS in 22m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bde1fa87a55a42849fc1e5d2deced500 : SUCCESS in 12m 41s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/3f20deec47d2420da0f11f35343dfdc4 : FAILURE in 24m 08s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0374ab58822f4cf38e0e50eb515d94f8 : SUCCESS in 26m 16s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83b62e16c91249aa8d986bf2cf392f2a : SUCCESS in 25m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4ff862fcd2c24484ba1b0ad963341b62 : SUCCESS in 13m 55s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/39abec4509094d3cbdf6655cf82264dd : SUCCESS in 15m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f3ce4aca6924f2cb10f9cab498fb689 : SUCCESS in 12m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/389ae4fb07d94e86aa7cb3eeec4fdacc : SUCCESS in 19m 46s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/996e062e09eb4a359ca64607fb049b2f : SUCCESS in 18m 16s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65c90d2db8a148a79ab5e4e612f25fcf : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":1},{"id":"556d2ad537b44ea3db68b6c1fb17d7fc2a049cc0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-07 08:05:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b56144316e150e63ea6b33d7e126c527ca275a5e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 08:40:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da361881e742e05bb7322ce75326c4c153c340a1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-07 09:40:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d97429a4a109ed835a15b19357d31518c26b48da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 10:00:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77464d78b90f4332912a416e714bb394 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9410ae807c79496eadcc428af4226167 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32aa71ed12eb4343b0138739e4398dde : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbb1eaed26c647f7881abb86970fc021 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bf5851156274a869ad9eb1e66e4fc04 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a184688efb1246efbd966cb8cbb9e4bf : SUCCESS in 7m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bf64a461bb9543089e328e4dde7ce018 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5063c9a56b3b41d3801c5b60ea778140 : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c520236ce8d04bda97d3a654eecf9d20 : SUCCESS in 33m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2ad7cbff8ecc45269f179fd65044781c : SUCCESS in 1h 11m 23s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/2d668316d5ab449f9f6d83cb96c4d162 : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/414adc26e9154348a76e21b11ded5969 : SUCCESS in 48m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a320fe3bb3940eb9dad26cff506c1e3 : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3da4f7c90b274ce5905ec9eda7c8ec5c : SUCCESS in 21m 23s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a53bcb19f0f94321830e2058e1f18e03 : SUCCESS in 23m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c0fd74dd49d4183a01f70f20c3a8018 : SUCCESS in 16m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c88fca5255474b2aaea308b59ec4949a : SUCCESS in 1h 41m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58a6cd7966f94b4c87815ddb58b683ca : SUCCESS in 1h 12m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ad34ebc991ba4e3eb6b429db821440d5 : SUCCESS in 21m 43s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/37790b66d80f40718131d1b8511cc1da : SUCCESS in 23m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35166d2659e94b58993c1e849d7fb557 : SUCCESS in 12m 43s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/f31f0ecb61d241c5bae177ac6af3396f : SUCCESS in 56m 59s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c93e0c15fc2a4150abb5f3e95c0dcb20 : SUCCESS in 31m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/374b533474c14ff19abc44adf3095384 : SUCCESS in 19m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/24314996625a426f9fb5abcd6502ea03 : SUCCESS in 16m 39s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/929ce1d422524c889a77bc0ca582796a : SUCCESS in 17m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b27a71dad0ad409c96cc7da3855456a8 : SUCCESS in 12m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/82bd6e70cc19448297ba79ec769bb7cc : SUCCESS in 19m 06s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/79c94d8950414829ad71ad0fc76fb6f7 : SUCCESS in 17m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4749af5ab4c247449dc8520481edd210 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"bc0b64de13d97ea8287aae0e8b5cb9acd9987bf6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 10:31:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4b5dc3503c04c295d2bf2b0cb36772f1f3a70af","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-07 12:27:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"164a816bb44e16f1ff40d69941b2426e415f7f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 12:28:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20e495dfa55c85cd081f9c34b37295987d52d982","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-07 13:04:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b85bd28061b9b7431397a97d58ce11c4cf3e9c0f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 13:45:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/3f90f67636a747878b3644b09ceea4ba : FAILURE in 21m 39s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e58f5ab3a7f4ea699db9e328475cc6d : SUCCESS in 36m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d823a6cfa9431aa6c22667244269e3bd59ae2428","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-07 13:53:53.000000000","message":"Patch Set 1: Code-Review-1 -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffcb9840aecac72a9c3952a0942cee61700e5b5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 13:54:15.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing requirement.","accounts_in_message":[],"_revision_number":1},{"id":"e7862e6455f7f31b0defb4341a3fe09e1c3c9707","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 13:55:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91a03b4370ff455e277c8ee67a106e4194955cf1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 13:55:47.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"5e23ca8b4b1d1cf73baeefb57cf51ecbea19fa20","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-07 15:58:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afa47ef197241bc28629dd0e8a40fee4ee7eabf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 16:05:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"689d832a5d4830b267f8593358b89d3c67d45ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 16:08:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3d6a4f45f209f380eb94585e069dae4c86fe73d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 16:08:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"49ad5ac6fab19539c1c2735348c587274c0c8fab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 16:09:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1a8bd0a0352e675fe6dd4de35e079c1c928117f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 18:04:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f77fba09776412fa5f1e1b078aff241 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189161221f56459ea966b92722d79bde : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0392c3f0e7bf41deadcf9830e5af0f20 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e3b60fd3144179a628e935bac55d8f : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc3a56ec48db4f8fa67f87eb428064f7 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6e9c920ce74a7bb79b87a5a9a928d4 : SUCCESS in 11m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3922194bbe344962a695690ad6ec864c : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b71c21ac1fd64a6ba2a3e5d5c76b5f2f : SUCCESS in 13m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f7d67a1dd9e4bfe8a6b86273903c75d : SUCCESS in 45m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/04f2001ebd01480ca0323462e6f92ebd : SUCCESS in 1h 04m 42s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/9a55491398434a199ec3ec53f39c1620 : SUCCESS in 1h 08m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8ccd46156d84146bf169346fb981809 : SUCCESS in 1h 00m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b8ff211f56304c87a9cc46dd01e6872f : SUCCESS in 56m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2684cc51183348baa30288489c9e9af4 : SUCCESS in 24m 33s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/07221ad3b7d94d909713a3cc6343ed87 : SUCCESS in 25m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5b34a52ba1d4e028097cf4f1f160492 : SUCCESS in 11m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/71da088406f1490da472de7b8f15d57d : SUCCESS in 1h 53m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49d7d1d8b4374008b2ae939ec4f966cb : SUCCESS in 1h 20m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/95c0c216e0334a6bbe70970b65527f3e : RETRY_LIMIT in 5m 25s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d3924e4e8ae5471896d18658afdc7d01 : RETRY_LIMIT in 4m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc99d4dcc2a34da6914105e80636a1e2 : RETRY_LIMIT in 4m 38s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/79623ed0015443529cdc24fc57747464 : RETRY_LIMIT in 4m 57s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f3995002a3cb4cf797d574c0396f788a : SUCCESS in 29m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b77be0034764d0b9e919c775941b20a : SUCCESS in 24m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/82b93a5f394643fc91f29d1b4eb1a222 : SUCCESS in 24m 17s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8f428255c95e4277aba60816b847ea68 : SUCCESS in 19m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7156bd92716545298115c1d79b223f32 : SUCCESS in 11m 52s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/20f93b515b3743439bd025414869ff71 : SUCCESS in 17m 53s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/796b2f0cec4d447a8127ec6f236d4a70 : SUCCESS in 16m 42s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a06947dbac44648b9f3218921ac6f1f : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":4},{"id":"e3f4988360e90379986e1806d90a3852f26ee3e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 10:05:35.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d18c0a287149dde6db1759d5e75b96ab9d832ffd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 10:05:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cfe7c467e31eb3f4cf4fe3bcc15770d4bd8f425a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:47:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/07db7df3f31c48868147b94c76b69f6a : SUCCESS in 36m 00s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b12fdf21ef84fffac2d554e1a21a7ca : FAILURE in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a3cf9af557226ce6e3ff7fce3c63e7b15ea995f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 12:02:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe80413a4ceb448a8e51d5a8f8b71f26 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dc7974f92d64ea5b549b338107d317b : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba9b60d5c5b049dd868b88e698022c55 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf6b70f4f08a47f5aa4aace4b4e72342 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0436ee310f754310be6b59270bc4443e : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aca95f8b00164e1789c1341a9b26ca8d : SUCCESS in 8m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae9b8e80545d42a9be8a82e742165dc8 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ec2aed395c048ebb2b5dfe325f05329 : SUCCESS in 11m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/de4cd79107a14048b7c777f2228bd529 : SUCCESS in 19m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/aef60cbcfa714f248b85c4bbbd1afae4 : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a70c010155ef48b384f5dab8a1ded951 : SUCCESS in 1h 11m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/530b4fb22a074dbc8056ea5a3c53f0e5 : SUCCESS in 58m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/00379635c5e647c8a8c63c959bb9a8e5 : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/644abae7cf5248caaa02ee670984f0b3 : SUCCESS in 23m 16s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9ed15c678a034a458d0600c7131460e4 : SUCCESS in 22m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99282d029c384f709720f57bb412c74d : SUCCESS in 16m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5d5b6fa996144bc5931b68af6dec51b9 : SUCCESS in 1h 53m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2a9a5e33ad042c5b6ea0064930952b7 : SUCCESS in 1h 14m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cbf16a24e5564e73906c51cbf53d8395 : SUCCESS in 19m 53s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d7fefae10d67420dbaf0a060ba9f6a15 : SUCCESS in 19m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a4300deb6fc4ed88ccd688b5f22254d : SUCCESS in 18m 36s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/27ecdf8abb5b4afbbc7a400b95b66d50 : SUCCESS in 47m 49s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a9cae65c41324a8193e480b87e928803 : SUCCESS in 27m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d5106defcbb41269b0f04491a5a6a33 : SUCCESS in 38m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/144e044e4d9844deb238733de7b707bf : SUCCESS in 18m 39s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ade5c6ec114441eaa9680b9b465e0504 : SUCCESS in 14m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0948560e44764af4a96adfd1c291c35d : SUCCESS in 15m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f827b968eff542e0aa2c25fb8edbf05c : SUCCESS in 15m 20s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f94316486f124c158dc497d91c3b1b7f : SUCCESS in 16m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc7e1f2ddbc24b6a9ffdb3333d4f0f3a : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":5},{"id":"8319341801edf4086fa2ea6c72a404a5b8328e3b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 12:17:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"17e2f5bfb7ccf1418a9ee9a812483b81afcaedfc","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:03:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/f0bf9f0a7414467ab898873dc4702040 : SUCCESS in 36m 25s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00e2e8fa2bdd4bd1b9251898d5a76752 : SUCCESS in 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e8714165661ebada0b4a693f14a3aa658159b173","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-12 13:37:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"25d58ee8878a84f65af3b354ac9e452d58dec1e2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 13:44:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b85e704c6f34607b620c0db2a3552220ce7930d3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-12 14:00:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a5eb700eb667e27950edb542c2d81813a3e0c890","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 14:47:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5d80605a85314da85da540f95af8e2fbc1d246b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 14:49:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bb889c23b694a6f4884172242867ed4c036cc03e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-12 15:00:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"41c497dda4d7580fc8c47c42c8e1c2bf5e23536b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 16:42:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee99f52149aa47ae9218ce58dc40bfc5 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93120e0eccdb4b9380b5221a4abc3eb9 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eae6a771621c48c9a522e0b55d165cd0 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c9699e6458545f09248c5e961da7651 : SUCCESS in 10m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dfd10e07f31e43a3b96a0ff1bd28054a : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b1a865807baa4acfab7ae60b91f90e9e : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/001ed6ec8517499498c8fdb0a81ad166 : SUCCESS in 19m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d00bb3aa1af446a4838a2c504edbbe25 : SUCCESS in 1h 07m 22s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d3fdcce7fad24013b80d06a8ef67d5f9 : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd85e389d9a24e559d7a4dc03679ad6d : SUCCESS in 55m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/765d540170a841baad832750f1bcf8ee : SUCCESS in 1h 11m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/497a533d91044ed79898229483ed5f35 : SUCCESS in 25m 37s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/6d25c1b3f082464987f014a9b212d1ea : SUCCESS in 23m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b565bbd2fa944b4490ebd231c060c3f2 : SUCCESS in 17m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/16355786907748cbbc63d35b1fa4121b : SUCCESS in 1h 48m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd55246f710546bba78c32fb921834a3 : SUCCESS in 1h 23m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3c816f9d9855477795505f31308b1efe : SUCCESS in 20m 11s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/7c8daa380b3b44499bf6929c0fcefd1f : SUCCESS in 20m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e25c160ce43c4865861107f4fef3ebae : SUCCESS in 16m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bea34f112785474cafe6aab6f8ec4972 : SUCCESS in 25m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ad49a18c9ac413cbaaa44ba71f0beee : SUCCESS in 23m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/294c1df36781427a90bcf5e6c138d48c : SUCCESS in 19m 06s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/148e20bad2be460e800129e6e848cb41 : SUCCESS in 17m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56cfca2396b64f888dc953f6e52f8a20 : SUCCESS in 15m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/987bd790bae44a0d9ab3eba40f3b2268 : SUCCESS in 19m 05s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/df77e076308f424db29a73b838fcbd9d : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abbf9f54283e41c59b6ba7df8f03b197 : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":5},{"id":"370e6bb1d92160154f0badafc36c0eb058fe42e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 16:42:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"bd96fe383587f7fa56d680b918a6fe462118cf00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 16:46:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4d38ede394749b79d6bf3d76acddf27 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"65ad855e79be0696943e1780bbc13142261d82f9","revisions":{"487a77ca958a09f9daac6fa924d057ba1e09b6a1":{"kind":"REWORK","_number":1,"created":"2022-04-06 15:46:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/836778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/836778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/836778/1"}}},"commit":{"parents":[{"commit":"8251b21237ca26edb63ac9b933b79c859cfa1a9a","subject":"Merge \"Add the bonding 802.3ad aggregation selection option\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8251b21237ca26edb63ac9b933b79c859cfa1a9a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-06 15:46:12.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n(cherry picked from commit a7ee3ac5c8739d7547ae430574a5ac8abc79c41d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/487a77ca958a09f9daac6fa924d057ba1e09b6a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/487a77ca958a09f9daac6fa924d057ba1e09b6a1"}]},"branch":"refs/heads/stable/xena"},"e07ccd8fb8fa3907ccb9c62afded512844bc54c6":{"kind":"REWORK","_number":2,"created":"2022-04-07 16:05:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/836778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/836778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/836778/2"}}},"commit":{"parents":[{"commit":"8251b21237ca26edb63ac9b933b79c859cfa1a9a","subject":"Merge \"Add the bonding 802.3ad aggregation selection option\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8251b21237ca26edb63ac9b933b79c859cfa1a9a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-07 16:05:01.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n(cherry picked from commit a7ee3ac5c8739d7547ae430574a5ac8abc79c41d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e07ccd8fb8fa3907ccb9c62afded512844bc54c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e07ccd8fb8fa3907ccb9c62afded512844bc54c6"}]},"branch":"refs/heads/stable/xena"},"e9d6a072d71f98f400cf320acfd520032ea95eb1":{"kind":"REWORK","_number":3,"created":"2022-04-07 16:08:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/836778/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/836778/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/836778/3"}}},"commit":{"parents":[{"commit":"8251b21237ca26edb63ac9b933b79c859cfa1a9a","subject":"Merge \"Add the bonding 802.3ad aggregation selection option\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8251b21237ca26edb63ac9b933b79c859cfa1a9a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-07 16:08:06.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n(cherry picked from commit a7ee3ac5c8739d7547ae430574a5ac8abc79c41d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9d6a072d71f98f400cf320acfd520032ea95eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9d6a072d71f98f400cf320acfd520032ea95eb1"}]},"branch":"refs/heads/stable/xena"},"b83348b5f4c315779b8b2cbcac78d76071c30681":{"kind":"REWORK","_number":4,"created":"2022-04-07 16:08:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/836778/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/836778/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/836778/4"}}},"commit":{"parents":[{"commit":"8251b21237ca26edb63ac9b933b79c859cfa1a9a","subject":"Merge \"Add the bonding 802.3ad aggregation selection option\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8251b21237ca26edb63ac9b933b79c859cfa1a9a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-07 16:08:54.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n(cherry picked from commit a7ee3ac5c8739d7547ae430574a5ac8abc79c41d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b83348b5f4c315779b8b2cbcac78d76071c30681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b83348b5f4c315779b8b2cbcac78d76071c30681"}]},"branch":"refs/heads/stable/xena"},"65ad855e79be0696943e1780bbc13142261d82f9":{"kind":"REWORK","_number":5,"created":"2022-04-12 10:05:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/836778/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/836778/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/836778/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/836778/5"}}},"commit":{"parents":[{"commit":"8251b21237ca26edb63ac9b933b79c859cfa1a9a","subject":"Merge \"Add the bonding 802.3ad aggregation selection option\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8251b21237ca26edb63ac9b933b79c859cfa1a9a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-12 10:05:13.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n(cherry picked from commit a7ee3ac5c8739d7547ae430574a5ac8abc79c41d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65ad855e79be0696943e1780bbc13142261d82f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65ad855e79be0696943e1780bbc13142261d82f9"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
