)]}'
{"id":"openstack%2Fkayobe~837033","triplet_id":"openstack%2Fkayobe~master~I955a2c12b5143a63422cab50c74191bc22c63932","project":"openstack/kayobe","branch":"master","topic":"forgotten-linuxbridge-agent","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-04-08 10:30:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I955a2c12b5143a63422cab50c74191bc22c63932","subject":"Restore forgotten linuxbridge-agent container","status":"MERGED","created":"2022-04-07 21:27:32.000000000","updated":"2022-04-09 19:07:01.000000000","submitted":"2022-04-08 10:30:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837033-forgotten-linuxbridge-agent","meta_rev_id":"0208858042b225797745ba747bb609988f853340","_number":837033,"virtual_id_number":837033,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 10:30:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-08 07:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-08 08:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-04-08 08:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-07 23:18:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 07:29:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-08 08:41:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"decb8c832b2f6b0369ef1d79cb074afffc4457b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-07 21:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c13771ad1c7713c17b63072ebeda34581ff0d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 23:18:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10855824a1bc471faf43fc04d4f39eb2 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1ee7a4510d4f61abbc8bcd6861f0f9 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c36379b41ed48d1804347559c55a2fe : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb442be29cc54a20bcb010ac00ac1438 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3f63422ae443b49018d9da15524a8d : SUCCESS in 9m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc624d720899456399fa2f7af613144e : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d1b48dc3ccf94f7c8fe801d3fab4d12c : SUCCESS in 12m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71aec28f384f4b6db727e0638e6c2e32 : SUCCESS in 23m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0ff672bd5e3f4d1da41b99e58283b73a : SUCCESS in 54m 17s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/861dfaa1f3d14a78a44e798d4bdd6172 : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20cae7f6e3fa4f0c990ef9347fabd176 : SUCCESS in 45m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f9238355f7ef4737adcccd80ebadbbf5 : SUCCESS in 1h 09m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/16089866bc0f42cbb0d69ace993e437b : SUCCESS in 19m 26s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b078fc16bba8421a8abb4862bd53b0b5 : SUCCESS in 20m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a87a60cd73f45c9939c7fed9b5a519f : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2a62231be06f4c369c485207bc82c55b : SUCCESS in 1h 46m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e6ff4e2ca6145008ed283b2d02539ba : SUCCESS in 1h 14m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5f6e48827040486291123b8c65f92e2c : SUCCESS in 17m 36s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c76ebbb93a1d45b7a9cc080f76528763 : SUCCESS in 21m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be6eb6cd6c214d29801f626aa2b52358 : SUCCESS in 13m 41s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/8b33571784534937acbe0e2f4110622f : SUCCESS in 39m 38s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8ff604025b1749998e46fdf578a0fdbf : SUCCESS in 33m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/034cdc18787e4edc93290838eb6d6f55 : SUCCESS in 32m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f827d15fb63540d4acec7b7d1df8a882 : SUCCESS in 13m 34s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cb03f24a37ee49008531bc99c8e78654 : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d3fbc9c464f447b96f9f1a4dccf49b6 : SUCCESS in 12m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b2583261e4884fd59e6df8bd0b07ac93 : SUCCESS in 15m 50s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8ad89810ec574b818020fd2b786a8383 : SUCCESS in 17m 03s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14877889e87441c7808db96719c30ddd : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c805c13a8f32bff38d7a88d130667d7c3ebeeb04","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-08 07:29:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f7ceb73bff5c1b2f7568bf7f3492449b29ceeaa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-08 08:41:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6952c2de46bfe05cc44dbf45e4c542241639c188","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 08:42:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18b8ba4f7619861aa350421f74d29793e87f2400","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 10:30:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a6d0a2697549089435357bfc96dbd1 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9396e0f97c064408ae2a5f18ac4d3831 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2665304eb0648848da1d89438caf0ab : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef418d954bc42429abe5d4bf16cb7ee : SUCCESS in 9m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/155963f4f4294b089d85ad0ce5bc7e47 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/18797939c9eb469c836c3afcb0626db9 : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3cf00602584540ecb2c216c77b349241 : SUCCESS in 32m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/14ed0a2deaff45d28b4d113f1c4998d5 : SUCCESS in 1h 08m 47s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4c5ff640b51642c0926d3ff73e496bf8 : SUCCESS in 1h 03m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14a22d019e484413a1c26bd1976ba1ef : SUCCESS in 52m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/344ece8d4a094d8f9db07e755304b8d9 : SUCCESS in 1h 10m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1b444b0f683f4944baba3b3bb0b7b533 : SUCCESS in 24m 22s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f2438dd20827491b939dd645397b3479 : SUCCESS in 20m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76a7e67c08a946b6a3dfc335ccaf44da : SUCCESS in 14m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5126795fa3e54d37b1c06e2d4ba87db5 : SUCCESS in 1h 46m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43d779a9fad9442199bce07c0086a5b9 : SUCCESS in 1h 06m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6cb2feab4a8c49e5aa31ee3b14ed340a : SUCCESS in 20m 16s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6d995f0eaff3490c9a40ff632e50f038 : SUCCESS in 18m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de299a99dbc44df69d189be2ae5aa643 : SUCCESS in 19m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/05bc6afa64b545e488d96bc450a531c0 : SUCCESS in 41m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65c78d1adde648bc87d35fec18eb9533 : SUCCESS in 29m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1e00824d123c428eacf30497d7530472 : SUCCESS in 24m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/db0e89f15e7e423b81c607b17073744f : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/980bbc7e70794f298659959e990fd3a4 : SUCCESS in 12m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bc5eca4cb5aa48e6a91d2843e5a1c628 : SUCCESS in 17m 11s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2bc4ccada69d42658375a661762d1492 : SUCCESS in 15m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4a320f279c046feb04ab8e07413f0f4 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bc4380abd4d43025e5393a3d1cac517360869341","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 10:30:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bee6bcb0cb8c29b12c9a3db324bcdff42d105e63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 10:32:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c15304de5743f2817c30a14cc63c35 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0208858042b225797745ba747bb609988f853340","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-09 19:07:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8860c4e8826689909088f95ff15f1cb2813647f","revisions":{"a8860c4e8826689909088f95ff15f1cb2813647f":{"kind":"REWORK","_number":1,"created":"2022-04-07 21:27:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/837033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/837033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/837033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/837033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/837033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/837033/1"}}},"commit":{"parents":[{"commit":"7b8c2fa6a634e2ac61c3e27e56736e82a1a84366","subject":"Merge \"Fix Ansible inventory generation when reusing group names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b8c2fa6a634e2ac61c3e27e56736e82a1a84366"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-07 21:16:49.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-07 21:27:05.000000000","tz":180},"subject":"Restore forgotten linuxbridge-agent container","message":"Restore forgotten linuxbridge-agent container\n\nThere is no mechanism to check which agent would be used and build\nall the possible agents, but forgot the linuxbridge one.\n\nChange-Id: I955a2c12b5143a63422cab50c74191bc22c63932\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8860c4e8826689909088f95ff15f1cb2813647f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8860c4e8826689909088f95ff15f1cb2813647f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
