)]}'
{"id":"openstack%2Fkayobe~837056","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I88bdf1e3d0c08271ac8938ae2f9ac3f9fee1efa5","project":"openstack/kayobe","branch":"stable/xena","topic":"story/2009927-stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-08 17:23:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88bdf1e3d0c08271ac8938ae2f9ac3f9fee1efa5","subject":"Fix Ansible inventory generation when reusing group names","status":"MERGED","created":"2022-04-08 08:37:52.000000000","updated":"2022-04-08 17:25:36.000000000","submitted":"2022-04-08 17:23:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837056-story/2009927-stable/xena","cherry_pick_of_change":834188,"cherry_pick_of_patch_set":2,"meta_rev_id":"b033595eeddafa10e2e84a048453575440c472ef","_number":837056,"virtual_id_number":837056,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 17:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-08 15:12:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-08 15:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-08 15:21:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 08:37:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-04-08 10:30:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 15:21:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"a93007009ac8d597609c90c49afb483a6b73249c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-08 08:37:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"df738d5ee4e235e801f244055df9bc3caa6fa24d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 10:30:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52b804272d6c4074a2c9caee2f053605 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c3160f2d2e64f29ab8c44d451a37839 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa5d766541474687bd6fc8786a51bb97 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9981cea20a646e6b7387031ec68028a : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41ef83e0973949088917a482b97cd834 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19639a6f6926489a995d0abf3dc42587 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/419dc4f8132b4f1881fc806290669022 : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75cc27c7c37d4cacb3a2d3d7cb3d8086 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/604baf6987f44fdcba44c1f51ab6701f : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c74eea76ea4441f18a0eff87f08b64f9 : SUCCESS in 35m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a57b08d6d850470b8a27b513f4d2a978 : SUCCESS in 1h 08m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/773c78a5d7a7473ba8e0d923e7d7d301 : SUCCESS in 1h 19m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee9ce67530cb4f098146ea62bffaf692 : SUCCESS in 54m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d15041e39a5843fa80ee0e3de40fa742 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/150fb7eb6f1a4251b02f59370e66fe6b : SUCCESS in 26m 11s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/d8dcc0f35b454c5da4da6dd6d8660942 : SUCCESS in 23m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/256f8b579aa145a79e0ce29eb26e67fe : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/95b7596cadc94bbfbcc533ad8951f359 : SUCCESS in 1h 48m 14s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5781d83b0f334d8e9d08e8a4743d21aa : SUCCESS in 1h 26m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/153091d5316b4a6f9f57c409fd9068c0 : SUCCESS in 23m 14s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2ad8172ec7e748e19d8fa4a2eeb4097d : SUCCESS in 19m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac2d1c70e12148b18436a001e1b75d30 : SUCCESS in 15m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/90d5e7bc7b08491899e1b073f30367f3 : SUCCESS in 35m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ba54835026e414f8a9488c0cc32649e : SUCCESS in 22m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0dc42205ac904549b42449738762d2d3 : SUCCESS in 18m 00s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0152e73812dc48b09805d5bfe1dc412b : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6d401f63d0b4c48a16c43f28a369898 : SUCCESS in 11m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/52b219f87ccc43259abc551a2f66a7e7 : SUCCESS in 17m 54s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/77535dba5ff042448ee1dbf0d7805ad5 : SUCCESS in 17m 11s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d964dbfca18418a8f9bfba40d5fb5f7 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6aa52a2fdd424515eb5a2cff80296c45f1533776","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-08 15:12:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c031c9739ca62450d191ce95b633be2026f1c06f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-08 15:21:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37ecaaea3f5b7a3829401b5d518359407ee646ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 15:22:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99dbe23c2c92d245083fb540ba1781c5e6c38d24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 17:23:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2567f3b2b57a466fba0093e2dcd61130 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3f6531cbb3f4292b3d17ea85a480d2b : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28cba0c5f0af45e29f0a1dffa97f0c99 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf467852819a44328be4a1429524a003 : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25df57436ef84f93924a6b79c4582c9e : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec8b8d508c2a4aff8a4edd16a6ca0b81 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc7b6a4c195e4e77a6c51afaa13fa060 : SUCCESS in 12m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/87e31df5fb4b4396b10686ea88b5aa66 : SUCCESS in 41m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bc3e109f6be945e7af107a08e219ce70 : SUCCESS in 1h 12m 58s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/9fe49443565f487c89a8eac3b804760e : SUCCESS in 1h 19m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53cd5ab1f2fe489ea6a173c19310bbdf : SUCCESS in 58m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cadfe624d4f94f928c05c70e44001007 : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9ab041961a224fd2b8edf32c41cccb51 : SUCCESS in 25m 39s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9ca38e9993324752b01797de51ec7600 : SUCCESS in 24m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b99f3a7ff93e43ca9f3bdd8a0db43288 : SUCCESS in 16m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/57d91d88552345bfbe8d3a2d5284c50c : SUCCESS in 2h 00m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1494b56d67cd45a09a093cc4110ef846 : SUCCESS in 1h 18m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4c10fe599f3042ddae843131e99a5352 : SUCCESS in 23m 47s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1cf3ccac4d8d4036a8ba964fc16a5651 : SUCCESS in 21m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6f0dea29c534a3f9fdcc925e36d429d : SUCCESS in 14m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6eaf4d88c3814835a8fb3773aff2b6c1 : SUCCESS in 25m 48s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9e86c9184394bfb9aa4d45e97d91469 : SUCCESS in 20m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dc601d9004f54884ba3606746b54f8ea : SUCCESS in 14m 02s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/54201dbecc99473ca27d90bebc719cd5 : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1aa11c78b4154846815d56f49aedb368 : SUCCESS in 10m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/58273668bafd43f19c5da5795868f6f0 : SUCCESS in 15m 12s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4cba530219504d249d82143f7af88145 : SUCCESS in 16m 36s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6fefb354a93425baa90ee80400caa94 : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"99c4f75aba80beac41943c365e3d116dfbf1423a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 17:23:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b033595eeddafa10e2e84a048453575440c472ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 17:25:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2a6247bd664cd3ad68b8f96648af19 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c5b37efe60a42c293d6d47e2e71ba95 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c96d8cfb5440aa494cff2e89948fce8974d7a43","revisions":{"5c96d8cfb5440aa494cff2e89948fce8974d7a43":{"kind":"REWORK","_number":1,"created":"2022-04-08 08:37:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/56/837056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/56/837056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/837056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/837056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/837056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/56/837056/1"}}},"commit":{"parents":[{"commit":"8251b21237ca26edb63ac9b933b79c859cfa1a9a","subject":"Merge \"Add the bonding 802.3ad aggregation selection option\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8251b21237ca26edb63ac9b933b79c859cfa1a9a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-17 15:34:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-08 08:37:52.000000000","tz":0},"subject":"Fix Ansible inventory generation when reusing group names","message":"Fix Ansible inventory generation when reusing group names\n\nWhen a kolla-ansible group is composed of a kayobe group with the same\nname and of at least one other kayobe group, kayobe would generate an\ninvalid Ansible inventory such as:\n\n[compute:children]\ncontrollers\ncompute\n\nBecause the top-level group should already be defined, we only need to\nadd as children the kayobe groups using different names.\n\nChange-Id: I88bdf1e3d0c08271ac8938ae2f9ac3f9fee1efa5\nStory: 2009927\nTask: 44798\n(cherry picked from commit 0c190a22f0f50385f598fa92d6962842bd9b96eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c96d8cfb5440aa494cff2e89948fce8974d7a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c96d8cfb5440aa494cff2e89948fce8974d7a43"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
