)]}'
{"id":"openstack%2Fkayobe~837199","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I955a2c12b5143a63422cab50c74191bc22c63932","project":"openstack/kayobe","branch":"stable/victoria","topic":"forgotten-linuxbridge-agent","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-04-11 10:56:02.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-04-11 10:56:02.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-11 08:52:39.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I955a2c12b5143a63422cab50c74191bc22c63932","subject":"Restore forgotten linuxbridge-agent container","status":"MERGED","created":"2022-04-09 11:46:48.000000000","updated":"2022-04-11 10:57:48.000000000","submitted":"2022-04-11 10:56:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837199-forgotten-linuxbridge-agent","meta_rev_id":"b2306f35ae42e6f2bd68a1624f56502435f6be45","_number":837199,"virtual_id_number":837199,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 10:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-11 08:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-11 08:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-04-11 08:52:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-09 13:26:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-09 17:55:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-09 17:55:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-11 08:01:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"a3c3d0a725091d6bcfabb1105e4c94826d62375d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-09 11:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f25d27203879f6d79956a09b8de1e07800fd5535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-09 13:26:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9b7381eea564a55b61af541ce411cab : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/657cf4010f894ce180bf3cddfb6fbd5f : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14c8345df849407fb12f8de31917be4d : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b85bb1da9734e669101dcafcaa97fa9 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b22a6ed8ee4b42b3afc669a4aa3868 : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e42aa27d4f924ef19279a0e09aa090e8 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ca95b5e59af473c8cf1240e96759a5d : SUCCESS in 7m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f114e3f075d74781a1c92e8212737bb7 : SUCCESS in 16m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/10db1cc0b3b34e49978b3618455b8eb1 : FAILURE in 38m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/30adcb881f8d47609c91fb23f4b98cc0 : SUCCESS in 1h 23m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8553af7f307b467ba6a9b25eeb14ad76 : SUCCESS in 14m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/60fdd257c8c048a2ad87fd8b678be12d : FAILURE in 12m 48s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9be6ea7fd3814e06a90a26122422057b : SUCCESS in 12m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/39b101ab2d0d4c97adb6a028fbc1e5e0 : FAILURE in 12m 48s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/58862500b3754ac297d808d9993577b7 : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7f4411e3dec1146f29efd2051e7f08af2530ea9e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-09 13:41:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6f05e64a029908d61b1ac115bdbf885c6f5282b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-09 14:31:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5123e0c77844f8b8882157206b30a4a : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd5641b226b4db181bdf397a2a359de : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2877190c90494d3eba5f8f2be665b682 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/734f1c5377194239b479d6517afe38d8 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac4ef294933443088053b4ab0ff0ce76 : SUCCESS in 5m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8357debaf92a450592ca7b9d26e4b896 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/67cf7dd948f94d5788148a7c21fd20ed : SUCCESS in 6m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d7f63690b59492a8e6edb1c9b72b753 : SUCCESS in 13m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/866ac5e8d8764012958ddab7b7d35562 : SUCCESS in 49m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f06f7598ec0344fe82148d21276e68db : SUCCESS in 49m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/268ad4a75a07492c93406b8252315099 : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/edff0f6a4ff1460dab4bc2d073bd3ebc : FAILURE in 12m 08s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ecd74810b6be459f8a046b5e21141d23 : SUCCESS in 13m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/053e9f70d5224465bb7a6dbdfebcf806 : FAILURE in 13m 11s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/166426c817af4f3e85ac87d532881c47 : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0537f08e5b82e9f6efdb6c0f5e87e8c5fdf6ea97","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-11 08:01:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04108a79b7ad455aefaf07791a71059825fd1f05","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 08:52:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1bfd7c8ac77d11df08beedc064b4ac41087ca9dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 08:53:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e9223750218ac53748cd37c86677d4181d7e6e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:56:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67d23e602904678b7d57b7cfdfc0706 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a03269d053b4ffba8e3964006acd456 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0722acb082a40289fe4a513908e2747 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55565f2d1d994b9aab02454113c0966e : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d3b6f092cb9450c8091d689d9d23641 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f2c5c618a8041c896361f770c28689c : SUCCESS in 8m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/054d96bdf59b4114aa178025ab9fdbb9 : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e36ce168434745c9a3fc725af9e4edf5 : SUCCESS in 59m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/af7399ca539049ee8dbf79b88dba6c92 : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3e4b636235124810adb8e07ddebc629c : SUCCESS in 15m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/725ae7909acb48d3b5bd003509a96391 : SUCCESS in 16m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/425f6902f7414c0f8549f64e8b2f5400 : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"01b6eb98cb3454e6f3c2599ee9e04d2c412d05ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:56:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b2306f35ae42e6f2bd68a1624f56502435f6be45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:57:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b75c057f7f6c4e62938cb7acf62f6ac8 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19faf95eec09d96d7a5460353ee7240358dc6854","revisions":{"19faf95eec09d96d7a5460353ee7240358dc6854":{"kind":"REWORK","_number":1,"created":"2022-04-09 11:46:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/99/837199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/837199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/837199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/837199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/837199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/837199/1"}}},"commit":{"parents":[{"commit":"a00892dc621ba065f42962c6eb85d3bf5a538c3d","subject":"Fix Ansible inventory generation when reusing group names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a00892dc621ba065f42962c6eb85d3bf5a538c3d"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-07 21:16:49.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-09 11:46:33.000000000","tz":180},"subject":"Restore forgotten linuxbridge-agent container","message":"Restore forgotten linuxbridge-agent container\n\nThere is no mechanism to check which agent would be used and build\nall the possible agents, but forgot the linuxbridge one.\n\nChange-Id: I955a2c12b5143a63422cab50c74191bc22c63932\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit a8860c4e8826689909088f95ff15f1cb2813647f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19faf95eec09d96d7a5460353ee7240358dc6854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19faf95eec09d96d7a5460353ee7240358dc6854"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
