)]}'
{"id":"openstack%2Fkayobe~837264","triplet_id":"openstack%2Fkayobe~master~Idc0cb70dec38aa03e090f0f03feb1870ab523d2c","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-05-30 11:19:18.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-05-30 11:07:08.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Idc0cb70dec38aa03e090f0f03feb1870ab523d2c","subject":"nclu-switch: always use become","status":"ABANDONED","created":"2022-04-11 08:40:03.000000000","updated":"2022-06-09 21:28:22.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"742dfbd5c8edd6de5a9f6682bcfd17d240fefaa3","_number":837264,"virtual_id_number":837264,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-11 10:30:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"date":"2022-04-12 09:58:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-04-11 09:39:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 08:48:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-11 09:39:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-04-11 10:30:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67d2718863eae29a08b944b7027a48ded6854b92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-11 08:40:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"276786afa1a61c81324432bb0a491ef028eb61c8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 08:48:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aee8edef32e88e6d3f0fb7005324a4e7d31f9596","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-11 09:39:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67a0c6b929b213846269cb4912fe03e48a260d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:30:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c1fc84601ce49ebbf84867a6fc80c8e : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12173b855a274c8fb21f46a92b5840d3 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/286a7eb3d30c47299c00947486914714 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/903aa791b9ab4e1cbc65762b5aa6482c : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc21377445d242779d85fda928ef5783 : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5cccbb8f61ef4d9ea4d4cdf1c5c98fb6 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f86cf0825bac42b5956e5eecc837d3fb : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5fb203ff26e447cb8f1484bc6881a8dd : SUCCESS in 28m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3c83e3238b2446679d6cd9125c866cec : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d19c62a8de6c43b99fe352bb4b5375de : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e767e235427409f9cf424d8a7bd5719 : SUCCESS in 46m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5199e40477b54cdf8f376aa09feb96e9 : SUCCESS in 1h 00m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/52b2fe5f25884d77b56b9bec2a89d9be : SUCCESS in 21m 36s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/d2e6d71496f843db9a1048f3dbd2b33e : SUCCESS in 21m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24aa05ce7ba841e88033154a772edd9c : SUCCESS in 14m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d3d1ec599bcf4afea7797b537a6e68b1 : SUCCESS in 1h 47m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc6c87d02f8e415e8cb1704ad96cb919 : SUCCESS in 1h 02m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/09b6a33e2c3247b6ba03075e00450756 : SUCCESS in 19m 56s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d34be172c7314c08b122f45e8e7a7d60 : SUCCESS in 18m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a5cbcb3b5af4dc0a93be2ff7d16010f : SUCCESS in 15m 24s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/7ab911e8400b4e129f82627f176cbb9a : POST_FAILURE in 41m 36s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/52e5d5f91d3149a4a0aed0c602b5b4c4 : SUCCESS in 37m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca88ae0cb1eb4bb8a28c2ab2c31ae8e7 : SUCCESS in 25m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8004b65611a84327a9787e20e4224c58 : SUCCESS in 17m 20s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0684855b656041e58257779a68a73505 : SUCCESS in 16m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4412461bb7594f2391d8f3d7f09997bd : SUCCESS in 14m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/60c1ca76dc37442b842742e0a2321fd1 : SUCCESS in 19m 15s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/63f129059a11452989b53f8af695057e : SUCCESS in 17m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08028bf85e01499cbf8be838a1d996ab : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8394cb6fc783b8b24643ed3c4bf2057a99386fa7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 09:58:49.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"32388f442776f2501ab50ddf000f4f99286212ff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-12 10:08:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59c4a692ed1dcc4301ba09a1ce9355b80e8b6e9e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-30 11:07:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6135d6549b88c875fb216169d7a65704ccbbdbaf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-30 11:19:18.000000000","message":"Abandoned\n\nnot required","accounts_in_message":[],"_revision_number":1},{"id":"742dfbd5c8edd6de5a9f6682bcfd17d240fefaa3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-09 21:28:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15f3818ca86a43bdd3dcb0e92e70bc0e0d5949d1","revisions":{"15f3818ca86a43bdd3dcb0e92e70bc0e0d5949d1":{"kind":"REWORK","_number":1,"created":"2022-04-11 08:40:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/837264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/837264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/837264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/837264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/837264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/837264/1"}}},"commit":{"parents":[{"commit":"a8860c4e8826689909088f95ff15f1cb2813647f","subject":"Restore forgotten linuxbridge-agent container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8860c4e8826689909088f95ff15f1cb2813647f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-11 08:39:15.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-11 08:39:59.000000000","tz":120},"subject":"nclu-switch: always use become","message":"nclu-switch: always use become\n\nNCLU net add commands need to be executed using sudo\n\nChange-Id: Idc0cb70dec38aa03e090f0f03feb1870ab523d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15f3818ca86a43bdd3dcb0e92e70bc0e0d5949d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15f3818ca86a43bdd3dcb0e92e70bc0e0d5949d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
