)]}'
{"id":"openstack%2Fkayobe~837306","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Ic779bb80f20fd72c73f0df05e048c851fe8491ee","project":"openstack/kayobe","branch":"stable/wallaby","topic":"fix-dependency-upper-constraints-file","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-04-12 10:54:43.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-12 08:39:29.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-12 08:55:27.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Ic779bb80f20fd72c73f0df05e048c851fe8491ee","subject":"Fix variable name for stackhpc.os-networks upper constraints","status":"MERGED","created":"2022-04-11 12:46:58.000000000","updated":"2022-04-12 10:58:24.000000000","submitted":"2022-04-12 10:54:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837306-fix-dependency-upper-constraints-file","meta_rev_id":"5eed965a04de957e91d6e5df2850ca25a1ee3bfa","_number":837306,"virtual_id_number":837306,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 10:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-12 08:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-12 08:55:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2022-04-12 09:20:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-12 08:55:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 14:23:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-11 14:51:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-04-11 14:51:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-12 09:20:37.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"67e33f54090b4284d18b9f2000744f6a6186ad16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-11 12:46:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46b2133ee05c82c232a52fdb446a00994bb63eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:23:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19f3a1d2f55749d1afb98ad33160e69c : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f01269617cb4e468ecb12e7edf5a447 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ef0652c5f6e4dc783f1484243c08509 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6db674ca909d46e3a4abdfd519796b35 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11cfc2b053894018bdb605f2e0308540 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/073d0241feab4a8da5a5a7ab23452d95 : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/851e25c779c2440ca65389f1b80d4697 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a196777cdc054cc9946024348673b829 : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70de578265e946909b07f63e9647f803 : SUCCESS in 16m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/56e6ed9bd10a4221a934ba3a70fc0bc1 : SUCCESS in 58m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdde432e92ac470bb5773962903da939 : SUCCESS in 51m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3f3fcb59b232426baec4cba71c00bcba : SUCCESS in 1h 01m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/554a89ee7c4743c3b4f33726cbbcbbbc : SUCCESS in 20m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84801efe508c4919b06181df6c5dbfa3 : SUCCESS in 15m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5f66870e38484ad59c7cc11f313ca88d : SUCCESS in 1h 33m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/50ada3c1179742a19970be2d1714e467 : SUCCESS in 21m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e51c5cfbcbe94fb0a96f03e1392e0c97 : SUCCESS in 14m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e8b3b6ea5784ec098bdc322058216dd : SUCCESS in 25m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/287fb26c833f4e61b6c8b59ebc19a735 : SUCCESS in 17m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/851492f3667c4ed391f8464cccca14b9 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a5402021384a4707017bfae99e2a8cbb0c5b9233","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 08:39:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d6f1a099ec110c0a6124f4f717410daaf3af491","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-12 08:55:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c270326e93a571a2c6837e5f88754331188d3240","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 08:56:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e5fb86894f0f5874a17cbe410f8d0ebeb22094c","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-04-12 09:20:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8173691127667d656eec5eaf29eaba22f932ef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:54:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b319946b901a4c769d280c7597fb9c0f : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0782bdcd9dd74f748b0f16a24e997a5f : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f497efc4c734b298e786ebef5f946d7 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99088cbfa33844c084589a517e73f681 : SUCCESS in 8m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/97307b428e9a4acba4cf71424865355d : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee037baa45d9424ca74d3b87aa916cec : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb12bbefa4c742e891295de4d9946daa : SUCCESS in 19m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9600bfc5e4a3451c9ae915e50802fa30 : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bae2ce15ff840efbe0f369eb273186a : SUCCESS in 52m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/13bfb02dc16b449d84b8172ea18d2195 : SUCCESS in 59m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7db0776b98be4a8bbe5781a771aa1760 : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/636a2781bd794d2abd9e7a7b6cc0b4f7 : SUCCESS in 12m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/92d7dc43c6e94cc2a0352734e3ce2a19 : SUCCESS in 1h 40m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d4e8a1105cef400294acced1694c6449 : SUCCESS in 20m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53784209707b4052952e6693f572d1b8 : SUCCESS in 13m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7b3e6af8171e4cb1beb95ebbf35a9be1 : SUCCESS in 23m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bfee0384b8b14ffbbee81341513eecdc : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b29657ac64e14cf89fee53ef91d8d1f0 : SUCCESS in 24m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e3ba747c925b2bcc6efafb701843ef6d112a8d5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:54:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5eed965a04de957e91d6e5df2850ca25a1ee3bfa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:58:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e56d7684b3b4a8f9a645005c86ac0ae : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09a4c2946d066b9352110464522ab4e7689f5716","revisions":{"09a4c2946d066b9352110464522ab4e7689f5716":{"kind":"REWORK","_number":1,"created":"2022-04-11 12:46:58.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/06/837306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/837306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/837306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/837306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/837306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/837306/1"}}},"commit":{"parents":[{"commit":"4d7e7a574e4fbd356752618026071722be62d39c","subject":"Restore forgotten linuxbridge-agent container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d7e7a574e4fbd356752618026071722be62d39c"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-09 11:54:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-11 12:46:46.000000000","tz":180},"subject":"Fix variable name for stackhpc.os-networks upper constraints","message":"Fix variable name for stackhpc.os-networks upper constraints\n\nUpper constraints should be defined using\nos_networks_upper_constraints_file rather than\nos_openstacksdk_upper_constraints_file because of [1].\n\n1. https://github.com/stackhpc/ansible-role-os-networks/blob/v1.5.3/meta/main.yml#L22\n\nTrivialFix\n\nChange-Id: Ic779bb80f20fd72c73f0df05e048c851fe8491ee\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit a224579ed168fd1fba428eb3f9dd6e2c7de8a181)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09a4c2946d066b9352110464522ab4e7689f5716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09a4c2946d066b9352110464522ab4e7689f5716"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
