)]}'
{"id":"openstack%2Fkayobe~837477","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Idecda342f3ab86733e8d59061458d44af834dbb0","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-05-03 08:43:01.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-05-03 10:31:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idecda342f3ab86733e8d59061458d44af834dbb0","subject":"CI: separate image builds into a non-voting job","status":"MERGED","created":"2022-04-12 14:49:45.000000000","updated":"2022-05-03 10:33:48.000000000","submitted":"2022-05-03 10:31:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837477","cherry_pick_of_change":836778,"cherry_pick_of_patch_set":5,"meta_rev_id":"2412ecd028a328e8d8df70c2d7c55aef260b3239","_number":837477,"virtual_id_number":837477,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 10:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-05-03 08:43:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-02 08:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-05-03 08:43:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 14:51:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-13 12:35:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-05-02 08:46:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"9eb3466d532bce55b4a7f8d0a7f70ff486096f82","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 14:49:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.\n\nThe following files contain Git conflicts:\n* playbooks/kayobe-seed-base/bifrost-overrides.yml.j2\n* zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"cc1cb6c691e72290c2859deab5c6f61177e5a10a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 14:51:01.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch stable/wallaby.  The error was:\n\n  while scanning a simple key\n  in \"openstack/kayobe/zuul.d/project.yaml@stable/wallaby\", line 43, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/kayobe/zuul.d/project.yaml@stable/wallaby\", line 44, column 1","accounts_in_message":[],"_revision_number":1},{"id":"43527edd53189df9213105ab87200e97e38788ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 14:51:08.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2c9c89149c2e176849eb0b3e5bfe53cca1d125d0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 14:51:11.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"8151ea6ed41499cb45fcd4f8f5e532922275012d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 16:51:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8b101d3e2ad492797fcfbd674db3a62 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/571284f3ee734b0e9ffad0ebe1d5f4ce : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/441b1566290a451e911adf51a69b97c0 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5238a6cc5feb4a189725275f2a099487 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a13313f8cc9425b8bb8f3599ddf5e6a : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8998164b315f4eb7a79a2c5ea7f86726 : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d7eabb6e6dad42e489735fe5a34e702e : SUCCESS in 7m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c80023d79bf436988a2ea72db3d73a1 : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/15b3e280bc3e4b07a3a427247206c4a2 : SUCCESS in 21m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e19c1d8664424d1ca967c282b268a0e8 : SUCCESS in 1h 16m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e705caae40014eb8af26157b8c808960 : SUCCESS in 59m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e067ea2d74e64038ad913daa06a46a00 : SUCCESS in 1h 17m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/56c8394dc34245e281c16dbb74f3d7b4 : SUCCESS in 21m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c3774e5d4a0462b9e08d1cbdbe856a4 : SUCCESS in 17m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/99dc27bc5edc47bc90d7ad90c9134da8 : SUCCESS in 1h 55m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7de740bb74bf40448393a31326f8caec : SUCCESS in 19m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/180e9b53a2ac49e9a002369ab820d8d8 : SUCCESS in 17m 32s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/86b543c977d44959a9999db71fe474bb : SUCCESS in 46m 33s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e94664f98bc64cd787da5b357e406838 : SUCCESS in 27m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fdeaa930f646418883525847e4fbd03b : SUCCESS in 15m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cb871201d2648b48150cf709f892c61 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0272bcd8262afed61efe729045438c3ba1154d3a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-13 09:22:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e18bdf0e237b03a32ecc94ee9597e205bbe98900","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 09:53:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/8eca5a77bdb94a7cae03a6db1b3e60c7 : FAILURE in 9m 43s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ae288e4a58c4cd7ad6548621eab8e0a : FAILURE in 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f494b968650c45d3c8b74e380320697f4527a48c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-13 09:59:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0db3fe27fbeb5acfebce810d86b5407bc28e4516","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-13 09:59:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"346424fd13c628fb483c565c4cb670803307df0f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 10:26:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58bb38323e2a46c7af9094d832d75898 : FAILURE in 22m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"02d4e6d7051da539449d01f7c3fe367f8d9c0e2b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-13 10:41:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef0d8d54eecd0d98c19eec78e5022bd0314a508e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 11:38:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a3dfd4c021e46359b7e7c65c748f844 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1604a29d04e458f912781b8fe556752 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cad20d5718a485eb78c27deaf94c4b3 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dcee4b111b442739a83d107c1535780 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19e2e7423928409ea3b7b5e7e5702b8c : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ecb4bd1bb544138b25a5513fa186e1 : SUCCESS in 8m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e1257b286ff4074b7bf21d5d89b52bc : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c4007368a70c4992a57901a258d99eef : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f2720dc9301409e9fb3fdd09ff0b3be : SUCCESS in 18m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/72afeabe57de4a6bb838bcc7cd3b466b : SUCCESS in 1h 07m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96ebfe6926294d8f9f9b43b980bee923 : SUCCESS in 58m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d3c686e53f5e444e97c93f1e879247e2 : SUCCESS in 1h 16m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/532bd6b1f20f4cef888a3b21bad83bcb : SUCCESS in 22m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/172f2973de1145ba8b6bba7d56da5dc4 : SUCCESS in 16m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f98a5e4a69dc49bc8cd72d33f5c5a436 : SUCCESS in 1h 36m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eb580bb50fcb469683356dfd9e189639 : SUCCESS in 21m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe8f74cc6e0a4f1285e9b9b90fa7862e : SUCCESS in 17m 09s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/62f12ea4e73243f18515bd55adabcd9e : SUCCESS in 51m 24s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2cc5686f82e4ed4815248dcf3f616f0 : SUCCESS in 25m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cf7f2a296c474684ae336a577c658b64 : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1573b347b4fb4d059c7dca4c499a5ee8 : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":3},{"id":"ae1bd7195d0c76c58bfa011cf725c1616321058a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-13 12:35:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"28b71bcb38c6cf31da18dec32d8b156258e4741b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-02 07:40:56.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"44f0f3d897a86f62a1c973666b4aa76995a00715","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-02 07:41:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7453f2d21e7a58b366e26c0b8a088b30158a4e11","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 08:15:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed6482bda4a246a8aed0855bbae37d8e : SUCCESS in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"881797b8c5587d91dd66343064b4bd31c4c2ccc4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-02 08:46:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a0347b5d29ee22d216dfac88b0e16657736fb8e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 09:05:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09743f7461e34f5eb2568c65ff591bd8 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fadf53b2b4934cfb90df3f1ca538076e : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5523a943516640c88bc7ac7c6e0ef525 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47fffcdeed424f018ebb2b76bd4dbfaf : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63f1352f7aff4b39978dd1b4973ca582 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b104629e72de497bb0174679590651ae : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a68e799ec06c4f308316b669c963731a : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f02db5cd7ae44ad8b7f320d1b19bbe5 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a74df3745e6d4663ba6b07d7bbddadf0 : SUCCESS in 17m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a1918109356e4be7ab5a1736b1bf14e4 : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ba94bb48fcb4c77a1994ed9c1f64926 : SUCCESS in 55m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/60e4ba9149ab4dc4a7167f695ea2610c : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ec22c49c985749b1b7da24de5df74772 : SUCCESS in 23m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbaa49716431403b9107f70b19fbd2e8 : SUCCESS in 13m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bf632949c2cd4c0683af83b0e09fead2 : FAILURE in 1h 19m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/85cdc95242fd42488e0d00d47744ef1c : SUCCESS in 18m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c088dc6fd9a94da08d5b54a8a28d26ae : SUCCESS in 14m 39s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/4b20f958bdf14af28f33d194b53ac1fc : SUCCESS in 44m 02s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f65154001cc24b99b20550e6df5f9b00 : SUCCESS in 24m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1585e22abbb640678ef5c543cac7b923 : SUCCESS in 14m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/575a528837a440d28aa507182d0822c0 : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":4},{"id":"fd6bf959d0e195294adcc902382ff988d854ed7d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-02 09:19:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"12d380241bf266133ef2ce6072b02d00dee5d051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 10:49:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/234c0077cee144aabc2214d10c7ca54c : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b69f4feffe774c89af1a00357d1bdc8d : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/633eb3d6f4cf4648b5cfab7d2ce84444 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/683505b75a9248c49687f656fbdc5876 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b44d98b3c5242ca95f678c58ea180d8 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b39ed57b9e4b49bf3ebfdc2c54dd34 : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4cc14569db4b469a8759240688d970a4 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a31a5552f6394d8da2e4134cf77e1c27 : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b4ddcd64c35431cbe672e68727684c5 : SUCCESS in 14m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e83f294879304d0ea3be8a1c61d22c09 : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d164f4fadb6241b1892d205dd8f4f1ef : SUCCESS in 41m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3334963ce0074383b08af5a6110c3f7a : SUCCESS in 1h 13m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/04344d8ec30f4323a6e12ce1a724a3c8 : SUCCESS in 18m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a77920b2e3a34a2c852f29e0729b4aff : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb4deba788744d8c9934b54048e7484e : SUCCESS in 1h 26m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b8e12905efad4097b4f9c4ded6dba6b1 : SUCCESS in 19m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/166ac3fc1c874e09b580da88e989cc4f : SUCCESS in 10m 47s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6ea81dcf7a3a47f6887b722c008c90f7 : SUCCESS in 46m 23s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/573cca3348164f428be9dc3c38fe07b3 : SUCCESS in 22m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/38c3bcd5b0f444c9b22761c50fe59651 : SUCCESS in 14m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/313f8cd82f3c43ca9d4340913b2a53d1 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":4},{"id":"e4f0227a7bbfd2f5b739c8bfefeda66756272298","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-03 08:43:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"39b4feab55a068a180effac6e8e9f756e87dee79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 08:43:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5c578ec9d9b68949573f39dadecfcb5e62536327","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 10:31:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0210fadbcff4444845523c9aaae9fa5 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f693535054b49b8bd3dbc3f7999b711 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5784937e8c5e4eaaa454c6a1fc3146be : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae12665251584d55977e38b3a05a8123 : SUCCESS in 8m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f18d64659034f54b8f6c57dc9aaaeff : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ab6c8b17ba7044beac2cc4f145f3ecfa : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75d2211a943c473eb3a165b02656ba7b : SUCCESS in 18m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/421e2769c594474ea4e5cb65e0f59dd7 : SUCCESS in 1h 15m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3229fe1b19b04b1db28995dc6d6cb144 : SUCCESS in 53m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/410091899d3e4b2faf20e4fd4cc9a08a : SUCCESS in 1h 14m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d1c03593e04c49ceb8fe34e85a5c2e6f : SUCCESS in 21m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6d447ea9793407586f204e38e167bee : SUCCESS in 12m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/49aa95b6496a46ca84229c2b3026aca8 : SUCCESS in 1h 44m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0044801c04b44fb7811c2ff707b628aa : SUCCESS in 19m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ed17f6fde6c49b6b6c60f13e74de215 : SUCCESS in 13m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/de5b76c5ce92471b883d50fefe75f0b1 : SUCCESS in 26m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/45735c4d5d954bb09fb413d452bdd932 : SUCCESS in 13m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92b531b0c2a947aca1fe54c9487ddb89 : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":4},{"id":"693ddaf28117049c7b8cf870dade7de390842e60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 10:31:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2412ecd028a328e8d8df70c2d7c55aef260b3239","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 10:33:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a5e79107e5d4b378b9f8939cc9c34b5 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c231848e3887e35ef897e81f8fb9370f9ecd42a7","revisions":{"4da13ca78169fe6fb60c787e8b97ba2e4316bea2":{"kind":"REWORK","_number":1,"created":"2022-04-12 14:49:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/837477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/837477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/837477/1"}}},"commit":{"parents":[{"commit":"09a4c2946d066b9352110464522ab4e7689f5716","subject":"Fix variable name for stackhpc.os-networks upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09a4c2946d066b9352110464522ab4e7689f5716"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-12 14:49:45.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n(cherry picked from commit a7ee3ac5c8739d7547ae430574a5ac8abc79c41d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4da13ca78169fe6fb60c787e8b97ba2e4316bea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4da13ca78169fe6fb60c787e8b97ba2e4316bea2"}]},"branch":"refs/heads/stable/wallaby"},"55dc15ea70cfd2e8b98fdf7b01d8e5e7a7d086ed":{"kind":"REWORK","_number":2,"created":"2022-04-12 14:51:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/837477/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/837477/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/837477/2"}}},"commit":{"parents":[{"commit":"09a4c2946d066b9352110464522ab4e7689f5716","subject":"Fix variable name for stackhpc.os-networks upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09a4c2946d066b9352110464522ab4e7689f5716"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-12 14:51:08.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n(cherry picked from commit a7ee3ac5c8739d7547ae430574a5ac8abc79c41d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55dc15ea70cfd2e8b98fdf7b01d8e5e7a7d086ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55dc15ea70cfd2e8b98fdf7b01d8e5e7a7d086ed"}]},"branch":"refs/heads/stable/wallaby"},"70cec7c998e4886a00196c8ba27440920da29237":{"kind":"REWORK","_number":3,"created":"2022-04-13 09:59:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/837477/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/837477/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/837477/3"}}},"commit":{"parents":[{"commit":"09a4c2946d066b9352110464522ab4e7689f5716","subject":"Fix variable name for stackhpc.os-networks upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09a4c2946d066b9352110464522ab4e7689f5716"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-13 09:59:27.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n(cherry picked from commit a7ee3ac5c8739d7547ae430574a5ac8abc79c41d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70cec7c998e4886a00196c8ba27440920da29237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70cec7c998e4886a00196c8ba27440920da29237"}]},"branch":"refs/heads/stable/wallaby"},"c231848e3887e35ef897e81f8fb9370f9ecd42a7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-02 07:40:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/837477/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/837477/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/837477/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/837477/4"}}},"commit":{"parents":[{"commit":"e449239a5793f5985bcc2ff48d4d192fee2b1c1e","subject":"Merge \"kolla_passwords: add no_log for password overrides\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e449239a5793f5985bcc2ff48d4d192fee2b1c1e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 10:46:02.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-05-02 07:40:56.000000000","tz":0},"subject":"CI: separate image builds into a non-voting job","message":"CI: separate image builds into a non-voting job\n\nDisk and container image builds tend to be fairly unreliable.\nWith 3 voting seed jobs all building images, this can introduce\ninstability into the CI jobs.\n\nThis change adds a non-voting kayobe-seed-images-centos8s job, which\ndoes the following:\n\n* Builds IPA images\n* Builds an overcloud host image\n* Builds a base container image\n\nSimilar Rocky and Ubuntu jobs are added to the experimental pipeline,\nand may be run by commenting \u0027check experimental\u0027 in gerrit.\n\nThe existing kayobe-seed-* jobs no longer build images.\n\nChange-Id: Idecda342f3ab86733e8d59061458d44af834dbb0\n(cherry picked from commit a7ee3ac5c8739d7547ae430574a5ac8abc79c41d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c231848e3887e35ef897e81f8fb9370f9ecd42a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c231848e3887e35ef897e81f8fb9370f9ecd42a7"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
