)]}'
{"id":"openstack%2Fkayobe~838404","triplet_id":"openstack%2Fkayobe~master~I6424fc405894514a63fb2b641637bbb9d5c070c0","project":"openstack/kayobe","branch":"master","topic":"proxy-fixes","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-04-26 15:58:20.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2022-04-26 15:58:20.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-04-26 14:06:13.000000000","reason":"Michal Nasiadka replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-26 15:58:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6424fc405894514a63fb2b641637bbb9d5c070c0","subject":"Fix no_proxy configuration","status":"MERGED","created":"2022-04-18 20:44:56.000000000","updated":"2022-04-26 15:59:45.000000000","submitted":"2022-04-26 15:58:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838404-proxy-fixes","meta_rev_id":"661d6f303c4aa3ffe949c2562a5831f677186ce1","_number":838404,"virtual_id_number":838404,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-26 15:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-26 14:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-19 15:35:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-26 14:06:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-26 14:02:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-26 14:02:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-18 21:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 06:36:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-04-19 15:35:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-26 14:02:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2022-04-26 14:02:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-26 14:02:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"36ecc5796ff16d9cf38e69d5a5614b78247767d0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-18 20:44:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b15de12494e65b0e9e658a5e2d487963892d276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 21:46:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/225d4a44a982493a9e6f2a9b231ee011 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee722d9bf1f641278d5c0abce16b6f1f : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/967c24fdf166410d892781dfa210e6b5 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca571889053e4fdc8b03755ef47818cc : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b0c7c4c0efa4446a49a7ad32d0ac39e : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef001156c8c415993368154bd8d0d36 : SUCCESS in 7m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d2f977d9ffcc40b0ad685277d11f00b5 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e63c42c854264feeb8df8635082c2f5d : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dead64b0166344c0b4a01aef18a1d063 : SUCCESS in 14m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a36ca32aeb984206943152a79a08d125 : FAILURE in 8m 49s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e82282e0d1a0451dbf8903b3785c0c7a : FAILURE in 9m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a10e025d08d44f0e87b99d3d46855f7b : FAILURE in 7m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4cc6fc404fe944b39885aef408a18729 : FAILURE in 8m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/42ab78da65c4428388314e1a2078db02 : FAILURE in 8m 41s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b625b0226dc74f198d25a29b73f3c330 : FAILURE in 9m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d193b6522cfb4202ae57819340624713 : FAILURE in 7m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/368b60a552a346ed89c78f4afac53c83 : FAILURE in 52m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8d184f7d650402eab85696d407a3f2c : FAILURE in 39m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7f70e45cf8be4e1ebb20482851af39e7 : FAILURE in 7m 53s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2a6a217ded4546fb9176335826017243 : FAILURE in 10m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/206688a88d464d64b7ac5dc212efd30d : FAILURE in 7m 18s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/56f61849b78b42b9b0c0b70c8205c448 : FAILURE in 8m 09s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e7dac1d645446d78ee2525962db4a48 : FAILURE in 25m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4371fb73d8d4c979d69f7c4a0134777 : FAILURE in 20m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2ddcc76a780e4901bbedcc00aeca7f56 : FAILURE in 8m 22s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7cf145333d5446f0ac5e4b26e1599674 : FAILURE in 11m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d35d95ee775b4ef3b8fe3abc905e43a7 : FAILURE in 7m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7713269b223c48b7b6e04bf04825219d : FAILURE in 8m 09s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/eabf6d25933d449caa659fa0ddb827b2 : FAILURE in 9m 58s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f49ac08b6c244e548f017800faa31d76 : FAILURE in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c510d8f1456eb30a40f006e0c18bf431efd4738e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-19 06:36:08.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"930ac6dbb8f64ecdefd9f067e008142209671dbe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-19 06:45:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27c72c542e0d89bcd6978fd42754143d93b816b6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-19 06:45:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d80087dc5220968b3d44797964b92f1d51bf2510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-19 06:58:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80e6d369a2a3ddc81ca10136f4d2894e08069dd3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-19 06:59:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"32393af63563b43925c3cc60a1d5548ec31caad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 08:57:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0269dd9b1a24510b427e5b2ec0fb915 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c6bfaa76dfb4c52ac7177df74e6630c : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad600148eee84bfb8b92a2b6a87da729 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a5b3b38d2244548a6001051cab48fd9 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fde2da6a3dc348c98b6d39a0e5e773ae : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/099e826d75ab4ca68d2e1b862b84ab8e : SUCCESS in 8m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c69896f4be914184a0ecd52d803ce9af : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa651287651c438da5097b848f04cad8 : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b47e6d977924a9eb1e9aecebd54b8c6 : SUCCESS in 18m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/daec2f4b5c3e4f11bd1080a288da097b : SUCCESS in 55m 08s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b858702ca67c4543a4b6f3671fc9f47e : SUCCESS in 1h 05m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06aba2166e94454198b5e2fa1f8c467d : SUCCESS in 59m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/39e5b6a5be8a425a993a70d5aa0d21a1 : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/75d55fd9f05d4f6ba726a2e86b1714b1 : SUCCESS in 22m 04s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7f92012f7477458697bdd8a7d4e6fa6f : SUCCESS in 21m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab326bb78fa142caa715b4e525cf606f : SUCCESS in 14m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bab6b4369ef5437896915c6d2ed43b2d : SUCCESS in 1h 53m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3aebe85fa4b440c938fd9d5d37191e1 : SUCCESS in 1h 04m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f3a1400ad77d46b3acc9c7c83710f495 : SUCCESS in 17m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f840c04416b145e2950c1a583336b6ad : SUCCESS in 19m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aa02a8de1e2487591283ade04ef5c64 : SUCCESS in 15m 21s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/414bb58ee6d44395ba3539f51316423d : SUCCESS in 54m 13s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eff2a324f65d42b1817f974248766993 : SUCCESS in 36m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d267526787241efb4cef1f61c4f9707 : SUCCESS in 25m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b93627b1033d48e39ed69707a3f3f6cf : SUCCESS in 23m 09s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/837ccb7f2fca4e43a140b8b6b1e7f458 : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/125161fa2b7d4ec4aa48874267465044 : SUCCESS in 9m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/47dcc89d091547d9afd422ea4e62557f : SUCCESS in 18m 25s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6acb8a301ace41beacc5df5294d452e8 : SUCCESS in 16m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6487789843e4540b2e5547d730f1885 : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":2},{"id":"17d5311359a8e7cb532babb2070e32fd6f963df1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-19 15:35:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"940cac2f7d5bf6e093405a5cee7ccb814d96caa3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-26 14:02:28.000000000","message":"Patch Set 2: Backport-Candidate+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6284620d6553c48ef736ac87fe37906487360a3c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-26 14:06:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9523ad287ecbdab94bbb08f3113da41ad3eabf47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 14:07:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0b3a5b53fcbf9be9063abc610f99444a3e6dba2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 15:58:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e73315e783642228b2476ce7272508c : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbcce66a39d148ff95ee7de9c8f33255 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a91b7de15cb492ab33c5d0f579d88df : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62d71912c9d4efe9c0a2a0576779b4e : SUCCESS in 8m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f86581436c2b4d9d98ee212329f3ba52 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f2e36d7e8da454f9e3d0c74bc229c5f : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/006c056eec5c4966898d4f3ae85c12d5 : SUCCESS in 19m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f98108496d294fb89e12f604c008941e : SUCCESS in 50m 21s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d51c1abe7c984f9bb68ef1b28e3697d6 : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d59cad33fcb34ce6902102e6fd70b298 : SUCCESS in 42m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5f50fdace1464499a6d44ef9ea8f49c6 : SUCCESS in 1h 08m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ecffd66296a94b5280dba9d1ab8b130f : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4549ba831f8d4776b03b1866d8876250 : SUCCESS in 22m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0454bafc4dd41a385bbc844ae23a6eb : SUCCESS in 13m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f6bb57c3adb74fdb82ef6207c5245c09 : SUCCESS in 1h 50m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9d78589ae6843e0a2d15de0da0931a1 : SUCCESS in 1h 19m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9744e195feac4925860d836366fc004f : SUCCESS in 17m 29s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/cd80a9bb10244bfbb34f4c4fe03c74ff : SUCCESS in 21m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cf4c01900de48baa341082faa2cfa34 : SUCCESS in 14m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0a4f3501c48546c7909c27f607f95fcb : SUCCESS in 28m 22s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8b241f64b054632850c47a57444ca0c : SUCCESS in 30m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8a7348bd37c849d4a7e9a657be1ab68e : SUCCESS in 15m 45s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8ed14fcf317d40b5a131b1590f43e64e : SUCCESS in 16m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4b0ac5ba87649bca8bada9968d1eadd : SUCCESS in 13m 09s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bfdd587a496d437c9a06592132be7ed4 : SUCCESS in 17m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/25667879a9954e8ea0e7dcbe2f148c9c : SUCCESS in 19m 01s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/257631be023847c68201acfaff049197 : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":2},{"id":"171846bccdef6c0d42ef093d075bbd2ba9c688e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 15:58:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"661d6f303c4aa3ffe949c2562a5831f677186ce1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 15:59:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edfe52d836074f8b9bb98e4374aa5a41 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"71eb21a3ddab249fe733b56c5f9b70d2947308a5","revisions":{"d4c7db7a73ffe85d4731943fbbc2597d6ad9782b":{"kind":"REWORK","_number":1,"created":"2022-04-18 20:44:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/838404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/838404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/838404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/838404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/838404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/838404/1"}}},"commit":{"parents":[{"commit":"45231fdd666833623f38d1ab1046870cf66b0f73","subject":"Merge \"Revert \"[release] Use OpenStack Yoga release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45231fdd666833623f38d1ab1046870cf66b0f73"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-18 20:44:11.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-18 20:44:11.000000000","tz":120},"subject":"Fix no_proxy configuration","message":"Fix no_proxy configuration\n\nThe no_proxy list should only contain domains, hostnames, IP addresses\nand networks, but docker_registry is often in the form ip_address:port.\n\nUse urlsplit to extract the hostname from the docker_registry variable\nafter prepending http:// to turn it into a valid URL.\n\nAlso add missing infra-vms to hosts in proxy.yml.\n\nChange-Id: I6424fc405894514a63fb2b641637bbb9d5c070c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4c7db7a73ffe85d4731943fbbc2597d6ad9782b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4c7db7a73ffe85d4731943fbbc2597d6ad9782b"}]},"branch":"refs/heads/master"},"71eb21a3ddab249fe733b56c5f9b70d2947308a5":{"kind":"REWORK","_number":2,"created":"2022-04-19 06:58:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/838404/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/838404/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/838404/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/838404/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/838404/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/838404/2"}}},"commit":{"parents":[{"commit":"45231fdd666833623f38d1ab1046870cf66b0f73","subject":"Merge \"Revert \"[release] Use OpenStack Yoga release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45231fdd666833623f38d1ab1046870cf66b0f73"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-19 06:58:22.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-19 06:58:22.000000000","tz":120},"subject":"Fix no_proxy configuration","message":"Fix no_proxy configuration\n\nThe no_proxy list should only contain domains, hostnames, IP addresses\nand networks, but docker_registry is often in the form ip_address:port.\n\nUse urlsplit to extract the hostname from the docker_registry variable\nafter prepending http:// to turn it into a valid URL.\n\nAlso add missing infra-vms to hosts in proxy.yml.\n\nChange-Id: I6424fc405894514a63fb2b641637bbb9d5c070c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71eb21a3ddab249fe733b56c5f9b70d2947308a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71eb21a3ddab249fe733b56c5f9b70d2947308a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
