)]}'
{"id":"openstack%2Fkayobe~838730","triplet_id":"openstack%2Fkayobe~master~I5273edc8265f115519f499e85fe12f8d22306c4b","project":"openstack/kayobe","branch":"master","topic":"cleanup-glance-conf-swift","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-04-22 10:11:29.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-21 12:40:42.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-22 08:00:49.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I5273edc8265f115519f499e85fe12f8d22306c4b","subject":"Cleanup old and deprecated Swift configuration","status":"MERGED","created":"2022-04-20 14:33:30.000000000","updated":"2022-04-22 10:13:07.000000000","submitted":"2022-04-22 10:11:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838730-cleanup-glance-conf-swift","meta_rev_id":"bcf94dd4ba384b9599238cb074828e48dbe21c0e","_number":838730,"virtual_id_number":838730,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 10:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"all":[{"value":1,"date":"2022-04-22 08:53:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":2,"date":"2022-04-21 12:40:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-22 08:23:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-22 08:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-22 08:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 15:54:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-04-20 15:54:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-20 16:35:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-22 08:53:05.000000000","updated_by":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"reviewer":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"state":"REVIEWER"}],"messages":[{"id":"a413430011af41bc262ec932927f160d3b6d17d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-20 14:33:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1c9f6db38b1203b9922c66b5296fd6b1ac6aeba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 16:35:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c3e3dabe0ad47a6a14a65e73b3d79b4 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f364f2a15dc141e4a2bdd2b131564ba6 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc566013775e42a298c98a57641e14a0 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b192a469916444ea792f700fa5d450c : SUCCESS in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e991073046a64357b7c5a4cddda82055 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48129cadd7f340c1a28ef905e54b76f2 : SUCCESS in 9m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b540bdc7bfd4f8a8d7d75f999deefa5 : SUCCESS in 7m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e6486e9c699044c2a0cebc30fff00c1b : SUCCESS in 13m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3185ec2801bd451284b5bda266990dce : SUCCESS in 18m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3db9fdd5d9064f0789cf21ecd80dec18 : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4a4943f1f42c4803a20f7121d6830d0c : SUCCESS in 59m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c852e17248974539bf12f2cfb9babc01 : SUCCESS in 57m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b9e7861a281943a6b4ad30578a594d2d : SUCCESS in 1h 03m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0190e46b39a24c998902e2317428f7b0 : SUCCESS in 23m 21s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/fa32720747d740729d741ef0563c8732 : SUCCESS in 25m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa564c5cbdb94ae39c71441e1a4041a9 : SUCCESS in 13m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1ad13dbdacfe471aa66f80055a062819 : SUCCESS in 1h 56m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90e40c52616d4c20bc294dc5382dfae8 : SUCCESS in 1h 12m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3d488ea240484bab87f5e1f01dc0fabe : SUCCESS in 19m 27s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/489c850588954f69aa102e057174bb82 : SUCCESS in 24m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba721d623b6d45409a1cd7455a22f25c : SUCCESS in 16m 44s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/65d4478fe671495bb75290bc0e5b3ebd : SUCCESS in 45m 02s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/df5ca9366c2e460380dfe120f1f9363d : SUCCESS in 39m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf32841a9117425a909e45e8e4948036 : SUCCESS in 31m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/27090477e9364b3eb16d9a3e22341cff : SUCCESS in 15m 38s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/619bc4b61a4443efa8c9aa2061a0f71d : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7393020eadbf4b85912132e20f1493bf : SUCCESS in 13m 50s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/13324281f63b47caa4730e4c7ad784b3 : SUCCESS in 17m 40s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c55e2a2518db4f5f93fd609a85200ea6 : SUCCESS in 17m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4dc2c4321d1469fa48d01dcffb99b66 : SUCCESS in 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"688c0e027e076c869faa08547a7759a4ff4d4375","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-21 12:40:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3538404464ba5b4f5d51fc21270416657b3c4e2a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-22 08:00:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f8c1a540e89d035343a83ade47c7d8949d2d66d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-22 08:19:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d084db150d5d559dc19d540dd7306fef79041353","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-22 08:23:41.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8c14368b841b6de43e64a08f8cac405ae6d1bdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:24:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8287572a4770db11b339e4f7f1d20653b9ccbe8b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-22 08:32:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c55eb81a8a570d41e87f71a34addc545da9438c","author":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"date":"2022-04-22 08:53:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"510ecafe4fd188a69c08d1c86a78475030e84a04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:11:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b91825fc10c4ff79d5eecd5df61a727 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eddc135558384862a41a88089de3f206 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1ca096f33544b77bd749eac44ae1c9b : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c7c3d85161417f9084dd587dac85cc : SUCCESS in 9m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d086b96d30b484db85bbfdc7b28de5f : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea90384267f142059866ba5ae9f5702e : SUCCESS in 13m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9ef77059a02c4682a5d1792725107d4d : SUCCESS in 19m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/28b06b69093e4698a16609629d26d33d : SUCCESS in 1h 07m 31s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/1497dd972599429c943a89e3b189c4b6 : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b98820a48aa44f795ed09bb35f5bb5a : SUCCESS in 52m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1b46ac89fbd8410eb024773a8f475562 : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7d99a9a66cbc4e79a3c01e65f9dd46b5 : SUCCESS in 22m 35s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/27bb4a9f09ad4c568535687038121427 : SUCCESS in 22m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e027926ce7ca4df782d145b349a2ee88 : SUCCESS in 14m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/76101210905f47ad8d3e45945d2f309c : SUCCESS in 1h 43m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc995866bf814e7eb295cc84a6512200 : SUCCESS in 1h 14m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6d99b5d5e5a543d1b2cdf45ee4a355ad : SUCCESS in 20m 52s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f79fd8f7c432430eb1b38596f48ca2cd : SUCCESS in 19m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70877b7c2af149a580620b0ecbffcdbd : SUCCESS in 17m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b1ade61426a545179da43a2464154d37 : SUCCESS in 36m 41s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f8a2323f21c4c42b106e7022e2f39ac : SUCCESS in 28m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1e42643acc0444fc8f213d3d8a40a027 : SUCCESS in 15m 35s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/310cc2e9624341b7ad59a58bd0723efd : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b3947841d424b1ca5e92d1f37c6b22e : SUCCESS in 12m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/740aefd405214ab89a11d3b9fd0e2996 : SUCCESS in 17m 32s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/06cd04f3d8674965b21d582ba7882c91 : SUCCESS in 16m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e91a6d36f8d4282ac03bcf2e43c1049 : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9a4b7480a64d5964a5096e5ae10131be52c3dfbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:11:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bcf94dd4ba384b9599238cb074828e48dbe21c0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:13:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b958b862fb15481986cb5fbf8c8b9ca7 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10b8c1ead01c5372df40d565ac44a18ea20edc5e","revisions":{"10b8c1ead01c5372df40d565ac44a18ea20edc5e":{"kind":"REWORK","_number":1,"created":"2022-04-20 14:33:30.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/30/838730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/838730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/838730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/838730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/838730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/838730/1"}}},"commit":{"parents":[{"commit":"45231fdd666833623f38d1ab1046870cf66b0f73","subject":"Merge \"Revert \"[release] Use OpenStack Yoga release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45231fdd666833623f38d1ab1046870cf66b0f73"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-20 14:12:52.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-20 14:29:35.000000000","tz":180},"subject":"Cleanup old and deprecated Swift configuration","message":"Cleanup old and deprecated Swift configuration\n\nThis is old and deprecated configuration settings for Swift not needed\nanymore because kolla-ansible correctly configure glance-api.conf and\nglance-swift.conf since Idddbf2ce741e0486d60e1de88c77a7f0332a5a2b when\nkolla_enable_swift variable set to true. This change can be backported\nthrough all releases till the Train.\n\nChange-Id: I5273edc8265f115519f499e85fe12f8d22306c4b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10b8c1ead01c5372df40d565ac44a18ea20edc5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10b8c1ead01c5372df40d565ac44a18ea20edc5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
