)]}'
{"id":"openstack%2Fkayobe~839038","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I5273edc8265f115519f499e85fe12f8d22306c4b","project":"openstack/kayobe","branch":"stable/wallaby","topic":"cleanup-glance-conf-swift","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-04-25 11:56:43.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-25 08:25:13.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-25 08:32:04.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I5273edc8265f115519f499e85fe12f8d22306c4b","subject":"Cleanup old and deprecated Swift configuration","status":"MERGED","created":"2022-04-22 10:19:48.000000000","updated":"2022-04-25 15:21:56.000000000","submitted":"2022-04-25 15:20:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839038-cleanup-glance-conf-swift","meta_rev_id":"59dead8860304b6c2536365ef10a201375a6be23","_number":839038,"virtual_id_number":839038,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 15:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-25 08:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-25 08:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-25 08:32:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-22 10:23:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-04-22 10:23:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-22 12:11:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bca48e1b0eccd7042d1d84828c443f5c7b0d7661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-22 10:19:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b32649198f543af03cd504be3235cf4d3a0c7cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 12:11:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1979a0adc3f4dac8691da35a8480e54 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb7cb7e135b4ea38769a8f49b8fe395 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/765204caa96d4b0fab1a5e8e0dc329d6 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6be44ea7af84ba3844d9017977519a1 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5ea5b499a7e423cb1c17445e4103add : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348e9c186e8545ac9991c07eb2654216 : SUCCESS in 8m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc3b45f7241741399f50b96b7e33080d : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/82a7149f81cc4b0ea2434dc122d1ab45 : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c38111fc4e2458b9fce928a50341d56 : SUCCESS in 22m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7c5c22e57c2d45978056ea02482cd9a4 : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30dbf5d000134f829365cf6557e986d4 : SUCCESS in 57m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f0acb020b5ee4f4bb9327da5f2046af7 : SUCCESS in 1h 11m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8e2ec31c252e4ed591f0133da01f4099 : SUCCESS in 23m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd875a468dfe4bc98e81f60481c07c8e : SUCCESS in 13m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b642417c4f4f4672b7bb6c1694b992bf : SUCCESS in 1h 48m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ed23108135904a8bbe2bf883cc3d7f6e : SUCCESS in 24m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58f62203b03f4c17b3f50495cc93d136 : SUCCESS in 16m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/16d011ddd4034c2c883d3461adb0961e : SUCCESS in 25m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/344645a78ac849a3a613bd6bffac4b51 : SUCCESS in 14m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fd7b58004fd4083a4e2674f6abb5756 : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7e719b49dafdbf746519d905c2cbd56733e38605","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-25 08:25:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"297aa268e7c8e8313cdfa22203b2f5449a2dd81d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-25 08:32:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2359dbdf9466487497bfdf0d5921588992ab019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 08:32:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"947171df441ef930aea9e89f65dd35b956258943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 10:21:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a42a235bb94ce3a1ad3ac5cc7513e9 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f98b67f734141e4b63fd3e74dfe2f46 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbb40b3d09d046fcac17401d57b9fc17 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d20a419435d4a5082b760740f392b16 : SUCCESS in 8m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a3a0f96d9d54a9f9c56aad8692d4aa5 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45fc25184a794f13ab11ef5eb34169da : SUCCESS in 10m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e6da31ce57c4c2c83206a236f905fe6 : SUCCESS in 18m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/932b0eb096054401a7441698e8314262 : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/067e84076ca943868cbbf22d96c76b85 : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/64caf60392444972862c12da1df955f0 : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4518002ad52d442f8e8582af266257c0 : SUCCESS in 20m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/483ac7c4a5e84bad9bcc9d7d8dd1c777 : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8835d596ab2043eebce2fc8c27d9215f : FAILURE in 1h 36m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3fd848f859f84901a128c64c4e474332 : SUCCESS in 22m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17b41d734b8942808fe3b1a32c853e0e : SUCCESS in 13m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/70404e2295b64446a1d3e296c949979a : SUCCESS in 29m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/84a54c14f88c43ef825a45fd74a5a7ff : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c712ca8904b49d881f38a145be9eebc : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"46c074faa2883665ff1460ae1855dc012edffca7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-25 10:32:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"602d01be055247ee3af8d6a692185d1fe5e1e8bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 11:54:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c637a18c8a9747108a076ce82ebb3caa : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b8f500705c4e08a33205a72f6a2d3d : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d902ffa37ed4c319c3a460aa462e505 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f731b5d8651f42da841333c29c7c437f : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b82795ccc284c21a51083ae40245486 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c96aba0db2a7462594cdab992898cca9 : SUCCESS in 8m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c29ed08ad6746d9992058401dbf87f1 : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b6115432f6c44d8839914aeadf61756 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c7ecb1ae5dee470b9d7151bb127518f5 : SUCCESS in 17m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/98b14d1bf84744a7b16763b80ddbf7b1 : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da3ebed32dcf40a386c173f08e2d238b : SUCCESS in 53m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/47d4523de0c644b09fbc2e60826fdc73 : SUCCESS in 1h 15m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/426ca8ae5c8e4c29ba7ad669ef20e8d5 : SUCCESS in 19m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4b7c1fecb0a489ab07b1e79593b8edb : SUCCESS in 14m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2765d2c5ff9c4e90b7cb89b6e8af24db : FAILURE in 1h 17m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e0e0441de3424cfc9e48c70b274d9e84 : SUCCESS in 22m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cacf95bdff7347c9945b9cfce642550a : SUCCESS in 14m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5c73174c3af54ee08b416287e9feb908 : SUCCESS in 25m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/53ebdd390453443391820ac952e6b8e9 : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad157579ca3b47c58a03cfc9bcc1782d : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8f264cd73fde6aa3d0b6513c9501e324b6ac0366","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-25 11:56:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c089ede9b12e875d23dd969688e45ed4337787e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 13:36:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7184131560b3456194f21fa5c7e8ede2 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9493a9b266ff458e8f0547f875f19658 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0fb63073a134157bcb7a4c52d953179 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4eb5ecc2de7d44fa95106db9c7348e5a : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9015712a3a3b4224ac2983137d00a198 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/690c6ffed1ac4ea28308f005fdfd5aec : SUCCESS in 9m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/914a66a791574b07950be754f29943cb : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/729452de9f6242fd997be1a55d339234 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4375b5248003495abf26291a15c5424c : SUCCESS in 18m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c82f8a01fe4d4404867967ee56f9399c : SUCCESS in 1h 26m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7105189981f4512851a19d4250d68e3 : SUCCESS in 57m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/79accd29d7a34b8c83acb3cc63b7215e : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/65343071c8424f5399ee1c559e2a36a0 : SUCCESS in 18m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/274f76804da340a4993016e05a48c613 : SUCCESS in 13m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a63c957c48c44fe397bfafab02865247 : SUCCESS in 1h 37m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6bcc1d29bf49477a8e2133e37cd34ec0 : SUCCESS in 21m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac6c103bfbf242bb9c037ac5a82aad25 : SUCCESS in 14m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/833ba2b259bc4f2a9888d5157a49ddff : SUCCESS in 25m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/529e82671b9043d3b862f102317f00ca : SUCCESS in 14m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9db5ee2423cb4150a0435cfc4fe9c496 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f2f1c04412005cd95a876ad1b9d7a81f422a3922","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 13:36:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0439498bf31f0a970ef4208c6db32ab2b82c8762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:20:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6eb81201ce4ec5a6af6ac05efdbfed : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95cb14c9f3d044e7b4d39c98828235ad : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe52978ff4b844eab1f24119ebe49fd0 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21de7d725124bfebce083defe42a65a : SUCCESS in 7m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29ac22e9f00e48c58a39e375bb53af72 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2694db3a2d4d4429a3d374e0fac5ceb0 : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b82f0698f0064dcdb3926035000df10a : SUCCESS in 19m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5a64b778a56d4ec580fff14d6910dad9 : SUCCESS in 1h 15m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0832bac413fb48abb3dacc6ce6cfd598 : SUCCESS in 1h 05m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/71aeba11e3c04de391009e21069c8c57 : SUCCESS in 1h 10m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eb646d856e0e48f2b0ec92a186c06ee6 : SUCCESS in 21m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19bf4d0aa4ba432d803f71e253bfc42b : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/86872915492d46a582b6377218dbff21 : SUCCESS in 1h 42m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/73f443228dc741cf98d5fc080e9178d6 : SUCCESS in 25m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b71e08baca94a2cb9466d50a226894a : SUCCESS in 16m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ec64ef3d01d64520bbd3e91bf5b7d5a9 : SUCCESS in 25m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1687421ad3254b68b2b40f3769820e62 : SUCCESS in 16m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40c32d0cadac4439a3af0f29d26848b3 : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"49396560e5b1930c3885e8c1796e39851e6eeec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:20:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59dead8860304b6c2536365ef10a201375a6be23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:21:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859b09d05d4c402f969e172418f61e06 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"712c41b4cc4c4d2892b5d2ba063283b56d14d63e","revisions":{"712c41b4cc4c4d2892b5d2ba063283b56d14d63e":{"kind":"REWORK","_number":1,"created":"2022-04-22 10:19:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/839038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/839038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/839038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/839038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/839038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/839038/1"}}},"commit":{"parents":[{"commit":"37f8eda4fbf87d039c8e7fcd7140052a883299b6","subject":"Merge \"docs: Fix custom LVM example\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37f8eda4fbf87d039c8e7fcd7140052a883299b6"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-20 14:12:52.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-22 10:19:36.000000000","tz":180},"subject":"Cleanup old and deprecated Swift configuration","message":"Cleanup old and deprecated Swift configuration\n\nThis is old and deprecated configuration settings for Swift not needed\nanymore because kolla-ansible correctly configure glance-api.conf and\nglance-swift.conf since Idddbf2ce741e0486d60e1de88c77a7f0332a5a2b when\nkolla_enable_swift variable set to true. This change can be backported\nthrough all releases till the Train.\n\nChange-Id: I5273edc8265f115519f499e85fe12f8d22306c4b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 10b8c1ead01c5372df40d565ac44a18ea20edc5e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/712c41b4cc4c4d2892b5d2ba063283b56d14d63e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/712c41b4cc4c4d2892b5d2ba063283b56d14d63e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
