)]}'
{"id":"openstack%2Fkayobe~839039","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I5273edc8265f115519f499e85fe12f8d22306c4b","project":"openstack/kayobe","branch":"stable/victoria","topic":"cleanup-glance-conf-swift","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-04-29 09:17:11.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-25 08:25:22.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-29 09:17:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5273edc8265f115519f499e85fe12f8d22306c4b","subject":"Cleanup old and deprecated Swift configuration","status":"MERGED","created":"2022-04-22 10:20:16.000000000","updated":"2022-04-29 09:19:26.000000000","submitted":"2022-04-29 09:17:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839039-cleanup-glance-conf-swift","meta_rev_id":"9f7d3be689f6fc968db971106a91d1df80047de0","_number":839039,"virtual_id_number":839039,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 09:17:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-29 07:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-25 08:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-29 07:10:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-22 10:23:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-04-22 10:23:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-22 11:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 07:10:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9cf16d93c3cec625f6f98d9afae908ab5948715d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-22 10:20:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ca56818c5bb5b15cc569b419ce9175d5a0d93a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 11:27:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cd892b476434b23823619a900f51966 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e948162856d141fca6492dcedfc92c1c : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8083bc55773a41bebe47c065edce8177 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e24551b527642e7aa2354ace1c8a938 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4646e477cd46cb865e626fd83fa6b9 : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a0318c7948e047d493a36730474886f1 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91986eb05bb2427286641e939303f525 : SUCCESS in 8m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e6cb9bf2c1444e69145383d78a4fb1f : SUCCESS in 17m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/883e0fa6142c4706abe5dc0cf537f1ee : SUCCESS in 1h 00m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5e91f2a3778f4924b5031fa4822d511a : SUCCESS in 1h 02m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ea4a7c3e231a420aae94cb23df81db7c : SUCCESS in 15m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/683fe328a5a64229a024e45cd6f1e86a : FAILURE in 14m 45s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8bdca81a6b5e470bbfa557c37531721a : SUCCESS in 17m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba902aa1f55e4f7f843809ca85d40c38 : FAILURE in 16m 09s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/61569d6261f747119a6beb93389fd861 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4bcd505bea05fce627e92d4901343cf3b3fc4fae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-25 08:25:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5134a91dec740504f565466cf67d2727d49bb418","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-29 07:10:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8eef479e19e8ecab85ddda40e98683a8660adde7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 07:11:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b948bf2975548838cc6927c0c5ba87480ede52da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 09:17:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a0aeff32a81421ba40aba9bc9f82a40 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3183570787b543cf802ad96c4c6fd5c4 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4afffb7483914db99774a97edee9716a : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820507fba92147189d11c3c1b131c352 : SUCCESS in 7m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79b4f34106e54929a57a082dd7bcede2 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33606b37cf3844df90dc8eb8f398fbb3 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25276551c7df43a88e64100f13cb1234 : SUCCESS in 17m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e9864b67b9f34e5e967b9aac6177c578 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1d5cab792dc74bb58b670a94a729e925 : SUCCESS in 59m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e186499e03f44b0fb244311a7718ca44 : SUCCESS in 18m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f0e3872ed08448ffb913ae3dfb3b536f : SUCCESS in 17m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/213b685b0b5d4180862303dae17ca27b : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"67c4db88da44dc519e52ad794af2465bf664f194","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 09:17:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9f7d3be689f6fc968db971106a91d1df80047de0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 09:19:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2b486f8fc64f3186512db95a25d6d7 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8173844a7079d0adc4be1ba1012a59521774351","revisions":{"e8173844a7079d0adc4be1ba1012a59521774351":{"kind":"REWORK","_number":1,"created":"2022-04-22 10:20:16.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/39/839039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/839039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/839039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/839039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/839039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/839039/1"}}},"commit":{"parents":[{"commit":"15937084a0d92167bb863a1a5a880a6b621fa745","subject":"docs: Fix custom LVM example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15937084a0d92167bb863a1a5a880a6b621fa745"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-20 14:12:52.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-22 10:20:03.000000000","tz":180},"subject":"Cleanup old and deprecated Swift configuration","message":"Cleanup old and deprecated Swift configuration\n\nThis is old and deprecated configuration settings for Swift not needed\nanymore because kolla-ansible correctly configure glance-api.conf and\nglance-swift.conf since Idddbf2ce741e0486d60e1de88c77a7f0332a5a2b when\nkolla_enable_swift variable set to true. This change can be backported\nthrough all releases till the Train.\n\nChange-Id: I5273edc8265f115519f499e85fe12f8d22306c4b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 10b8c1ead01c5372df40d565ac44a18ea20edc5e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8173844a7079d0adc4be1ba1012a59521774351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8173844a7079d0adc4be1ba1012a59521774351"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
