)]}'
{"id":"openstack%2Fkayobe~839077","triplet_id":"openstack%2Fkayobe~master~Iab33b38b2f2b9f7f2e07ebbd54aa1ffca7aeda4d","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-04-09 17:47:19.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-09 17:47:19.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-09 17:47:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iab33b38b2f2b9f7f2e07ebbd54aa1ffca7aeda4d","subject":"Remove stale configuration files in sub-directories","status":"ABANDONED","created":"2022-04-22 15:40:32.000000000","updated":"2026-04-09 17:47:19.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b7135f1aa5131686cf56608d5cf35bffb692c155","_number":839077,"virtual_id_number":839077,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-29 15:21:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-22 17:30:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 07:50:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-02-07 15:30:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-09 16:19:18.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"2836e6f08e43cc4eb4ad5fc2f2febdd74389578e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2022-04-22 15:40:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc750e89a813d72a760a8720823dbb59c14fba36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 17:30:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac289bf097fc49709b2d1ca26e6433c3 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aed1d9362e874fad9947b5ea19a0a88f : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39e13bf0c3644f4faa72dccbfde3869b : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6449e3c918534f218b27df4382e6e2a3 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0c96a4084a34d7fbcb0999214c6b48d : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d50a0284e84b54b79f3406635dd5e3 : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/762aadc847724bf1b16eaee8e95cddc5 : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c0d948c142794a26ba1eb7162cabcda4 : SUCCESS in 7m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9cd214dde6ab4bb2ae440a100236576d : SUCCESS in 14m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9bf5fdf6ab064ac381872005c2a08374 : FAILURE in 17m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0cc692ffdfbf4bb3bf3a4ac775c68089 : SUCCESS in 1h 15m 39s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/71cf296135474890a93c024a06e91b4d : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/605575e3e26748da8c4396ad5a9ba6d1 : SUCCESS in 42m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5b28b8d986b8463493d9dfa6249ee992 : SUCCESS in 55m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/33f92b776a4043ac88df88a7431036c6 : SUCCESS in 22m 45s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4646c156932b477b86856edbe39099af : SUCCESS in 19m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f7e8a0f7bfa452a98058c986957a205 : SUCCESS in 15m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0f8ade6240a444d79315d5739b5798c4 : SUCCESS in 1h 45m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/876a77a3020e494e8cbd7df4c73712be : SUCCESS in 1h 21m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/47ac13bc40544282b99834f4b2f70664 : SUCCESS in 21m 29s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/46fc312ff58d471a91d674ed98864a0f : SUCCESS in 21m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5065242cefac457e88dd809d75f9625a : SUCCESS in 16m 04s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/8dbf0feab7364bd1b15ef4983384d5bc : SUCCESS in 47m 08s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8b921e70f2ef47d0848366660c66136a : SUCCESS in 40m 37s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef6dbc71984c418a929cbfa940607312 : SUCCESS in 29m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/409bd99556094589a051d6d78d4fb39f : SUCCESS in 14m 24s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/dcc22d901bc84c15a43049b92ea09bf4 : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85340a8741704f21972d5216d98c3d27 : SUCCESS in 13m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2bffe63aa5a042ffb4150640b118f0a5 : SUCCESS in 16m 17s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6bd21633bd7d49d6ae51e7e91956150e : SUCCESS in 21m 51s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c2e2b2c6994a53bf9f7b5b21fbbd8f : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6e3f61cba26091e99691cde67e6817c3d494a954","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-29 13:38:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"100a67bd5515b67a4e3903981bb795f6f71db163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:21:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d248d6f822fe4a44b804d78b044d7ccd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55e7c6064c0849ccaafc92084d511699 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f12e469acbc41cf833f9658fe80676a : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26aaaa24b290448f8d175dddab7e9caf : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c20917c4cde44fab52db7f833c5a5e8 : SUCCESS in 8m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfb7c8d1155842ef9d8fff7803c30339 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b479b11db84593bc4d04c5edef85f6 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b545de2c7ea64ec898d9792df3b829ad : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e358d2d1c02433c9c3594bbd62000b6 : SUCCESS in 7m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/89d4a7aa6bff4e3d8836eb454f0608d1 : SUCCESS in 13m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb8d28f5307a4f4fb9274ef90beb0e86 : FAILURE in 16m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ac68c156576d464b9c308814a6744689 : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/336289b8292940508eed3473ed982ae9 : SUCCESS in 1h 10m 01s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd3dff0375ce4350b1674c2a6154fa41 : SUCCESS in 45m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6c4c45927d4c4a9992c297086d4d0343 : SUCCESS in 1h 09m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e7abcedaf87d425d977f7e1d02add216 : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a8ba473298394638993c21e73ea89c13 : SUCCESS in 23m 17s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2d28f683d8a426f91d4c764e3500370 : SUCCESS in 22m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cf5bfaa0534e4797828e0ac6d1211e6c : SUCCESS in 1h 36m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d1baa78665a4ebc9e0cd55f2152e3b1 : SUCCESS in 1h 09m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2e2447cb1ce143e38c84c2b12751405a : SUCCESS in 18m 46s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/920f0b1daf25499ab4a2ffc8d33ef7a2 : SUCCESS in 21m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b930e55c6184388a5b8c55761485824 : SUCCESS in 20m 43s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/7c08ef65c22b4b168b6f3ee00660427f : FAILURE in 33m 27s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c09b19460627427dab0441aafaf3a5a6 : SUCCESS in 41m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f469e7dfa0a48799424afb8b5f49c0d : SUCCESS in 34m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/25bad675d8d542d4b8cc3f427a674633 : SUCCESS in 15m 55s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/de8b0c48133d484896f3c93b119080bf : SUCCESS in 18m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75a1ce7cfea844c383df2aeb78ba01b2 : SUCCESS in 15m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7985d36d831d4be1b667dbfb6b66ec8f : SUCCESS in 15m 44s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/50f81a1276814797a51d4472ad7cf827 : SUCCESS in 20m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0f8e680e19442e084bda48d35a9f208 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":2},{"id":"349a7f610c5cdfc04f8593070fb5cd62cdf57ce5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-30 07:50:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21359cf77ab88bab2f46508795e92294eb97fd5d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 16:19:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1eefc434cc1fcfb8543a1b7fd753089b7b9aef39","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 17:46:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7135f1aa5131686cf56608d5cf35bffb692c155","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 17:47:19.000000000","message":"Abandoned\n\nSeems to be fixed in newer kayobe. Reopen if this is not the case.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4b60f283ec2eb8dd61f4a7eb53321dcdd828720d","revisions":{"44be38d01de821ac2a1ae031d117f45d6a8eb85e":{"kind":"REWORK","_number":1,"created":"2022-04-22 15:40:32.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/77/839077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/839077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/839077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/839077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/839077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/839077/1"}}},"commit":{"parents":[{"commit":"10b8c1ead01c5372df40d565ac44a18ea20edc5e","subject":"Cleanup old and deprecated Swift configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10b8c1ead01c5372df40d565ac44a18ea20edc5e"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2022-04-14 16:26:27.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2022-04-22 15:39:11.000000000","tz":60},"subject":"Remove stale configuration files in sub-directories","message":"Remove stale configuration files in sub-directories\n\nIf a user places a configuration file in a sub-directory,\nit will not be cleaned up if it is later removed from Kayobe\nconfig. This change modifies Kayobe to remove files in\nsub-directories in the generated Kolla config when they are\nremoved from Kayobe.\n\nA further step would be to remove unused directories from the\ngenerated config, however these have no user facing impact.\n\nChange-Id: Iab33b38b2f2b9f7f2e07ebbd54aa1ffca7aeda4d\nStory: 2009992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44be38d01de821ac2a1ae031d117f45d6a8eb85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44be38d01de821ac2a1ae031d117f45d6a8eb85e"}]},"branch":"refs/heads/master"},"4b60f283ec2eb8dd61f4a7eb53321dcdd828720d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-29 13:38:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/839077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/839077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/839077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/839077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/839077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/839077/2"}}},"commit":{"parents":[{"commit":"51b8b6f43f5ebd1f7ab509b5c5a6103222d888a4","subject":"Merge \"unit tests: Use assertListEqual and maxDiff \u003d None\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51b8b6f43f5ebd1f7ab509b5c5a6103222d888a4"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2022-04-14 16:26:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-08-29 13:38:04.000000000","tz":0},"subject":"Remove stale configuration files in sub-directories","message":"Remove stale configuration files in sub-directories\n\nIf a user places a configuration file in a sub-directory,\nit will not be cleaned up if it is later removed from Kayobe\nconfig. This change modifies Kayobe to remove files in\nsub-directories in the generated Kolla config when they are\nremoved from Kayobe.\n\nA further step would be to remove unused directories from the\ngenerated config, however these have no user facing impact.\n\nChange-Id: Iab33b38b2f2b9f7f2e07ebbd54aa1ffca7aeda4d\nStory: 2009992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b60f283ec2eb8dd61f4a7eb53321dcdd828720d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b60f283ec2eb8dd61f4a7eb53321dcdd828720d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
