)]}'
{"id":"openstack%2Fkayobe~839344","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I6424fc405894514a63fb2b641637bbb9d5c070c0","project":"openstack/kayobe","branch":"stable/yoga","topic":"proxy-fixes-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-04-27 08:06:05.000000000","reason":"Michal Nasiadka replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-27 07:54:22.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-27 09:52:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6424fc405894514a63fb2b641637bbb9d5c070c0","subject":"Fix no_proxy configuration","status":"MERGED","created":"2022-04-26 22:15:56.000000000","updated":"2022-04-27 09:54:01.000000000","submitted":"2022-04-27 09:52:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839344-proxy-fixes-stable/yoga","cherry_pick_of_change":838404,"cherry_pick_of_patch_set":2,"meta_rev_id":"5da63e2046bf1afd7cb3e5946c8b0f70575d8eec","_number":839344,"virtual_id_number":839344,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 09:52:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-27 08:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-27 07:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-27 08:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-26 22:16:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-26 22:16:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-26 23:44:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f97e877e68754d360b8dbc9a445a79945050714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-26 22:15:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6cf315e1ef407fb8307ca1254865e4d0de3b0a31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 23:44:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/854d4b9928f34463a9db5b5ef04f40ea : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e002f29be874820986312237b20083e : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/972be1d0100f404ea15b4bf2ef6cbfcd : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64268df47b79477ea6328e4b1da53ce4 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41137a73c9e447b581c08051924f8b21 : SUCCESS in 10m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90c84b502986482aaac8c7e8f71d4be3 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7151a1114ce44598d85c6750341a51e : SUCCESS in 13m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4d78296342246ecb73fb8e48b211168 : SUCCESS in 16m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0e7cc35447054ea0a1bc8e7d20ead87d : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/47845ca6412546ec9404631e31943b2e : SUCCESS in 1h 04m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6674261b79340e396de544a38ceb963 : SUCCESS in 52m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c27a600495d347a399ee894e32c90cca : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/363cd7b5f4f149c39d5ae39760cabf8a : SUCCESS in 22m 22s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/465ad36d9bc94b5480a16b5e527c3343 : SUCCESS in 22m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df1c1c9b60e54aaa8e7e4ee94242018a : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8be80a464259442ea705675f87ae5046 : SUCCESS in 1h 27m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb84a8a1c89a4785a75847f17a9da79e : SUCCESS in 1h 16m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ef48cd4b5cb04886a53fd809905fc734 : SUCCESS in 16m 21s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/91cce6f66b6a4fc6898592e4915a8add : SUCCESS in 21m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b161b9daa6d54369b27cca927ccaa1ad : SUCCESS in 17m 15s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/d99d0eeb0ff849d88d48cd4555d4ea46 : SUCCESS in 47m 17s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1bcf3b25e79b48d998cd3b56cbde6dd2 : SUCCESS in 33m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e57796709ef94eec9427503799d14a38 : SUCCESS in 31m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/405c55785d974429b85b6b828636ebcc : SUCCESS in 13m 20s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6d1c713876ec496586b808fc3f4532d6 : SUCCESS in 15m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72181a4e420c46cf94eccda63ea621d4 : SUCCESS in 9m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c25dbaa716e74515af6e084e002f174f : SUCCESS in 18m 30s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/080b910c4cab4dc9963d78fa68d90582 : SUCCESS in 15m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b38824988cf444d3a0b1da9c789ee2d2 : SUCCESS in 16m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ee1a861f4c2830dd39420264ba5bf9662f1b52e4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-27 07:54:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5637ee36ab41cef4aa7fa5ddb0890d840fdd5f7b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-27 08:06:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53d1125768a20bd3c53966c38c5a6a2bec22335b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 08:06:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1cfacfd1a0d555e5120b1dc4c63aeb99eb7927c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 09:52:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2db5d5722414010993ec683a1b7ec1a : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af33108d076044289a72a75517bd92d4 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be080ba80b0644d8876c080d73933991 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e9127a262b46b0a6bd488eece67539 : SUCCESS in 11m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cee2b9bec9fe44e4a103098729f09075 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed8c311691434420a4905ceb8a380f1a : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/748c2e9d76b1478cbf770bbd22774d8f : SUCCESS in 16m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2a7726027eaf464296f46b4833244103 : SUCCESS in 56m 32s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/da6ead0b85b9486ead39e9a4018a42d0 : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/815a9f09e57446919c56ecf6ca8f4fd5 : SUCCESS in 46m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/97d799fcd0a44dc98bf9956efcbb4afc : SUCCESS in 1h 00m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/725a17982fdb406783b4437be4c7862f : SUCCESS in 22m 21s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/58b652faa20a4d6bbb37059b6a792fa3 : SUCCESS in 20m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56df79323ca14ebeb7696a9ba7541b3d : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5a636d4628e7460da9171d9aaccff175 : SUCCESS in 1h 37m 44s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c07dd608fa2e4af0b3aaec6e84f4ca02 : SUCCESS in 1h 06m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4b6ae585de3540aabf22d31e5dfad018 : SUCCESS in 20m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0abf6140e2974e1b840e53fe289b2af1 : SUCCESS in 21m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44c40bb4e9614efbb7614fda683daeaa : SUCCESS in 14m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bea32376c1244fc88a08326f6b305312 : SUCCESS in 35m 48s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71492c42ae4d40858306f033b0df54f5 : SUCCESS in 28m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/72ac4b2581a5449eb5909e75d14e7cbe : SUCCESS in 15m 51s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/aa708544a48b4685a20efe7af597317b : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01eaa0d46ac5468da2f1cf349f7bf619 : SUCCESS in 11m 52s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4bf42457053d4d0aa1c4c0b28bb250c4 : SUCCESS in 17m 27s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/670b314c8d2848d7b438a92dbe117fef : SUCCESS in 18m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b77f677b4cf840efa744e8ed3f01afb7 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"08380ca45363da064288e0b0a419819fb9dac82e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 09:52:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5da63e2046bf1afd7cb3e5946c8b0f70575d8eec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 09:54:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/386abc11a0fd453490d7fa3bb6f2e9a9 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd191e7011e23871c50f700e924a59952abfe3d4","revisions":{"bd191e7011e23871c50f700e924a59952abfe3d4":{"kind":"REWORK","_number":1,"created":"2022-04-26 22:15:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/839344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/839344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/839344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/839344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/839344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/839344/1"}}},"commit":{"parents":[{"commit":"bf903dca6ba270ef8e7eac4ae0790eaf08d64339","subject":"Cleanup old and deprecated Swift configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf903dca6ba270ef8e7eac4ae0790eaf08d64339"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-19 06:58:22.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-26 22:15:56.000000000","tz":0},"subject":"Fix no_proxy configuration","message":"Fix no_proxy configuration\n\nThe no_proxy list should only contain domains, hostnames, IP addresses\nand networks, but docker_registry is often in the form ip_address:port.\n\nUse urlsplit to extract the hostname from the docker_registry variable\nafter prepending http:// to turn it into a valid URL.\n\nAlso add missing infra-vms to hosts in proxy.yml.\n\nChange-Id: I6424fc405894514a63fb2b641637bbb9d5c070c0\n(cherry picked from commit 71eb21a3ddab249fe733b56c5f9b70d2947308a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd191e7011e23871c50f700e924a59952abfe3d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd191e7011e23871c50f700e924a59952abfe3d4"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
