)]}'
{"id":"openstack%2Fkayobe~839652","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I9c6e261dbac067ca3f9218e6ea81813ddc2f6520","project":"openstack/kayobe","branch":"stable/xena","topic":"stackhpc-drac-bump-stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-29 09:17:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c6e261dbac067ca3f9218e6ea81813ddc2f6520","subject":"Bump stackhpc.drac role","status":"MERGED","created":"2022-04-27 21:44:29.000000000","updated":"2022-04-29 09:19:34.000000000","submitted":"2022-04-29 09:17:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839652-stackhpc-drac-bump-stable/xena","cherry_pick_of_change":839056,"cherry_pick_of_patch_set":1,"meta_rev_id":"08d91abdd1a78e52e54c1ccff6838b68905cb078","_number":839652,"virtual_id_number":839652,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 09:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-29 07:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-28 08:00:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-29 07:10:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 23:21:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 08:00:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-29 07:10:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ae6fb0b1ba00b312beca05c0e81ae14476486d23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-27 21:44:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"034f3c906e1fb8bf774b45aa0f05a526f1ac4a98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 23:21:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f4aa8a32fea45be8f3404a10630f35f : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09221e4f5a9a457ebc1a81f71c071b3b : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aa6d70b4e574a45a2c0daaccb9faab6 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b30eae35869142e5903da0b1b7993725 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/232b4547471e43e9a69b23f464742b32 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3287db3c17b347f09946625c9d69ba67 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c70179d3c58644e3afa5acb026754a68 : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b3dca18b6ee449e7a408bca848d2b368 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e8121d9ffa04a3ab9040256725243cc : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/916c156be82348179b655415bc4ea8a4 : SUCCESS in 18m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3a1a16af39824087b676005a314b9303 : SUCCESS in 1h 08m 06s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/6b78a16edcbd461e9bb1d34ab4a66534 : SUCCESS in 1h 12m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09a3b485edd147df9b7d44e3817584d3 : SUCCESS in 56m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/02bdbd02c2ae41e7b248b5ec06eb1b6e : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/94803759aa8441c5bc941be93b8ac9de : SUCCESS in 25m 03s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/485d62a4fe1a4c57a35c22b8bea95db5 : SUCCESS in 23m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13f15a2b85a04fd1a12a45470c19d6bc : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/31293244e8cf4282a944d97819659ffd : SUCCESS in 1h 30m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03bdf7ef1796493299d188d1cddf8388 : SUCCESS in 1h 22m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1c06758513cc456ca8508a545ae6e746 : SUCCESS in 29m 08s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/ae100691f4344b1a8592aba956454e34 : SUCCESS in 20m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dab1b5ba6a14b6e9a67912f169bd6a0 : SUCCESS in 18m 49s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/b6fb8d4dca794650b91386c435d1e68e : SUCCESS in 52m 14s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/27a17fbd4d3b4bb7b5026a42a657562d : SUCCESS in 31m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27065b13e017496ab6b24c497b975a50 : SUCCESS in 24m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/091ce9471ea14e8e94d76a6a1ea81b9a : SUCCESS in 17m 15s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f1439612309d4d9f87d44648a8955351 : SUCCESS in 16m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24ff5d96465f4c0cb93546df6e0c8cee : SUCCESS in 12m 39s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e68ad00f0aa547b58b2ac18f40c65d34 : SUCCESS in 18m 30s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bd406434b76e4f3f8ea7ad1a40710a4a : SUCCESS in 17m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e216871c4714bc79e90f346327e1025 : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f1edcba8546c6816e5d2ace836ac856172cda6c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-28 08:00:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c957025f4a862e39c1a903b714085e023e61df03","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-29 07:10:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1de3503803bbaea5b51fa7fd9ec7755f574fb512","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 07:10:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b7191b743e1b5b2bc8ecac51eff0ca5377f9b02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 09:17:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35bd8600e574a79aed3a3c4ae333f2d : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e4a7e5644da4a498ecf7127ca73a729 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c11f7f5e370647a8ba965ac9eefcc44d : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc7ae56d27b4d09871f9227e18f6c4a : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a0dcd34c9314df497328cf993e590e8 : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d9312f53a614f699994f879ea6fcb47 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16760184c88247e388e9c6bc26bc4f38 : SUCCESS in 13m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1963ac089e5440648a54a48f790514c2 : SUCCESS in 19m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5b286f9e940d4d559d8f84a931962814 : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/eea556fb7e744ba4910889131dbaf41a : SUCCESS in 1h 10m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ad416b1659c4779af46867ee93df9aa : SUCCESS in 56m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6889ba132879414eb3b8dd91ff42c9a3 : SUCCESS in 1h 07m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8a907b6724214c6fa3a6ebf690e98414 : SUCCESS in 25m 59s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/1f570338c7db498598d0842adff8bbbd : SUCCESS in 25m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0038e584205a491ca221931776cbffe1 : SUCCESS in 15m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c4a498786f74239bc6df403dc140338 : SUCCESS in 2h 02m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80683b62663a42f18cad219e7f886398 : SUCCESS in 1h 10m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b359ec71f1934cf9a8c83436b01b5bb5 : SUCCESS in 20m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c1cdfdd738d341049bb9b4c8a40dbcec : SUCCESS in 19m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/405191925df04ad4960e43fad9d4acfc : SUCCESS in 24m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/475bb7039d9a433caf10decba0adc6bd : SUCCESS in 31m 04s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/025beb94e03043828f05aa500275e783 : SUCCESS in 31m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/73c3d5f223694af7ad95b05c4fda2044 : SUCCESS in 26m 12s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f4c48247ced8443b9ce0f991b72b3f17 : SUCCESS in 15m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11922b07607344a39f6594f0d047b353 : SUCCESS in 15m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d4333c8b028442578a8b594f9ac335d3 : SUCCESS in 20m 49s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/aa0dc01a7d244e76bdb1e4230fdad2cf : SUCCESS in 18m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c4fe6945882488b84bbfb7bd56dbdd6 : SUCCESS in 22m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1188b6698479984b00fac13172b17ba7e70ef63b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 09:17:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08d91abdd1a78e52e54c1ccff6838b68905cb078","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 09:19:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be19c923a6e458b947f98fc027550f7 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa6185e450d240b2b633c05f8bc0ae2d : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e3c0405cde0dfc2259ddbb70a574050d1c7bdec","revisions":{"4e3c0405cde0dfc2259ddbb70a574050d1c7bdec":{"kind":"REWORK","_number":1,"created":"2022-04-27 21:44:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/839652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/839652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/839652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/839652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/839652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/839652/1"}}},"commit":{"parents":[{"commit":"82193e8ba88136000fd9f2393bdd20bf5eabcc14","subject":"Cleanup old and deprecated Swift configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82193e8ba88136000fd9f2393bdd20bf5eabcc14"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-22 12:18:14.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-27 21:44:29.000000000","tz":0},"subject":"Bump stackhpc.drac role","message":"Bump stackhpc.drac role\n\nThis new version fixes the evaluation of check_mode, which is causing\n`kayobe overcloud bios raid configure` to fail.\n\nChange-Id: I9c6e261dbac067ca3f9218e6ea81813ddc2f6520\n(cherry picked from commit 5d6b9714ff3d84afd02d7923ccd5145dd66b9774)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e3c0405cde0dfc2259ddbb70a574050d1c7bdec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e3c0405cde0dfc2259ddbb70a574050d1c7bdec"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
