)]}'
{"id":"openstack%2Fkayobe~839699","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I9c6e261dbac067ca3f9218e6ea81813ddc2f6520","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-29 12:39:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c6e261dbac067ca3f9218e6ea81813ddc2f6520","subject":"Bump stackhpc.drac role","status":"MERGED","created":"2022-04-28 08:01:40.000000000","updated":"2022-04-29 12:40:52.000000000","submitted":"2022-04-29 12:39:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839699","meta_rev_id":"12e295fce009fbb5fc18cea5fc2fd3a7bc4a4c1c","_number":839699,"virtual_id_number":839699,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 12:39:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-29 07:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-29 10:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-04-29 10:41:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 09:56:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 07:10:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-29 10:41:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5e7653f99f8a2c518a49c18b541002c87f86715f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-28 08:01:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbb7a0332f3876c5a3cf19eb7b00b8563192ff9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 09:56:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5f17231feec47a7a0891318b95603f9 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de7430ec70743d8a1999250b45fb9e3 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/266b93e68ae941f5942bbd5e4497f4f5 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ee9297da35d4cf38fe071973141d85c : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48855a59a51b4ad3b45c7b17e34fccaf : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e825786e35042d5b091e4aa7b3dce9f : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bc95ff9a8754ca8bca1176316332a9e : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01fd0bb214d34b3994f926af1261512d : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96513a369cfa4841a8ae8e3993d393be : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/371b49d5574940a88d9a70df7cf26053 : SUCCESS in 18m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3e4151938bf843dabffa47ef4b59ce29 : SUCCESS in 1h 12m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c32d99a63ab24c48834655c9bbd09d31 : SUCCESS in 53m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/55fcac36fb914a86a71413492014a88d : SUCCESS in 1h 14m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/99988df0e76f4cdeb7a64005915f3aff : SUCCESS in 18m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61bba5b933654618ad83066e78bd544b : SUCCESS in 16m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4b291528cca9415bbae30604c96c0a30 : SUCCESS in 1h 49m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e6359161cc51475fabf4764aff37405b : SUCCESS in 23m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72d5cb01c83a44ce9526f1dd7b7d57ca : SUCCESS in 12m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a362246b180a4f3b8eaba74c4330df22 : SUCCESS in 27m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/804fd11698c849a59e175ca7223bbe6b : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23eb926042c841389583612255d6b473 : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e9e4da6b0e2bf3c576a84ca7d6b75ad95eb428c9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-29 07:10:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e09d23ffece6a88b98e9cc89d752951e42a8470f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-29 10:41:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5532dfdcb81292c98e9bd2deba2ecb3e48b689f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 10:42:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7625db5593590621b98d66fb7b3044b981fc304b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:39:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4590e904114db7ba589535eb13b8ca : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61009ebb568d4400958026396c2cf945 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fbd0747888c479d9af965e3e62b0235 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fefaf37c0dfb438bbf28076f7728049d : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b51d477ec4d458186519e36ec1d7514 : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e54fdd38d6d04e61b139d3906ac45f77 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66fe5e6ffae34ab49b8c31c5db5ca27b : SUCCESS in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f8dc6e9e775746849eda203801a5cb00 : SUCCESS in 20m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d423530e582b4b03848f64bb32821cf0 : SUCCESS in 1h 17m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e964bde7c19c4c5ab266112300a4aeb4 : SUCCESS in 58m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7f4d8b4cee5d46368f924c4ad10b7fce : SUCCESS in 1h 13m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f9876266f21a4a6583aa83798004f484 : SUCCESS in 21m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edfc5bc1c03145b3abff544c8a85192a : SUCCESS in 13m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f45257eac53b47b8b964ece0b8ab1a3f : SUCCESS in 1h 51m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/23e58bacc60b4661bec20a93a0e83979 : SUCCESS in 23m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c8b5960c0df48d59d29d8372fb24ae6 : SUCCESS in 15m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/54079d8a4e32423cb25d0524e265f0f4 : SUCCESS in 25m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9d349bc7742c4e16a53c4d1776b8f78e : SUCCESS in 17m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c77be16b7cf44bd4be7a06ff2374d764 : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"09c1122559c7cc3f987b4c6c068be0c0751a6522","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:39:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"12e295fce009fbb5fc18cea5fc2fd3a7bc4a4c1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:40:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/579185059d4047ef9665d33638f7ea92 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08f030bcec43429daeb7c0902bf8440c : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0988335b1aa81cefd1368d1e6f6faa346de7bc6","revisions":{"c0988335b1aa81cefd1368d1e6f6faa346de7bc6":{"kind":"REWORK","_number":1,"created":"2022-04-28 08:01:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/839699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/839699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/839699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/839699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/839699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/839699/1"}}},"commit":{"parents":[{"commit":"712c41b4cc4c4d2892b5d2ba063283b56d14d63e","subject":"Cleanup old and deprecated Swift configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/712c41b4cc4c4d2892b5d2ba063283b56d14d63e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-22 12:18:14.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-28 08:01:30.000000000","tz":120},"subject":"Bump stackhpc.drac role","message":"Bump stackhpc.drac role\n\nThis new version fixes the evaluation of check_mode, which is causing\n`kayobe overcloud bios raid configure` to fail.\n\nChange-Id: I9c6e261dbac067ca3f9218e6ea81813ddc2f6520\n(cherry picked from commit 5d6b9714ff3d84afd02d7923ccd5145dd66b9774)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0988335b1aa81cefd1368d1e6f6faa346de7bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0988335b1aa81cefd1368d1e6f6faa346de7bc6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
