)]}'
{"id":"openstack%2Fkayobe~839744","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I3f499d63d19ba7f7372b401bd2da23ce627f18e5","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-29 14:47:25.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I3f499d63d19ba7f7372b401bd2da23ce627f18e5","subject":"kolla_passwords: add no_log for password overrides","status":"MERGED","created":"2022-04-29 10:01:36.000000000","updated":"2022-04-29 16:51:12.000000000","submitted":"2022-04-29 16:46:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839744","cherry_pick_of_change":837553,"cherry_pick_of_patch_set":1,"meta_rev_id":"9a28fc8cbde89e90310f261251346f81f14a0746","_number":839744,"virtual_id_number":839744,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 16:46:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-29 14:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-29 14:47:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-29 14:47:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 10:01:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-04-29 11:52:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 14:36:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"7fa7b3ea9ee359d083fdb0a8380b50ff9563f8ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-29 10:01:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b0a67038cf094b194e09dfb63d2db10f0f9a2d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 11:52:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c4919ab226b4d1ea0d18d1ba06b6f52 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/395a0d6a4a614bf4982181c65f5899a1 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c4509f8e8674c1abfd7b59b29cf5763 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5349548315448e78380da90c9c39488 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0265be35d309487c8e3e0582eed1ce6b : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a42fef7007114e40a213788705fea18c : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/acea97df5e4b45a6b01eae5f4e8d4474 : SUCCESS in 7m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8847f6f9a8314b35aa4a7369d4b02e88 : SUCCESS in 15m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fc1136b92e3b46e6a60cd676e8eac801 : SUCCESS in 17m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/34d3d1e25d6246eba9551535d122fdc3 : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b4a1eea5d1c7410ab0bc5c22de6bd517 : SUCCESS in 1h 17m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e65c56963a604e8aae4026c0e8a56bdd : SUCCESS in 51m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8c48d6ffc39840b19c1066490ca54d4f : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8f8ecfb2df054038bb0f9dfbb3822069 : SUCCESS in 24m 16s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/30c3fbbf727f44079ceb15c766ce1d95 : SUCCESS in 21m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f49f23218f3d4b868becdf663551caf7 : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/948c4e7853af4713aaa16da63e1c4c8d : SUCCESS in 1h 45m 47s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b233e9bf9f34420695ae477f4359c632 : SUCCESS in 1h 20m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d8a253c67928404a8bd08153323436ec : SUCCESS in 19m 16s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8693805a926d4fb2843e35e8f8d68a74 : SUCCESS in 23m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35e5eeaa1bd74c7981b2f886d726742e : SUCCESS in 16m 36s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/c0b16f167d5047e3b384a4c5020c3381 : SUCCESS in 43m 38s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/049b05376701425d9cf69561a6bd8ac9 : SUCCESS in 40m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceb76967450d4c0d9614ec0b21379212 : SUCCESS in 30m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a578c3bb771b4b77a637960689da3850 : SUCCESS in 15m 22s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c7c6716835a046848b3bf8d28f856474 : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46e310528e764fa88ba8270a8bddbcaf : SUCCESS in 12m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bb6595e875ac4173a72113d2e9830e9f : SUCCESS in 18m 10s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8d2d2817a0cf4f8c8e115da770902c26 : SUCCESS in 17m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e815241c281c44d79038c93776a03a35 : SUCCESS in 14m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f1043ac3a2700145df6d02c76419f16759d7c004","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-29 14:36:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0836836c044cc8f4744b18064792174f0bf26282","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-29 14:47:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"475f99c257610742bf6a1f6fafd37f60c20d6527","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 14:48:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05722763ea3a8f5375ca1bd77c191840a71432d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:46:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c72ca799ed54d39ba4c3f0084920c9f : SUCCESS in 9m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bac1b865c0b44829b2a31739105c766 : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0284304d99f423ea5511732df7e4cc8 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb2c1d012f84a3d8c9a44806d890d52 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c38ca3f9af54935818e9bf5bee3ae57 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c9a761f96f542c9aeba11188cbd7405 : SUCCESS in 9m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3c614a819fb45bbb6bdce96b94b9c1b : SUCCESS in 13m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2eef129b3dd3432aaa9967bde9ce5151 : SUCCESS in 22m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/535aa41500b148e3beff14638c8dd9d4 : SUCCESS in 1h 04m 37s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ef0b3f61e2a847ac8d01f97b3ab3d708 : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce4f13c4b9064c9e8392c275dc98f64c : SUCCESS in 53m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4021b78acb2f4bdeb271016757d8cb92 : SUCCESS in 58m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/21582de8a16b4b9d98f7805ba7b5a63c : SUCCESS in 23m 03s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3ebc6f9835ac44d48458505c476aaa77 : SUCCESS in 21m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5192f8f78182448ba2de4007169ee7f4 : SUCCESS in 16m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/210168829aee42dc9a1fdb21d618fdcb : SUCCESS in 1h 56m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa1e512dea494da3be8df2a19c5cd0b8 : SUCCESS in 1h 18m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/298bd64bb580473fb77eac8681633335 : SUCCESS in 21m 16s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9e7adca7be7c448a98c81f2a5a6e4350 : SUCCESS in 20m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9278390dc1e64ac9b16e4a7335854592 : SUCCESS in 16m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/428d9b5394d3403788fd977cf0c2e0b0 : SUCCESS in 41m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4004d34233741228898093129c9ed8d : SUCCESS in 31m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/de0710999dd64e148e7009bcbf1e3d0b : SUCCESS in 16m 33s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4caa68699d9547e98c4a4ca850867b88 : SUCCESS in 15m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b44ebf12740c4b68b5d6a43e576cfe49 : SUCCESS in 12m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9d6102c44145408e8239fba76d169023 : SUCCESS in 19m 12s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9fb056ae44b5420689a4c73b0e1c41bf : SUCCESS in 20m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71f96d3c59954ef6a41dff482a7fe86f : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"258d7fa2a17d5841748db08b16837d722ca4b738","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:46:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9a28fc8cbde89e90310f261251346f81f14a0746","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:51:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1940723eb8f447198a19d9b49bf45e9d : SUCCESS in 1m 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8bad621b2c14c079ef5a8fd12bad63b : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55f2aed3f251d67f84b4615f3ae58b693cec7468","revisions":{"55f2aed3f251d67f84b4615f3ae58b693cec7468":{"kind":"REWORK","_number":1,"created":"2022-04-29 10:01:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/839744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/839744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/839744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/839744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/839744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/839744/1"}}},"commit":{"parents":[{"commit":"a54725cff755192b7500af4469e88aa5e607e8bd","subject":"Bump stackhpc.drac role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a54725cff755192b7500af4469e88aa5e607e8bd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-12 10:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-29 10:01:36.000000000","tz":0},"subject":"kolla_passwords: add no_log for password overrides","message":"kolla_passwords: add no_log for password overrides\n\nThe kolla_passwords module overrides parameter may contain sensitive\ndata, including passwords and SSH keys. It should be protected via\nno_log. Without this, the parameter value may be exposed in Ansible\nlogs, or if level 3 verbosity is used, Ansible output.\n\nThis change adds no_log to the parameter.\n\nChange-Id: I3f499d63d19ba7f7372b401bd2da23ce627f18e5\n(cherry picked from commit 51a57394bee0aa98365d1fc9f441a505bdd72b27)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55f2aed3f251d67f84b4615f3ae58b693cec7468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55f2aed3f251d67f84b4615f3ae58b693cec7468"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
