)]}'
{"id":"openstack%2Fkayobe~839745","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I3f499d63d19ba7f7372b401bd2da23ce627f18e5","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-29 14:47:33.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I3f499d63d19ba7f7372b401bd2da23ce627f18e5","subject":"kolla_passwords: add no_log for password overrides","status":"MERGED","created":"2022-04-29 10:01:44.000000000","updated":"2022-04-29 16:50:55.000000000","submitted":"2022-04-29 16:46:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839745","cherry_pick_of_change":837553,"cherry_pick_of_patch_set":1,"meta_rev_id":"285c2518c257cafbcb972bb76b7cb6bc27dcc5cd","_number":839745,"virtual_id_number":839745,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 16:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-29 14:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-29 14:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-29 14:47:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 10:01:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-04-29 11:42:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 14:37:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"14d2d239b7b58d97c35bc59276d9e989d657a537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-29 10:01:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"caf5fdaaea21ed0e2dd4e0fef252f596eff2f126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 11:42:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f707fca46e374dd199b0ee4af93015c1 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a5d243ffb641239eed8ef1cd1795c3 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c705fb90e23b4c90a958039dd31f6799 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2149d83d6c8d4e00949a6303dba90fd8 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/903c924e4f394a3dbe5e381b81ee1238 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c3e938d01a346b98134104b9903dd8a : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f596400b4b3847e988b8e3d6e1be74b9 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/308ee706242e467a9aef380b3f08e258 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a2972f33a2bd4f1a9dfa54222ae55b23 : SUCCESS in 14m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fdf5222a81f24fe09f7cf6d65c103b62 : SUCCESS in 20m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9cfec00d491c4b95b2b09be4d33b6d2b : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/36c7d4fa019a49e29d1dbece85a09dbe : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/217ea65590574d5badc969828d76e984 : SUCCESS in 1h 02m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6da498903797433893ec034eeef57a27 : SUCCESS in 1h 09m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1e611f5e0ce8468ab4836d20d9111769 : SUCCESS in 27m 04s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/fe2e679cbb6641f1aaccc757b548c6ca : SUCCESS in 25m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b611686183eb454aa4d0ad93f130a899 : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b7a75be4370e4fdd8d2487bd01d25d85 : SUCCESS in 1h 38m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db53039c103a4745a17bd9d8234d004f : SUCCESS in 1h 13m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/321f3f5daf2445d48d1e91b976813b35 : SUCCESS in 20m 48s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/234bbd5508bc4e3688ac735ad2141013 : SUCCESS in 20m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc30f0ffb457485fa56e287b6987cf11 : SUCCESS in 16m 21s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/13153181675b4e7bbeac8c6478f0af65 : SUCCESS in 1h 14m 31s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/47df4cee3c524590a40b7e7e80e07465 : SUCCESS in 29m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/950bba24a38648fbbce9283386b5a281 : SUCCESS in 46m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9f44746995a24fe3b2fd2e501bf79378 : SUCCESS in 14m 23s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/59e5eeda84e64ba6a5b25bd8f56d99bf : SUCCESS in 14m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7d69323b969497084fc18d99f158ac2 : SUCCESS in 25m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/29be6c3a7d41450790ced5ceda250080 : SUCCESS in 16m 45s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/eb9367f167a84c3f8d28e2880525e67f : SUCCESS in 16m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84ba61f2d3f04f098ce7860050744985 : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"83966a8addd0646226482c23d208caedb34fb850","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-29 14:37:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a8e2a916ce3e216d864f597c06e48508dfc2c53","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-29 14:47:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"434c8a18a6f036254690bbf6df4989678cfe4ac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 14:48:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"119bf73edab24809b3f8e8848ac57196bde6d338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:46:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9b9c850d83b48909ceb65baa74ff5cd : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e75611fb733a44dabbf526d1978aca61 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/729e1855dc384dc18a5337b3f7184d3a : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85afd1396f1b4f9fb283e27db26f2c1c : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acd17aa1f4a64b6591901b175a8362e4 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d61de66baea4a10bdb4c54e97f0a9d8 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/47700be287564f6f9dec00a2e2b0dc82 : SUCCESS in 16m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eee86b9c2270469e8cbef2e211715ef2 : SUCCESS in 19m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e4c139748bf149ef953f9dfcb51d6c74 : SUCCESS in 1h 28m 23s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/53212faf49804ee38f6baef3e5a09a3e : SUCCESS in 1h 15m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f266ab8153314effb146b93d159b17ef : SUCCESS in 54m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5db1d40f70204ac1bed051a4ce3ae331 : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5b416664ab4d423aad52705e82dfcc70 : SUCCESS in 27m 13s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/eb8e52208e07406b92a902b3936fcfd0 : SUCCESS in 23m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/678db2d8cc5f4a429cd99046461d4faa : SUCCESS in 17m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8eee94a8e079479cb0e5316b28fca50c : SUCCESS in 1h 47m 17s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31ced251a0cb4b02a600d0cbf438ab56 : SUCCESS in 1h 19m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dc46e3ef88b14833984a575099eaad3c : SUCCESS in 21m 10s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/23e052d7178f4bd7a94dd079d6256bc3 : SUCCESS in 21m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35554841de494decabcdda0fb6f940ac : SUCCESS in 15m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0e500ee16fb64da1b6af4cfc08841bfc : SUCCESS in 28m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42dc1a7892a44931a89733f4488323b2 : SUCCESS in 27m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7559c9cafc0244d99be76af21cbf54f9 : SUCCESS in 17m 01s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/950a3835b67446debfbf12607ab8726d : SUCCESS in 15m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edc119cc01844c938dac49601d8b0475 : SUCCESS in 12m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/27c201c22809445ca192d2ee337acc82 : SUCCESS in 19m 23s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/39ea74b3a6ab4028aa8c3c20db1b029b : SUCCESS in 16m 43s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/464994ae3b48446ea82d5508d9455b64 : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"21c22c2456b03884acf6ca5d5108f4f82055f488","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:46:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"285c2518c257cafbcb972bb76b7cb6bc27dcc5cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:50:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31303b7375840a4ba24a5ad4cf88102 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9f8bd91942549e0aa3d122ed7e11bc7 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5fe785215877f8051fc409394c42f0bc15eabe8","revisions":{"d5fe785215877f8051fc409394c42f0bc15eabe8":{"kind":"REWORK","_number":1,"created":"2022-04-29 10:01:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/839745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/839745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/839745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/839745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/839745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/839745/1"}}},"commit":{"parents":[{"commit":"4e3c0405cde0dfc2259ddbb70a574050d1c7bdec","subject":"Bump stackhpc.drac role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e3c0405cde0dfc2259ddbb70a574050d1c7bdec"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-12 10:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-29 10:01:44.000000000","tz":0},"subject":"kolla_passwords: add no_log for password overrides","message":"kolla_passwords: add no_log for password overrides\n\nThe kolla_passwords module overrides parameter may contain sensitive\ndata, including passwords and SSH keys. It should be protected via\nno_log. Without this, the parameter value may be exposed in Ansible\nlogs, or if level 3 verbosity is used, Ansible output.\n\nThis change adds no_log to the parameter.\n\nChange-Id: I3f499d63d19ba7f7372b401bd2da23ce627f18e5\n(cherry picked from commit 51a57394bee0aa98365d1fc9f441a505bdd72b27)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5fe785215877f8051fc409394c42f0bc15eabe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5fe785215877f8051fc409394c42f0bc15eabe8"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
