)]}'
{"id":"openstack%2Fkayobe~839746","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I3f499d63d19ba7f7372b401bd2da23ce627f18e5","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-29 14:47:35.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I3f499d63d19ba7f7372b401bd2da23ce627f18e5","subject":"kolla_passwords: add no_log for password overrides","status":"MERGED","created":"2022-04-29 10:01:52.000000000","updated":"2022-04-29 16:51:13.000000000","submitted":"2022-04-29 16:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839746","cherry_pick_of_change":837553,"cherry_pick_of_patch_set":1,"meta_rev_id":"b515d8f5d36ee76d8d49057aabc85f3a6ab2a62e","_number":839746,"virtual_id_number":839746,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 16:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-29 14:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-29 14:47:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-29 14:47:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 10:01:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-04-29 11:53:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 14:37:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2b36a472eac6cb1dc60b7bbe06a261f00e3ba4ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-29 10:01:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2d7d04cd3d396fe50daab7772bc5df86626337ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 11:53:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b8c43f4feb34572b7f66d2b8d440abd : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ac16a82936e4e1f9fdeb43bfe79fb26 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/291c9a578b05433f961876ed5be3675f : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b60993a43564b27a2540cca655c6a4e : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebf5b76a224f4300be199827b71be9df : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6846602b5f3a47e4aa41fafe41619a29 : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c6d2165a9f840ccb12fb33fe4748809 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/db15fd16798f4166af81c73a8883b39c : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/01920e3e6212442282a3c057cf6f834e : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2122674e42d4e23b9b3af831048474d : SUCCESS in 18m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/103c0ad1e0da47daa57d7e40b8ad644c : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7ee6ab30b5e49428b253fd30a02ec9b : SUCCESS in 59m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0df956f6f1f34544949d0fbf5b99e451 : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2e411165af3b4ac39d451b12f630a2dc : SUCCESS in 15m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1947a5c5765b4f0c8efa9916f7f76f70 : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b712278e0e874d2897aff2d9378d780b : SUCCESS in 1h 46m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f033eb4af1a14f89b8cfa062a80471ee : SUCCESS in 22m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b21c2169d6b343aba6f5e372e080a661 : SUCCESS in 14m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/be538b98307448649b2a42fc43d82420 : SUCCESS in 24m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3ab3c3e9ee174f3cb66ea9ff18dbd2c5 : SUCCESS in 15m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac159c75819a4f9f92bcc0ca25ca288b : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4e14a771d885bec93004b9ad0c321a9684012cab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-29 14:37:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab8e9653e424bd4d7740cacb3b95bbf3e9930aaf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-29 14:47:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c4612ce0dccc9e66a755b880777d9be46f3284c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 14:48:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a593b7d43b46023ae4773d08b4a51ea4834086a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:46:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62581f6823cd400eb7e6de3c67ef2278 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe0c9cbff53c41ca914dbc032821cd04 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4ad4664647042febfb518af55faad62 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae1868ac0ba49d1b34ce5384ce67692 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0ca3331fe534686873ddd668ff1581f : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa163a0e0e0d4e4cba622089beb8802c : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c76f959ba33a43009dbcdd84b0aa48ed : SUCCESS in 12m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fa4465eead340fdb6df0584e68d867e : SUCCESS in 18m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/584c759b7ec245d48b087ba14ec979cb : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7379acbf6a04a4096e3e50f5a5b2774 : SUCCESS in 1h 12m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0fb50d3c1bf142a485ebcd2f9cd7ca69 : SUCCESS in 1h 14m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/44cdf54cbf9f48a8bc040681de17d4d0 : SUCCESS in 20m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99a265689d484ca4b00a6ba4fbf63b96 : SUCCESS in 14m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/73801dcc067849e9bba00b45c282830a : SUCCESS in 1h 40m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a8e83d46d6cb44e8b73a060ea3598899 : SUCCESS in 22m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d049ff6c05b141c88a022409c5605aa7 : SUCCESS in 15m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/35dcbba655ed4780861b1ce82f40ddd0 : SUCCESS in 26m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f8f8361282db49f3a7ed66e10942d858 : SUCCESS in 18m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1081a319c0944f1cbbb21bbe01fa9d91 : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c846ef7b2609a1a08d3dbe9857ef51d3bd73c1fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:46:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b515d8f5d36ee76d8d49057aabc85f3a6ab2a62e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:51:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da603fc0e49456fa0a8848ad0c66f89 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2022b09d374c47d0b84e691a92e6b14d : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3beff4ca461728043e7660cda63cdd14c60543d3","revisions":{"3beff4ca461728043e7660cda63cdd14c60543d3":{"kind":"REWORK","_number":1,"created":"2022-04-29 10:01:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/839746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/839746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/839746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/839746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/839746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/839746/1"}}},"commit":{"parents":[{"commit":"712c41b4cc4c4d2892b5d2ba063283b56d14d63e","subject":"Cleanup old and deprecated Swift configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/712c41b4cc4c4d2892b5d2ba063283b56d14d63e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-12 10:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-29 10:01:52.000000000","tz":0},"subject":"kolla_passwords: add no_log for password overrides","message":"kolla_passwords: add no_log for password overrides\n\nThe kolla_passwords module overrides parameter may contain sensitive\ndata, including passwords and SSH keys. It should be protected via\nno_log. Without this, the parameter value may be exposed in Ansible\nlogs, or if level 3 verbosity is used, Ansible output.\n\nThis change adds no_log to the parameter.\n\nChange-Id: I3f499d63d19ba7f7372b401bd2da23ce627f18e5\n(cherry picked from commit 51a57394bee0aa98365d1fc9f441a505bdd72b27)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3beff4ca461728043e7660cda63cdd14c60543d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3beff4ca461728043e7660cda63cdd14c60543d3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
