)]}'
{"id":"openstack%2Fkayobe~839748","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I3f499d63d19ba7f7372b401bd2da23ce627f18e5","project":"openstack/kayobe","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-06-06 11:19:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3f499d63d19ba7f7372b401bd2da23ce627f18e5","subject":"kolla_passwords: add no_log for password overrides","status":"ABANDONED","created":"2022-04-29 10:02:07.000000000","updated":"2022-06-06 11:19:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":837553,"cherry_pick_of_patch_set":1,"meta_rev_id":"637c406af9af1daea044e51cdce33efb707ceb7f","_number":839748,"virtual_id_number":839748,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-29 10:34:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 10:02:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-04-29 10:34:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d62d5776c4e148e1907c376cbcdc63582a48ed03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-29 10:02:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3afb1281f0c48a18f838a8a2ca9caf72c108745e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 10:34:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b4e38679a224a66a758939039b3ddba : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7184588805354fde8d58d124688a31b3 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/530e2e3d6fdf4a1388b2bb43120d549c : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6acb534d7ac5449a8c18cfbe8d909c6a : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9547fc06e0554669810702d5f1d12ccf : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68297ded2cc5440b837c4aefb0546610 : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cbdba19e51e14b498724675be2428979 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/210b8cc2e27f480eb6d2ba28d0abe5d2 : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bd9d22d0a57c412b89bbc646980ba69f : FAILURE in 7m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fe27041bd6924968a2ea994c48360177 : FAILURE in 13m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/103f32d3ae854d508f53aac5bcd43a57 : FAILURE in 10m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/731fc85ee0ad465a9f6f95d64d94d525 : FAILURE in 12m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1253e07645fd4cc1946d60cb14e04525 : FAILURE in 19m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3248a95ebb494085bec149933a5b12f0 : FAILURE in 18m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ba5b526ef93e4b5b8b8c70802acf91a8 : SUCCESS in 11m 22s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"637c406af9af1daea044e51cdce33efb707ceb7f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-06 11:19:07.000000000","message":"Abandoned\n\nUssuri is unmaintained.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18a15055c1ad4c973b1c5a7095b70b5d93273d2d","revisions":{"18a15055c1ad4c973b1c5a7095b70b5d93273d2d":{"kind":"REWORK","_number":1,"created":"2022-04-29 10:02:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/48/839748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/48/839748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/839748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/839748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/839748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/48/839748/1"}}},"commit":{"parents":[{"commit":"01708a9a9c254d224c477c6c904294180ad3962b","subject":"[CI]: Remove CentOS Linux 8 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01708a9a9c254d224c477c6c904294180ad3962b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-12 10:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-29 10:02:07.000000000","tz":0},"subject":"kolla_passwords: add no_log for password overrides","message":"kolla_passwords: add no_log for password overrides\n\nThe kolla_passwords module overrides parameter may contain sensitive\ndata, including passwords and SSH keys. It should be protected via\nno_log. Without this, the parameter value may be exposed in Ansible\nlogs, or if level 3 verbosity is used, Ansible output.\n\nThis change adds no_log to the parameter.\n\nChange-Id: I3f499d63d19ba7f7372b401bd2da23ce627f18e5\n(cherry picked from commit 51a57394bee0aa98365d1fc9f441a505bdd72b27)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18a15055c1ad4c973b1c5a7095b70b5d93273d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18a15055c1ad4c973b1c5a7095b70b5d93273d2d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
