)]}'
{"id":"openstack%2Fkayobe~839883","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~Ie1d5ecf32c637b77757e56fbe2fae1ff7c0bf000","project":"openstack/kayobe","branch":"stable/yoga","topic":"bifrost-dhcp-range-netmask","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-04-29 14:54:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1d5ecf32c637b77757e56fbe2fae1ff7c0bf000","subject":"Fix Bifrost inspection through DHCP-relay","status":"MERGED","created":"2022-04-29 09:33:36.000000000","updated":"2022-04-29 14:56:14.000000000","submitted":"2022-04-29 14:54:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839883-bifrost-dhcp-range-netmask","meta_rev_id":"a7572294b5255579049014e66a114626ded68c74","_number":839883,"virtual_id_number":839883,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 14:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-29 10:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-29 10:01:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-04-29 10:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 10:01:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-04-29 10:50:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-29 11:33:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15eee17c33cfd5d0f8eea0f4e5180e2d44dd917e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-29 09:33:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70d763eee4490f9e0d9756213eea3fc0cb7f02c4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-29 10:01:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b52719718cd66f4a229245598d0d319ad7c44a38","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-29 10:50:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3037e368019b23c4fd9d394668ad37314c8b6cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 11:33:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cef99701f5374d4491fff44965a6f723 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/722205c485cb43ebbe2edae5be912d71 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9964fe9f2dce438880ab6b3c7a77b49e : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cbbe37bf4964e6caf34ee34a74f7c4f : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc30b1ffd5024a03b250296c6bcbaf1d : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e8eb7d234d641f380644cb6c7f932cf : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eeef3e17b96047a3b797829c8b60b5bd : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/40e9e2f762354a9eaa01a98b2b19ff7e : SUCCESS in 13m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c0ed0a9f7a88402b903f6ed4a7b382d9 : SUCCESS in 18m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5d27f7fe312b48b5b348c6068455f764 : SUCCESS in 1h 08m 14s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/db0d0b4080bc4523b1eaf69d36190062 : SUCCESS in 1h 01m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/326e848d20cb49b3ba35cc8c53e93546 : SUCCESS in 55m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/74125bbaeafb4d1b97aefbbab8de0334 : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/03c6bc6157a34b76874aa0a9ec285a65 : SUCCESS in 23m 08s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/8bfd71bc0be64e2497f352cd9c1a9a68 : SUCCESS in 22m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bff68eeb14c041fdbf0215d6a9de4bb2 : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba5bfc785162426091feee4782888525 : SUCCESS in 1h 53m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88f5a6cc49884cb59cbccdf4c95e2902 : SUCCESS in 1h 16m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0fef1379e1da40e4b8f602816a8130b1 : SUCCESS in 21m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9eeb05b363be45cea8dfd835ee1fcef7 : SUCCESS in 19m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28523179a8604223abd4ad4e21ba1c87 : SUCCESS in 17m 00s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/31cfd3989d2c4760a27675958188ebfb : SUCCESS in 47m 12s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/40e0f0c612b14577a03a997795594ae2 : SUCCESS in 40m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6989e225ca2a4434b6d1a6a98f2a2948 : SUCCESS in 31m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a490e4230df049a5afd38ed3ffa92d17 : SUCCESS in 11m 46s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/33299095530a4e2198825000ba5b59bf : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f21a65c322834a1f8c88de550be50dfc : SUCCESS in 11m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2babe521f5854be19b44e6aa01ec4f2a : SUCCESS in 18m 03s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/98b6152a02d849ccb28d57824020d6ce : SUCCESS in 16m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5ed8215ef5b4fa487486674dedb9792 : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"59fc5b7945f17cae023583613026e6caa97306ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 11:34:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5615db57f8d79beb8f9fd6b830e4ad85611e2672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 14:54:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ddc7ee476764a769dd4566d7ed37dff : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/034ebbc1b0c140dd967bd7b2646746bb : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfcce7b0e48d4f20b0cc28357254c703 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7df6f457e69d4f1b95e66c3cdc47d266 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33eabd47f8774eefb2e986a18eaa9b70 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/636e9a86b220459a88f24f3b47c12a73 : SUCCESS in 8m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81d6a97d60ee42a2b4252f387c3609a0 : SUCCESS in 17m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db4f9afe6fd34eafa48bc71da21ad238 : SUCCESS in 20m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7c8f02c100b640c98cc28824c5e4a519 : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d5381b8698f64b6eb2b282fc7249ea16 : SUCCESS in 1h 11m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59b767f2518b45e1887d6853eef8ad7c : SUCCESS in 53m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f1d59ebe058c4f43b94f521d251e70bc : SUCCESS in 1h 12m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/846a5b0037104c3b8ba6fd11035b01c1 : SUCCESS in 23m 06s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/281775eb84684075a6047892592c8fa4 : SUCCESS in 22m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63a8875743354ee59fc79875d6f37675 : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9f41f2b938b147e2b9bc4b736491371c : SUCCESS in 1h 53m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b66a83c96b94e37ab2d0820d8c04e4d : SUCCESS in 1h 19m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/325c9faaa1d54367bb5b7a199dc73cf7 : SUCCESS in 20m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4d07639cf7c04b23a66fcbcd4614be2e : SUCCESS in 23m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82ed62b3861541fca8a91719c4277f1c : SUCCESS in 20m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cd6c81f1801849589ea37c2db3eb9b4b : SUCCESS in 41m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7445bcf64c14205a478f8713901629c : SUCCESS in 29m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fb40f8e87ea54786a9291b58440ddcbf : SUCCESS in 18m 07s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a3782d0fd6ab47b3aa7876a0acb99f32 : SUCCESS in 15m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3003ddcf90954f13a0bee89f9deb5c24 : SUCCESS in 12m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/65239c4e2f4e4eeaa553a361944a5baf : SUCCESS in 17m 44s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/37fdb1d664a5433fa462b229844bb036 : SUCCESS in 17m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/198d9b33be7c4729a21dd8f0bd1fab08 : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0058441da861d9dd5a9c1a66442e8f99585324c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 14:54:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a7572294b5255579049014e66a114626ded68c74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 14:56:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c5939196fd44f668fd4ec27b07d3230 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b70a33d018574695b4e75838889fe1fc : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac63fac1ba2c82aa8667caacb8b8ed18870cbb6e","revisions":{"ac63fac1ba2c82aa8667caacb8b8ed18870cbb6e":{"kind":"REWORK","_number":1,"created":"2022-04-29 09:33:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/83/839883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/839883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/839883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/839883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/839883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/839883/1"}}},"commit":{"parents":[{"commit":"a54725cff755192b7500af4469e88aa5e607e8bd","subject":"Bump stackhpc.drac role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a54725cff755192b7500af4469e88aa5e607e8bd"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-17 17:29:39.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-29 09:33:10.000000000","tz":180},"subject":"Fix Bifrost inspection through DHCP-relay","message":"Fix Bifrost inspection through DHCP-relay\n\nThis is continue of the changes to support Ironic/Bifrost provision\nthrough DHCP-relay same as I9488a72db588e31289907668f1997596a8ccdec6\n\nDepends-On: I74af38dc555b7edee8331e31dfd1a2fbfe4f1151\nChange-Id: Ie1d5ecf32c637b77757e56fbe2fae1ff7c0bf000\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit d271e705b94f0f82a14dead99c6fe660ea077de8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac63fac1ba2c82aa8667caacb8b8ed18870cbb6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac63fac1ba2c82aa8667caacb8b8ed18870cbb6e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
