)]}'
{"id":"openstack%2Fkayobe~839965","triplet_id":"openstack%2Fkayobe~master~Id9418e4e88c52056412daa22462aa611bfcc59ae","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-05-27 17:47:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9418e4e88c52056412daa22462aa611bfcc59ae","subject":"ironic: Set MTU on provisioning and cleaning Neutron networks","status":"MERGED","created":"2022-04-29 16:15:51.000000000","updated":"2022-05-27 17:48:52.000000000","submitted":"2022-05-27 17:47:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839965","meta_rev_id":"9a32450c486716ee8b63528f7d897ed6d4a8c5ae","_number":839965,"virtual_id_number":839965,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-27 17:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2022-05-27 15:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2022-04-29 21:44:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-05-26 22:18:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-30 09:29:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2022-05-02 07:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2022-05-27 15:34:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2022-04-29 21:44:37.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 18:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 21:44:37.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-04-30 09:29:50.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-05-02 07:44:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-05-26 22:18:27.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2022-05-27 15:34:38.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"81db4b20cb04c75e8ddd0f02ce04d06003e43c4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-29 16:15:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea90473004c2a0a5b93deae59f28cf81bd73cd49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 18:03:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e38c0cb2e64448a2912b2c6e0c192c06 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c87b3e2e8844508abe70744010409e8 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d8b2e33efbf48ba8ca5666af0664893 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a0b90f30218485c96809ae1aacd3767 : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0eca51b77e374625968dbe348598143d : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f56ef788f4b4089b45f34f8b2668cd4 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/202b09e0803c44d7aa6f86038251447d : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ab142bb65b6451696bbebe967ffbc3d : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84f90def764d4b93963c711dc607c4f5 : SUCCESS in 13m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71cd31123a3b450bb9c002ad7e00ef5b : SUCCESS in 18m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bc3b25adcf62494ab47f574c2178e11e : SUCCESS in 1h 09m 29s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/dba5f9329e434870b3e6dc2bcf60d100 : SUCCESS in 1h 04m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d24b704a81ac4f14acd2701ef0e9a28a : SUCCESS in 1h 01m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a3275d9aa2294eff9c75872ccfa4bcfd : SUCCESS in 1h 13m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/920e715c38444b2c9feb37d7d7a1947f : SUCCESS in 22m 43s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/1961ac69e3534b88b853dfc6780a83d6 : SUCCESS in 21m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70784ec5b648413d99169459f88ebbec : SUCCESS in 15m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b1edb8671b4a40a1879fe75c69b65606 : SUCCESS in 1h 45m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4afdbd266aca4727b3480f55d1143e52 : SUCCESS in 1h 16m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3c6ac0dd6d6e4fecb0a10cf88f6e163f : SUCCESS in 20m 55s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e37b9e9f3bd5408ea08b6a0c80a0877a : SUCCESS in 20m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10b4d2522c904edaa0147510ed1b28c0 : SUCCESS in 16m 43s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6694dd8474604fb9b360d9abeb007b85 : SUCCESS in 50m 01s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d4ab345fb8e740c2b95b56a33e5a4c9c : SUCCESS in 41m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eef2f5977f54c8aa6ae01686778ddbf : SUCCESS in 31m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2962592debf3415291b5b7126041313b : SUCCESS in 17m 29s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f2fd8ccd2c6c42f8a57f2d7dba4e181f : SUCCESS in 15m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/817676456a7246aba58db5c002a28e07 : SUCCESS in 13m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1bb3c943b6bc4784ba258d4993c1a0f2 : SUCCESS in 20m 13s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3a7cf2bd34724685b3b58ffe80faba6f : SUCCESS in 17m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59b5f85ec6e84c7fb37b3d260b6ac92f : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1951d3ed4b09f4b689cb22c5c60a05aedbf984a7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-29 21:44:37.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e001925b2367e90a8bd49e5398bf899026379ad2","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-04-30 09:29:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"351d514baf33bb2af7535c48123c85c5f57016be","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-02 07:44:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec56fd2dd2c09291305be5b0a241fb4b2c24613e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-05-26 22:18:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"659a7bb2a5101ed3eb41d7637db8b68035382517","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2022-05-27 15:34:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e0761273486cf664f371ddbbc65d3f84063ab499","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2022-05-27 15:35:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"112553a8ad463e647e0811cd88386bfb4209fb57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 15:35:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0002e3e783c4e77b22346f901df06df3f40ebec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:47:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2457632e0c948ad91b468c868276c30 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c7424ba25ce453d86dc38353ccd6edf : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b1f7a60aa0f469ba317110c8faf2e8f : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b74a60de4843ae8c6d2b96affd965e : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/638e07b3f6e34a548d57e9ee5315f29f : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9fd7d4981c014a44b38394059bbd16fe : SUCCESS in 7m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2aec9d938609491eaa84f0e52f242b7f : SUCCESS in 16m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a7eb0b5ccd44e1eb2a53139d23a8bac : SUCCESS in 19m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/972dbea5f1594923bbb156a97a46aea8 : SUCCESS in 1h 06m 02s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b4b2ad9150a347b59dbb436e5499cc1f : SUCCESS in 1h 11m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3b5867e98b64978978079a42849802f : SUCCESS in 48m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/60828756f49d4424ae8ed903ef09d773 : SUCCESS in 1h 07m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3ad6809800eb43918a186ad9be7e1030 : SUCCESS in 21m 34s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/85b54c7dd3fa43c7bab2ec37370b84ff : SUCCESS in 25m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/905b0c30fcee49e4a937bc3f0c910ed9 : SUCCESS in 15m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bd3c1a5a06024ece9fa8b9d429d34f5d : SUCCESS in 2h 05m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe0fda61c2214356a0eeacf30cca5e12 : SUCCESS in 1h 18m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b3a6cfe021184da8b2d4b7eb63f291ce : SUCCESS in 20m 08s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d7e08c9242d046c4a3b38a1b7e1d3b3b : SUCCESS in 21m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6240f4a0992546869b29e3c66a61d8d3 : SUCCESS in 18m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d82a94eaa0d949abbb051d76f268d5d2 : SUCCESS in 44m 14s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fadca45302614ca1a16e22228cfa53dc : SUCCESS in 30m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f00aaff5cc2c4634b617bdbdc725ea1e : SUCCESS in 17m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1ae57b1855c64f12b75d446e44600ae7 : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85940577889e487e892295c31dc39116 : SUCCESS in 12m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8e21bb4a3a9d4ef1a9a36b71dc406e96 : SUCCESS in 19m 09s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d408e538e1654e64b8da9ee6b2dcedbe : SUCCESS in 18m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5390ad3636414dbf9752433975b5ca3c : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ad07db375e001432c9e35052c81f7a4b09d64b67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:47:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9a32450c486716ee8b63528f7d897ed6d4a8c5ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:48:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97cee404bfd24d56829a67e20b905850 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08fb85cc5d6f466c902e636699dd0aeb : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd24b6d98e1b82ab9c255e3e5ddbaed99c49067f","revisions":{"cd24b6d98e1b82ab9c255e3e5ddbaed99c49067f":{"kind":"REWORK","_number":1,"created":"2022-04-29 16:15:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/65/839965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/839965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/839965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/839965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/839965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/839965/1"}}},"commit":{"parents":[{"commit":"341bde30b9b7e8e924fb0ce7a5918f5eb2119d79","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/341bde30b9b7e8e924fb0ce7a5918f5eb2119d79"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-29 16:09:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-29 16:09:49.000000000","tz":60},"subject":"ironic: Set MTU on provisioning and cleaning Neutron networks","message":"ironic: Set MTU on provisioning and cleaning Neutron networks\n\nPreviously we were not applying an MTU defined in Kayobe networks.yml to\nthe provisioning and cleaning networks in Neutron. This could lead to\nissues when nodes communicate with the Ironic and Inspector APIs.\n\nChange-Id: Id9418e4e88c52056412daa22462aa611bfcc59ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd24b6d98e1b82ab9c255e3e5ddbaed99c49067f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd24b6d98e1b82ab9c255e3e5ddbaed99c49067f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
