)]}'
{"id":"openstack%2Fkayobe~840033","triplet_id":"openstack%2Fkayobe~master~Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b","project":"openstack/kayobe","branch":"master","topic":"cleanup-orphan-interfaces","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-30 17:25:07.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-16 11:13:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2026-04-16 11:13:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-04-16 11:13:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2023-11-15 16:01:43.000000000","reason":"Change was marked work in progress"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-30 17:25:07.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b","subject":"Cleanup orphan interfaces","status":"NEW","created":"2022-04-30 20:19:00.000000000","updated":"2026-04-30 19:05:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":68,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4be1d84ecf4b9e5b3fb05b1bad82338d02b828d2","_number":840033,"virtual_id_number":840033,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-30 19:05:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-30 17:29:52.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-30 21:44:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-03 19:48:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-01-05 14:45:40.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"CC"},{"updated":"2023-08-31 15:42:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-31 15:42:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-11-15 14:52:45.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-04-13 08:41:25.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-04-28 08:57:07.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"CC"}],"messages":[{"id":"e83d033f2b30f3a7cee8ab53a6c8f34575f43679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-30 20:19:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7d55dd99ace02bd9e3189e2429646065cdc7ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-30 21:44:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8602c03070f450aa48ab7acce022172 : RETRY_LIMIT in 1m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96143f126a4b43dcb85a346914410386 : RETRY_LIMIT in 1m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/912cbd1726554502ab6535dfe7ccb065 : RETRY_LIMIT in 1m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/533285a740b740c3a7ec356720f7f7cf : RETRY_LIMIT in 1m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/880763d7484f450bb8b1f3c9d9fbb056 : RETRY_LIMIT in 1m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee7dc9d62dc344f8b5d3cc79dc719eab : RETRY_LIMIT in 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a845ce4147746a5a3bada86912a052c : RETRY_LIMIT in 1m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77c883ca1a174c6d9e79eab4a8b4fd7e : RETRY_LIMIT in 1m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd7d430790304c5aaa3f0587c6e0be91 : RETRY_LIMIT in 1m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1e459f1ed1f84d98bf56a00be5039a4d : RETRY_LIMIT in 1m 04s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4ec670637f3e4b67921bf4ed600bafc8 : RETRY_LIMIT in 1m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8946fc512a364dcf95a336623bd56508 : RETRY_LIMIT in 1m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c63ca74ba06241fb97fda6b1d8b862f1 : RETRY_LIMIT in 1m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8e0d98243a15446398615ceaea28f76d : RETRY_LIMIT in 54s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5eed4dd4dbc34fd5ad9b32057feb3071 : RETRY_LIMIT in 1m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c53b46cd02f43248d6a42afdb52be53 : RETRY_LIMIT in 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/78ee9c824083403aa6a0dbfe126d2373 : RETRY_LIMIT in 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a43aef00c60a4879806b510f6af328bc : RETRY_LIMIT in 1m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a0e4265750f04b899664b0f43a0fce90 : RETRY_LIMIT in 1m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/7da58a8171ce4685a80a5c12f621ea2d : RETRY_LIMIT in 1m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/632a7f01bfeb49e681f032a2b8862c2e : RETRY_LIMIT in 58s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/e1ea9b32ce2242628fbf251087f86ca7 : RETRY_LIMIT in 58s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d1a31614e4c147f5ab3397a21e0e2445 : RETRY_LIMIT in 1m 04s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bc2ae600c4a45d2b2cc676eb2887f6f : RETRY_LIMIT in 1m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9a5bebe350024dbf972c5faa011e8395 : RETRY_LIMIT in 1m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a4df5e19e42e48aa9f613c5d3b7dfac6 : RETRY_LIMIT in 1m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b04c819351fb449a87b67cdc68ef2ab4 : RETRY_LIMIT in 1m 09s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bc7e3d0e677e4290ab29bc10517a5f1e : RETRY_LIMIT in 1m 02s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/677f2d7a864d4ce2bdb47d5591a651da : RETRY_LIMIT in 1m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5aa3627a78043398bbba15a3ed68611 : RETRY_LIMIT in 1m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b5b4835379994f08a3384a0148696875995dc377","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-01 17:05:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a79f2e6b35f347cad67b3d5777a3a9ee4c8cb76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 18:58:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e88a6ceecbe34998bc3ea5e60050d658 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a1fb3385544e5894ac58ce140bea41 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cf607d2635546feb28b83aaa6a3e4a5 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b32f1c8e66c462591f13b52c943e3b0 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bc46650a602458ba86b47460126b47f : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f876a350fccd4e5e80cc757b74052173 : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c1b1e10cc2094fc28bc4d82e30235ac5 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ab3be2351344d099ccb02a3d1a9fc02 : SUCCESS in 13m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30de25c1e6f6487cb4b56d8b5540ed4c : SUCCESS in 15m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5e929dbc581744258729b8562da3ab85 : SUCCESS in 53m 28s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8b1fa7e663424acbadfc389430b94417 : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64d80f6ff0b1483291b996c49134b1f2 : SUCCESS in 41m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fb589bffb09641d186d883b342f72bcc : SUCCESS in 1h 00m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fde7778d8bc34b269f046edae9e3caf3 : FAILURE in 11m 41s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/ceb002a72d1c4311be9ef5631df0b1f4 : FAILURE in 10m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2470870f5c4846bebb241c4dfc48a802 : SUCCESS in 15m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e64a5e84fa594d16878281aa27fed4b7 : SUCCESS in 1h 51m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c39e4d41d9584f77ab5d5d19492819ff : SUCCESS in 1h 21m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d10011f7a6f84930879f160d78e3dbaf : SUCCESS in 20m 42s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f58167a216fe4c8fb90b2c5cd623815c : SUCCESS in 19m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0800a9246e524b60bb35495372c6428c : SUCCESS in 14m 45s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/73afa0796ce4450481e2e45dcc518f96 : SUCCESS in 1h 03m 20s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/28a8e80ef3bf41798c43f498ab0c6e27 : SUCCESS in 42m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d894d2e559e743dc8e1d5da6c4d0119d : SUCCESS in 26m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4b6e2cf8bd884509a50456848e2ade6b : SUCCESS in 16m 43s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fd85db2bd1d14b2d89aa5fe2108bf958 : SUCCESS in 15m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83ba8abb7d254cd2a5ddebcb22ab10ff : SUCCESS in 11m 23s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bfcdb5b338464ba58a5a1eba42e6cff0 : SUCCESS in 14m 03s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fec790605de44aa29354c6a2294520d8 : SUCCESS in 15m 55s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc44308c42454917a1d1363cc96f27d2 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2aba0a747008ce2144661c644ad8767516af2340","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-01 19:04:34.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98777359804ec3a2135f296e61cca6917165f7fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-02 00:53:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"000c36559b7181ea17b989b9094fec583afc675c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 03:07:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c720c9bd6f7438f904bfa5781b0f8dd : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c41654498f114576bc591fb22dcfca4c : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/311286388f094dbb8ee1fdd2f36e00d6 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6c9c6b21c454af8b10787750aff6ee5 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/205397a6a69a459eb86d204b10010dd0 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6820620fa8f24661a85e691f9eefee08 : SUCCESS in 8m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0ff02b50ade34ce5b6d4be7582da81bf : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff677ae2765446bb83ff26a6e085b3f0 : SUCCESS in 15m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63c8bf64241d4c34bd2a74b1903f7ab9 : SUCCESS in 20m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cf1fb98771b64b4392c2c38e4d784305 : SUCCESS in 1h 12m 59s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/c7c619466b7d4d7eb9f48bfaf57cd097 : SUCCESS in 1h 09m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88b7377769e5415889f44bacc229fded : SUCCESS in 51m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/41013011fae54bec89bd627e3c8a02e0 : SUCCESS in 1h 14m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e000d06ff7924efaa2ed77ac1dff9315 : SUCCESS in 24m 16s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4b1eabd1c87b4c669b00021067495c3c : SUCCESS in 21m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71f59964679c42e39b8f3b564122a324 : FAILURE in 11m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/32196a11de8641f182d81b969170b5d5 : SUCCESS in 2h 05m 05s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/790fc2c7b4fd4ca9b2ffd099d2edee56 : SUCCESS in 1h 24m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/df252ca631ba4081a3236212b845cb60 : SUCCESS in 18m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/deeba480df8f4952bdf50312ffe3c079 : SUCCESS in 20m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d89bf1c53af4da4a8d5f7542eeb5517 : SUCCESS in 16m 27s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/c6738c6e98df453a8aa736afecee7322 : SUCCESS in 44m 35s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6b78df262f484d6d989913c8000dfe7e : SUCCESS in 41m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84d0b3fe0ade48a3895fe0227079e512 : SUCCESS in 27m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/26a1101c2a484b1d80c1c3762452346e : SUCCESS in 14m 55s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e7d83635e09a4a4ba9f6af24238efe10 : SUCCESS in 15m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e39247e18fb4be4ab95a70aeeab8678 : SUCCESS in 11m 39s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/17cfb96fa85048d195a96c2ff7e6d1cb : SUCCESS in 19m 28s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bcb186bfe55d4b388b36d66a83ac4f8a : SUCCESS in 17m 29s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4e305bd02584966ba04ef76b809291d : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ff9e45acbbde23a14c9793afd7d6d286ee6c2874","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-02 08:09:47.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbf2507828f2b806793f353ace684540a4b591e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-02 08:19:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"384b5340eed58924209b0e9e34aa2ddb04c46cf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 09:47:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bea52f9a45c84adbbb10dcfe02dcc253 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d7551c36f14984a15ba778fb6bc9e7 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/136798490b0d41e48940babac4a7db0e : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dc38e3ec059401bb2cdf9643ebfa742 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bfecbf9a50c4ea6a56db69462d5ce66 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30cd2516801849d8bd7ddd4aa71c8c56 : SUCCESS in 8m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f985e18484244ea1b73e8dbc7bef330b : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e361d42c178e4990b82042a264597bb3 : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/241c6168c1894001b7ba27c24bccda91 : SUCCESS in 15m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4b2abca2f8cc4594acee655378d14575 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/69ae438c7f3043dea4a136ffa817c98f : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d855e30295d44e081744f075c38c255 : SUCCESS in 51m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4be20e1155a449cab0fa1dee74c46a2c : SUCCESS in 50m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9b88ebaee1a84a97bc5e40a0c1a59361 : SUCCESS in 23m 30s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/11de7b3f7a1742fc8509eb63b3f5130d : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e60985d78cf4a17ab66b1319b5c459d : SUCCESS in 16m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2f180760688c4f148484dae922c1c9a0 : SUCCESS in 1h 22m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0134590f9f29445d8b44637568e226f3 : SUCCESS in 57m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2fb94440cfba48a4b9e59ed446d1a8b8 : SUCCESS in 16m 15s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/85831b85cc1440039ad540b2db8cbf86 : SUCCESS in 19m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa81bb5ce0a84be3b837cb315a52f09a : SUCCESS in 17m 17s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/ee38c25eede94b5c91ad249704947533 : SUCCESS in 39m 05s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c92603c6888143aba35f3b79c8d51c05 : SUCCESS in 38m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3e02e41b4dc4c62a0a8606eb0d93efd : SUCCESS in 30m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/266831f2d39f42cdbf3b6bec4e9ecda4 : SUCCESS in 12m 18s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0136d1d118264074a6d3c5e233dfbb40 : SUCCESS in 14m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c15670762f2c4c32bc5f01e00d4e2000 : SUCCESS in 14m 28s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/18c89213952b4820a8a7e586250b72df : SUCCESS in 14m 02s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fde130ed33d745b9804d088758a2192b : SUCCESS in 17m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/472a4d90935a411e9a7176c5c199c94e : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5af0fdb51f3ade1e60177e5bfa0d092e296774aa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-02 20:22:27.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eec682a3546bd6ae005cc48bdadccba47ad87083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-03 11:20:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"64f4f2c2c7cd762bcc0cf3356167f22fb7e2b76a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 13:52:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f4a75dd6ae0442f952188df29402005 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44744e48be9c4e06a9e3910d10062a5d : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a47f2c7d8d684f1aab80ca64ddc6a08e : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e92683229af94f10b2bd8b20219295a3 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60d02111bdc144fa9f660937d1e47107 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b6bda8c8ce487d8fad797998a08d93 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cda44da3aedb4d1dbd8197a6c52d455a : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f9e926e3b854d728faa77d18c6ddd47 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/82041e1b269443da82f78f300163cdeb : SUCCESS in 13m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e6a85984baea4f1ea427b326c77f9088 : SUCCESS in 19m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/64d2934d1e1841bbac766e7c3fa87b81 : FAILURE in 22m 08s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/231506b29a714467ae9f213a79746678 : FAILURE in 24m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eff2fe726b345adbc9aad1ed775c5fa : FAILURE in 1h 22m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8c3233e5766249d1af71d411d6f75041 : FAILURE in 1h 30m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/446692ee46f2493f9f41da2df937c69b : SUCCESS in 23m 10s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b27c22b239894eea86247b5819ff8883 : SUCCESS in 21m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0c1abfd40e4465c88534de4a57dea9e : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f0350c069c6d4a6689b5f2eb3790a2c5 : FAILURE in 2h 27m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/292402ef72d642409d63403dd95722bb : FAILURE in 2h 10m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f5eb76ebedf1427ca84dd99c25e05eb5 : SUCCESS in 1h 22m 32s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8452d2b10c2b46d59584e653b794630d : SUCCESS in 18m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4b73e2bb07a41019c6cf16f2b86814b : SUCCESS in 15m 24s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/ac92c57704774ccd8d3fb44f62c93e74 : SUCCESS in 45m 19s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b32d45748cf24fabb1ede73bd35f48d5 : TIMED_OUT in 1h 33m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdd3937c0dea486392234cbf24563747 : SUCCESS in 31m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dbe3e91ec0cc475ba59ab2a8cab1e0aa : SUCCESS in 14m 49s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2608e15d8c7f4576ae10fbc9d1e35795 : SUCCESS in 14m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da54a8bcc4324bf8bada25b98d135664 : SUCCESS in 11m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a8a4d733150d48a7827e8e6f165dfd33 : SUCCESS in 22m 16s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f3b4f337c486412b84e73046c1743585 : SUCCESS in 17m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ea26869b17c4ac9ad175ad7b17681dc : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":4},{"id":"a6daa825f0113866a9b5556012e69e852faeeddc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-03 14:16:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0cffce22f3890389cd4fe68f102fbf6c52e69710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-03 14:30:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6d251242a730ba13c3597660414958bb7bdb42a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-03 14:46:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4c0ed12f84f458cae134c07787b8eb64c8d56b2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-03 15:20:48.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"cbfb8819e92243512aa587bb021ac1cc72a8cb23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c4177f139cc47c3a4df677ebacf5938 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe63c0779c7d413884d6074e2fa87fcd : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce79d392b75141048a6297029c282a74 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/585035ab472c4863a31eaf5355cfa567 : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45564ad9f29648cf94782cdf00e7b733 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59710643c3f04d369a137ae4e5dbf412 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7c2628d221140be97e6f886b54f49f0 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f55b96f26434974be7f33d7203d2f2f : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa515bd284594065952f43578c7c5b68 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fdb72f2b405241318e42d9db2393c99a : SUCCESS in 17m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1c1050d6dfac43f9b34bf7cbab8e577f : SUCCESS in 1h 01m 12s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e335161cc5c44e7c9a1ee5236a8fc62d : SUCCESS in 59m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59df24a7e34f4edaa47f490b79bcb664 : SUCCESS in 54m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/19212c5c04f3455f9b7ba1f2b6466638 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fe94cf5be9d34340936b4c00a22532c3 : SUCCESS in 22m 43s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/25bf342e3589405097396f2624e15924 : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/548fb38543f34b609ceb288dfb84a0fd : SUCCESS in 15m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bb78ce0282274927871894b98729f461 : SUCCESS in 1h 40m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ac28be4bd724eb9bc81f5c346a6d9ec : SUCCESS in 1h 20m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d0b9b9572a7a4ce58d8e4076e1ad6f30 : SUCCESS in 18m 30s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/806f9732d9ab49ff843d14ef6f9ca82a : SUCCESS in 21m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/156656964d3d484f96e251e92b9526c7 : SUCCESS in 16m 32s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/2fc17b50e8114e4ba3c80edfcb38c51f : SUCCESS in 45m 46s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/50fe40efabcc451fb2dc3574130bb691 : SUCCESS in 41m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbf5f8e5569242418b77edc51d0c0a43 : SUCCESS in 29m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b559bbbd413c45cea5bd78164e75681f : SUCCESS in 18m 28s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8ef0643791024e0d8e601bf7e21f24b9 : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3ba12d27f704813bc7ebd1c87e2c8ed : SUCCESS in 13m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/77c13c9ad6064ce3b597485a3476af41 : SUCCESS in 20m 12s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0d3e96d226624cb6815da4a6382c5d91 : SUCCESS in 18m 30s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3401ebae58224dcdb8ef051112e123bc : SUCCESS in 16m 27s","accounts_in_message":[],"_revision_number":7},{"id":"9a77f475bfff74e84fd4d108e9eb3064da7211f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-03 17:57:01.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"fa5248a0a7e96bb1444210055ef69fc170a056d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 19:47:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad6f3344671b47c586f3dea65a053316 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3022d82c6576486e8f8648b3fa3b04eb : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43e3c6f4028c4c8fb12c2f5acbe50d80 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9be328748fb54330bfc6fd141fce06c5 : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faf554c0b184472eae0bc8b368154bc3 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/626a14bb5e6b45b6be7058affe23099f : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8235704338744bb88a075a60019e1c8 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29cb8e5be0d14937a769ca8391861fff : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/734bff27eebd4523abe791bd56ba897e : SUCCESS in 15m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6b3a46a0ae441719e3679a337c4bded : SUCCESS in 19m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a1ea009cb62e41faab375a5ba6f66d89 : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/527c5d736aac4d969302bc648011a40a : SUCCESS in 1h 00m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db49f56fb35f4729a3bd68a75bfcf8a8 : SUCCESS in 52m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/03d0e2699eb74b7e993662e1e84fafa2 : SUCCESS in 1h 13m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/60a809293b794613b1c98b7ed19e6c97 : SUCCESS in 20m 52s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e3551085e5d3434494f49ffcb47fd779 : SUCCESS in 20m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e05d82a6be5c4271baf794f6de2274ae : SUCCESS in 13m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fa1e02eb47ba4ebf97cb333115f867d6 : SUCCESS in 1h 45m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/804b3b574abd4c88b958de43958f3ed9 : SUCCESS in 1h 11m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5c539ac3f64a4a06bd76116df1051f19 : SUCCESS in 18m 25s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/44eea27f71214b4eb524c32d30fff49a : SUCCESS in 18m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f955106981284472a7e5408789d3a183 : SUCCESS in 15m 05s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/07f305d622cd47dcb311ac7a1f754205 : SUCCESS in 45m 25s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/10492170d8484342806c9fc1c0e1dd30 : SUCCESS in 36m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/371ff110d10b458c88559a3e96bb08f5 : SUCCESS in 27m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/75a8abfc421043158ad87a6fbc413078 : SUCCESS in 16m 02s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/33308e5958d042618e4f56037adbc33c : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c2e1a6229e54d52b7b0f161262af328 : SUCCESS in 11m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c14ae76f385848c6abb5444f01a17b1a : SUCCESS in 17m 26s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3ce6996a9ee74b9c953140b7ef148700 : SUCCESS in 15m 05s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bf684da0d6748b8967f974785f99f7c : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":8},{"id":"bf4e2bc83e9a03b252929aa72b337434a61af34c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-09-29 16:22:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5952ffcebe37f9de8615424530aee6ab8ec9060a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 17:40:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5557f867a084d23b4048050fd0eb7c5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdcf7386ab5647408865b73c7f3b2ebd : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d19fc6d1da400fa283e401233d23a3 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b1fa3c79cd54240b4949159adf67305 : SUCCESS in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d8c21874305437691d864c4322fa9f4 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1938e3967e974cc4abd4c89ccffa6011 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d66ac857ac48e58f7e2cb40fb831ee : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ed15694584940f399185f6d9b6f0f01 : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b067ab5a30e64d119d8f3df35ad9e2c6 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6933c639a3764743990c210e4eed07c5 : SUCCESS in 16m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9912724629e44067988f5875c7cc7159 : SUCCESS in 16m 54s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c7395a53cac461cbe83d1874e14b67a : SUCCESS in 58m 04s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9a5f51bd6f94308b4d52d9fe5647259 : SUCCESS in 18m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68bd4224d1514556ba5c1d5b6b4ed38d : SUCCESS in 1h 13m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61416781b2ff4ab5935fadf9cd6b5707 : SUCCESS in 20m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8507ffc64d2488cb0be7fc4070c75c1 : SUCCESS in 28m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f4a027ccc2c4e40b7b9df25a4b61165 : SUCCESS in 14m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a320d2d0308042e2b1d7d4fa43041557 : SUCCESS in 17m 55s","accounts_in_message":[],"_revision_number":9},{"id":"7f6e5a309314d64f51a3bc23095968bb7820b9a3","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-05 14:45:40.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"06af4f45590b76a64c88797ec46efcf612699c72","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-01-27 22:41:27.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"11f780df6651e7872693e32318cf8a3af486e96d","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-30 13:48:49.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1fd4baab9835d7ecfd8926053fa41475b8de327d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-15 21:20:58.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"af2dc22c605380c0c108f6f76207a8fb6792d09c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-15 14:52:45.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e0ad7043dc33f13fb8caf3ed396d7df8934f4e8d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-15 15:58:57.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"67b0114b73d7a7753c6b485fd7abd3361c7e0496","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-15 16:01:36.000000000","message":"Patch Set 9: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"25da0af3a83b9c855c444009860062e33a2a864d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-15 16:01:43.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":9},{"id":"a44250eb3ce025411799333d628ae26e9be1348d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-30 23:01:32.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":10},{"id":"1eb1f29cdccae662b5fa12b33724f213e89b94c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 01:34:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fdb1981ead740c78d59d09c8f2a9c7c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f6371827b784be085d62d1e4570bbee : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a8f6c864e6499c8d4d6d5e5db571ed : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7257a34e942e40069e8eb0067f799bdf : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f495f1ba8f449a3835e739e750de2f5 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68f2fac17388466183b089d21a9d1669 : SUCCESS in 14m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cc822b4cfb7c4a58a0a4a5280588c8a4 : SUCCESS in 11m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e89dd6a77b7f44f6b4a81d93c549fb3c : SUCCESS in 47m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/930c04484f56494291808d07780f1cd6 : SUCCESS in 1h 01m 59s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74010d8e7b154e85b485ee1add9b01f0 : SUCCESS in 59m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/91487c21d109455680991cc43ea9312b : SUCCESS in 57m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/94b25e15ee3a4e589a5ef664cbe96ad2 : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cbc9526abf9b41b2b85671831c50ad5f : SUCCESS in 18m 40s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a28d144fae94c67beaad2f3c68c5477 : SUCCESS in 11m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ca2dada7924c4ab488e8ec9e924308cd : SUCCESS in 2h 30m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d3f4c444b3e4b6d98ed20c39855eb59 : SUCCESS in 1h 59m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/04c4ff45786b41f4be7413ac6b236e50 : SUCCESS in 18m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51a8da876a044211adbb0770131c2bb1 : SUCCESS in 16m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3b7cee8ce5954c0681c56014cf5d617b : FAILURE in 27m 59s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9eac02fd415049098281cf6e3e70fc7b : SUCCESS in 31m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5382ab5542c04058a901f0208a069137 : SUCCESS in 28m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d8a83682293d46bd894a30b206c0617a : SUCCESS in 15m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1483b282f60c44669932adba5948b565 : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1849673aec4b4a6da1329a58025279f9 : SUCCESS in 21m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebacc6e1e30145e0a9bccf41dac5620b : SUCCESS in 13m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bc1fc5bf8fe43ec878e8369f937ac0b : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7145d39a8d5940928882eed4a3a9e68c : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2d30d0739ed402e8ee7f51a4fa50371 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":10},{"id":"6e4be5c4ad7155712597b898b219df6c99f7485e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-18 10:28:52.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"d56ebf2788c77bf8ac3cf41489957d3e8d0ea5a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 13:01:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13f5901ba2ec4b7f8c791cc99b27bd43\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f26d2caee331493a9841100e5196b5bb : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40cb59fb47924bd180ec4ea1d98b4388 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb99860ccb3c412b8f02eea862d2cf17 : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01d980b09e3b4ca6ac8392c8242081f4 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1dcda9c618e34ebd8ad693326dcb0254 : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dcc3dbd25fd64e59ada4b201603fdc29 : SUCCESS in 8m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/28179a0ef76d49a38ec110bbbfa209a6 : SUCCESS in 56m 56s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/67fb2ab8c23749adb655ba9dd7c457a5 : SUCCESS in 1h 14m 34s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77ce49f7595d4a479bd93fbe1fcd24d8 : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bc2b57a626684fda922d4d028566d9ec : SUCCESS in 1h 03m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ce77d37c50d0496e89d21f96e2ae313c : SUCCESS in 18m 02s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/af7b722d4c954fdb8784949eea8c4e9b : SUCCESS in 19m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7efc982e7f1b4a29b68a3f4d7dd7c607 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8f52e053e1b942e59fb58120a8409951 : SUCCESS in 2h 28m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97004d5b896c41429aed875b248ac6d0 : SUCCESS in 2h 24m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2b2b6ce2f86549e09231ebe9663dd982 : SUCCESS in 18m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e49bee1d4c3a4670a98dd671a9487b74 : SUCCESS in 16m 53s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7e2aacfc9c3f41fd818590c661bbe6c9 : SUCCESS in 36m 53s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/20ccc4e2e0274b4990f86c72d5c200a8 : SUCCESS in 31m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bded7e5d21714a9a9dcbe45ea6781fa8 : SUCCESS in 26m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/94614c9e0e4049a6b891c99fb70f1ffc : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f58ff36f05fe40baa0b54561efe306ff : SUCCESS in 11m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f82a89e220c64b79bb08f32791af4cdc : SUCCESS in 18m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48f0f03a28d549189bcedd6810916d7d : SUCCESS in 14m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a488f7f0ed3d45f98e284ea6bd7bb6a1 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31cbb200d6094f5b9b4d0fc0173520e2 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b46a2930dac0427bb6f535fcbc20a068 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":11},{"id":"14898b65986f72d22f3ee95228902dee8ce6a82b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-10 12:14:41.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"1ada932b6864be96e6689f09c5d5058e1b996a09","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-10 13:34:46.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"41634ecda90b02df863b4411238e8f8884777e2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:15:25.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a23eb0305c6547608d6e59a96136853a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a9b6318f79461ba219fefcf765c218 : SUCCESS in 8m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3bd43a8ba80e4318a4c01f7549dd06a1 : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b4eebb31f494dbca33c152e994cc24c : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ae3d23dfc9a42ea9e56acfc4c9a56ee : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/047e6a1db5bc490c9f88c0b1f61eeb60 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1d3015a78e8348259549ac26eefd6f6e : SUCCESS in 6m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acee1265c5ee4a2d950bd6a7255dd4b1 : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f3bbbc00a42841b4b7fc5f229163b795 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7413bdea1bab4ad48e30a96bafdecec0 : SUCCESS in 13m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aeb361314e184426b46ded6db4ff68be : SUCCESS in 9m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d331ec34c8ee4d3d87e417d41bccd20a : SUCCESS in 20m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f193b58fc6c4e1885545cbf8f546c8b : SUCCESS in 14m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/062e5167738a41b7ab7c1f35224a0ba0 : SUCCESS in 11m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e6afd9be74b4d2b9253a282d70535e8 : SUCCESS in 11m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/aa835658202a40a99e9ad4092564a730 : FAILURE in 46m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e67bb8f9f46042f29b6cfc6c8d056e0a : FAILURE in 48m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fb3f3c62ac174f809b38fb50a354763b : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/956d6c3dfe8c4711a683b34a68d3a38f : FAILURE in 47m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fbb1f3173a8c46908f1aee484ed0b664 : FAILURE in 46m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d0495a8fcd2d41cc9d22b01ef304737d : FAILURE in 3m 58s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4f85a1e8394c45a6ad07391ce62b7922 : FAILURE in 4m 11s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74513928eafc48ff90668f44bdc53439 : SUCCESS in 1h 27m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b21c90caa0a4a8a90e5bc41a0434369 : SUCCESS in 1h 39m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2609d44926a44e93931530c743e4b750 : SUCCESS in 10m 23s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8fc9e65890274e97adece0d01c93aa84 : SUCCESS in 20m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50c31767a5df4763a18e1b6503e9a8ab : SUCCESS in 14m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/64a0ed92ec99466ca461817881d4773c : SUCCESS in 16m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bf577a8c966041b0bd8a04c06f589f4b : SUCCESS in 16m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/52f894e12bc6477d91620a24bdc147fa : SUCCESS in 28m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04b5fbd781f5434cbc122467f85aa741 : FAILURE in 11m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d30af0f93811444cad63334d8edf8ae5 : SUCCESS in 21m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2eca047f67a04b918432ab3c8ff359dd : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/519a1a317fe3443fad54998ef923fa27 : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":13},{"id":"a7d9b4d346859ae1776994b5cc5ab2585ab77d35","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-10 16:40:02.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"7b95f3599c50af30ca06e765723224fd2143babe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 18:21:46.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ab383ebd6ad4001a42a647487c1f38e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619a352b30cf4c8f8697c8155f52dd64 : SUCCESS in 5m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/150a8e9400e24107ab5287baef97ee28 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f81c376ce6e44108ac52d30efb0b3b23 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1894f1863694379bb5b542228da6779 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b211e0d9fd124b7da05e98da07bf3857 : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b61b8613894b417dbca98c7c5f758622 : SUCCESS in 4m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f35b83087a84152b0c09df5fbfc2f96 : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b8ea18d46f14f03a10e3145af669973 : SUCCESS in 2m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d7c5df244a646e6b81388ed4cb39cb9 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78f7db3b13d24d3b89ccc945981dd03e : SUCCESS in 6m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/df4b7ef44bd94b3bba31471040f49bf7 : SUCCESS in 14m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2564a0fdcc7a4e9ebd8518149400bbf8 : SUCCESS in 11m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fde875e379204a889a403999e66beecc : SUCCESS in 9m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd6a7792b64d4ded845c410bdcde2f70 : SUCCESS in 9m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e4315c6155ae46abaf6fa3f1074d26e2 : FAILURE in 48m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/06d77a2a1f2e4886b0110d6aacb21b9e : FAILURE in 48m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/46bb889bc4cf46d9aaf1c323360a646b : SUCCESS in 59m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28f6fb1502cf4e9ca8f9e3d8dcf0d1a8 : FAILURE in 44m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ea02fe0098314e0fb28b4bab39712707 : FAILURE in 45m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5202e88433314a549b9aa93b2542d3d9 : SUCCESS in 1h 40m 33s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f4332d7a03b948bf8a5ba92be648eeb1 : SUCCESS in 1h 33m 04s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4e0eb520bca42d7a388d0d79a952530 : SUCCESS in 1h 29m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/028667f8ea8b43bc8209e657ebe4827c : SUCCESS in 1h 39m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5edad20a5eb34e50b5a3ec6a0cfbb483 : SUCCESS in 10m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/53264b74c25846858a8deb55b7433677 : SUCCESS in 17m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9624fbfc4f3416ea30a77bedf8c30e9 : SUCCESS in 9m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f193201d7c9482cbb62d1eec860c0f9 : SUCCESS in 8m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ad0c3729738d476c90a219c1de41d4a3 : FAILURE in 2m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/091e7064aba047568c69c64099013887 : FAILURE in 2m 24s (non-voting)\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d19b1ef4f194e7dbfd7fa4865019ef6 : SUCCESS in 32m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/626865a6a4bb4dedbb74c0a71274e187 : SUCCESS in 20m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f988b2977be841f79202d0e95e920791 : SUCCESS in 10m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83656b54fb5241de88d3f9b828098531 : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":14},{"id":"b7ea54428ecbbac7121948587b8526040515b650","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-10 18:40:14.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"193997720d6d176a19ef58082ef4d1fd5b5f4ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:35:30.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27934080b871445b99f8bc165d800eea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ecbad0eec994b78958fb196ff2d714b : SUCCESS in 9m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/01ac886af5cd4435b43f056049cde44e : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d7dbb6c9054f05a0ef8be7ff260004 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488956aefa5d471fbf53ade06f386a6f : SUCCESS in 5m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f76125a4378247f6a23e2a2c329a61fd : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9994f49c4ae4aaaa60637ea82d7c3c6 : SUCCESS in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24c2492daaf94f4391045ee8ab887371 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bdbe8033af75447784b669a25b6e6913 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b22b4f56dde1433cbd6685861f2cde89 : SUCCESS in 12m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9cc7869a86504b37ae3604902ceb8b75 : SUCCESS in 13m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b85194baea5b4d8c995506b500c48e46 : SUCCESS in 20m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c23fb8e4b71646b0a03f331eb0b5bde0 : SUCCESS in 15m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3d07b5ad89bd478992d59fe175309ef2 : SUCCESS in 14m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4209dea2948d44c3b3c3a1efaf91e6fc : SUCCESS in 16m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fbfdafaded3e4ba29c675b719da71d11 : FAILURE in 47m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/52d799f250f340c696be74ca47353634 : FAILURE in 48m 07s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/00ded44ad2984ebdabbb83d35dfc2c07 : SUCCESS in 1h 02m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e33f89a1a364d4daeffffbc87e55415 : FAILURE in 44m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/260eba6bd13f4cc2a74648484c9f4080 : FAILURE in 45m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5ce230f9329b44f0b14ff0be4e73b36b : SUCCESS in 1h 33m 54s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1f93690dff404bedb00579b6c20f614f : SUCCESS in 1h 29m 34s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27a06e8d62a24a868e2db8a456a45359 : SUCCESS in 1h 27m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/732b98a6f51a44a4a57eb63d654086d3 : SUCCESS in 1h 34m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/570587e935de420b916737e3509111c4 : SUCCESS in 23m 35s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6a6c0bc575654e02890dac0571c6fdec : SUCCESS in 18m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cfcf79b29ce4563bff9702c18a2abbc : SUCCESS in 18m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/754224db46fd41c69d52150e373136e2 : SUCCESS in 20m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c0ba70ee37f244afa44e49bcdd4070c5 : FAILURE in 16m 15s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8903095beaf349e2af06e987d09f35ec : SUCCESS in 33m 24s (non-voting)\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26438322b52644af8a663a39ca21f0d3 : SUCCESS in 32m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4f8e0546c7e456bb6311bdef9389635 : SUCCESS in 29m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c950d7059d5a43d3ae73467210329ae8 : SUCCESS in 11m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7492d9f2fbe42989e46451258719cb0 : SUCCESS in 17m 04s","accounts_in_message":[],"_revision_number":15},{"id":"43918fa86a3fe383b0edd625011e83b3f307a9a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-11 00:06:26.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"30f9252bd1818fbe2b1e711a87cf0b5a7e76d167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 01:43:48.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f7a04c387b646a4812cedcda8df8244\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae8225c0dda94887abd05c52b4d71241 : SUCCESS in 10m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/62a97eafe9ef43d7a8ef7aa0c8f8613d : SUCCESS in 6m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/254dbd6f0c9945b2bd8ee6d8fcf61348 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f1d8779234424db968107320204821 : SUCCESS in 2m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3a04010a06f44ebaef960461968a91b : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/599822480fd24f738e4c036c9578d0fd : SUCCESS in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e390c1e2a9e4a6a80516c8315065443 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81bf86b79f674dccbaabee04d2061e1a : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4c689dbc3eb49f682d06caebd7979ec : SUCCESS in 13m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1e08d078e1f4995b73f1ef4d2f1f123 : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1ad073b4a3904e3995e294bdf6ce6ebc : SUCCESS in 14m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cedb6239bee4ee49d98be290c67b08f : SUCCESS in 12m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9d30ee4cea3b4dafb47f031a58289120 : SUCCESS in 9m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16832f9202ec4a4699259979bcb940dd : SUCCESS in 8m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/410b774396dd4e7493f7aa54b7bad6d9 : SUCCESS in 56m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/04ffa5cbbbf1468ca5dc41dd1e8f9bb5 : SUCCESS in 58m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/502c56b14d334b6c8f0fbb91b40fd200 : SUCCESS in 59m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68f92b21380f4242bb9581ad354f3917 : SUCCESS in 57m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d28df27f7004145b0d341c5e7578c65 : SUCCESS in 58m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5dd51034f7e449d6bbd31f7d218aea8e : SUCCESS in 1h 36m 24s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5f4d8cbd035a4d4d8431a74dce7a3f92 : SUCCESS in 1h 31m 01s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8456ff2dfd644a885c4c4c68b2d66ec : SUCCESS in 1h 25m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ce11fb22a9449e4ae1888335961211a : SUCCESS in 1h 30m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/29344402f7054a87a76d832882b503cb : SUCCESS in 16m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a9ad805e28f44589aa0c453a0b51db55 : SUCCESS in 10m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41a9291d1578495fa98a3a18c2d9ac20 : SUCCESS in 18m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b0fa7e209cb4ccfac4db2e973ae4b3a : SUCCESS in 9m 05s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/af7f61a9fe1b454aaefa3546722b4c88 : SUCCESS in 16m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c5638bdf71974b0eb65bb627aaa28f95 : SUCCESS in 27m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/973261f580a148b1a838ceb4471a4302 : SUCCESS in 15m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a9cf7015d3f4b84906d1afc3b7fda1b : SUCCESS in 27m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f34660319ad1411e8adff171adad25bf : SUCCESS in 20m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3011f353a21842758aaf0284975d69a1 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":16},{"id":"3ec258e9ad4e5b4779e3350bdbad14dc753626d9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-11 11:16:21.000000000","message":"Patch Set 16: Backport-Candidate+1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":16},{"id":"475ba39c0583d97544165fa38a8852def4e97815","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-13 08:41:25.000000000","message":"Patch Set 16: Review-Priority+1","accounts_in_message":[],"_revision_number":16},{"id":"129e89f7833e52a7a6e0f1f3d8e1271ce0160fa4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-13 10:41:58.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"176effc716179114d9c45a3acfac8e49995a2fa2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-13 10:54:11.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"ef04d0ac2baa06dcb691b0ac80b2df5f83abb853","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-14 10:40:53.000000000","message":"Patch Set 16: -Review-Priority Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":16},{"id":"3caaf54a54b81409b6339b43943694a9cd178e6e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-14 10:51:55.000000000","message":"Patch Set 16: -Workflow Code-Review-1","accounts_in_message":[],"_revision_number":16},{"id":"4d30aaea47ff1688f18c14996c5595574945b753","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-14 19:29:54.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"cd29ec5013b550b5654c04bff82fd4aef8efe97a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-14 19:30:12.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"43f895c36acca03883dbbdffa41191b8ae765450","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-14 19:34:47.000000000","message":"Patch Set 17: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"09a732146432e9cb9aba7be18b5b7510d22bdb2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 21:06:11.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5199e2f5624b42149f618fb7c7082521\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e1e752d47045efb1ae6e88f67cb451 : SUCCESS in 4m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b54606ab69024cf4bbbfe57566ce1c9d : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5aaf67f877e2417895f38e07d4a3ccc7 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d74e03ac2e6e4a228c188f7d12430ccc : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a38b26a988c4e2f82d87fe0930c4a13 : SUCCESS in 2m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9571832438c546baa9a43d6d316718fa : SUCCESS in 3m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e40d187db703410fb3ef9c293fe6c82a : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd5c825b2abf4252bd6d98b3bff96d18 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/88fe28f63cd34dc8b10bbc634ab0709a : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/edc5cab185944d8499bd0931250bf4b5 : SUCCESS in 7m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b75987719280406badbeb7c12b35ecab : SUCCESS in 19m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a3e5d6d8a7b4980a2d03e20e8603c2a : SUCCESS in 15m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5294fb48aafe4f61a63f94a9539875ab : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ab7c86633884c91ba6e657d02478914 : SUCCESS in 14m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ad7a94531a5645aea27081ad98bca9fa : SUCCESS in 1h 03m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fc62fb0412f14c8f946e9ea49c03090c : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0ea6a9b7ff17433b967bb7be7e03a582 : SUCCESS in 1h 05m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d23476dfef1a4f39823e37b8de945142 : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/24f35b20ba6f434287565159b87e2982 : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1484d720edd246bab7eaf1008adf16c6 : SUCCESS in 1h 34m 38s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/979be3009a9641d686b67f88d12b89f5 : SUCCESS in 1h 29m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45fdcd3510a240d2adb6ca8de94fc039 : SUCCESS in 1h 32m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0be18f059d2840499088215e482f06e6 : SUCCESS in 1h 33m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8c5fa44d83b44243b44d0354546be2c4 : SUCCESS in 21m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c06d1b9af02148aba442400088a3373d : SUCCESS in 16m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/007397da7f1846b694d9a3a23795b599 : SUCCESS in 21m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d5d8378ea7f456a84ee71ddf75ef789 : SUCCESS in 21m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b4f2cf0293ba44a287ad0367adad871b : SUCCESS in 18m 40s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d3301f527283408d8470c86f66dab811 : SUCCESS in 18m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/528bb259b7ff42a68564d16052f2d97b : SUCCESS in 31m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc2a2df327714a7c96ce0f006c44d7aa : SUCCESS in 26m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3fe7456da02c4a9c9a479c2fba194a78 : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e0c70f01235410c829f0c8ecec680ef : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":17},{"id":"f0b243659cc4ab53068d6a681bbfcaf5ea0ab9ee","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 08:46:00.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"1a174a71e80596e8939531d2f36aafeddc28f049","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-15 10:34:25.000000000","message":"Patch Set 17: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"bdb27cd0883b33fa266487ce0acbbf06c0d82ecc","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-15 10:34:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":17},{"id":"f7cda912b24162094770a81c66558020240fad90","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-15 21:34:22.000000000","message":"Uploaded patch set 18.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"71fd8e4ff1214fcb1a3070bf6f1bc833b8b7c5c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 22:38:23.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb446cb4126f48d2996238c0c33cab78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14dc99de60f243609950810eba818d30 : SUCCESS in 10m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/de3a1430dbbe4b1e8b9eecabbef96f46 : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c4b5684985d4f13a1f5a7b74bf7eef5 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead14c16fc1545f389181c1e82de1cff : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64627c12c176482abdd65f73bde54752 : FAILURE in 2m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/878b4088a360406b9fd6e70d2d204d26 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7bb3c654426245cc8833fc0d17684c7a : SUCCESS in 7m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e03fedc548db4cc79620a368b3c780b9 : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2740f02ffc7d4f2c8550317c1aeb3bae : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/18eb17454b0a42c9a90e2750b79ce65d : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a75678f2080c48f88687d28e671e266c : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1744528d1aca4f2fbbdb407f728dc81d : SUCCESS in 18m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ffcb34398504aed8b98e46fca34e6a2 : SUCCESS in 20m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f946cc5c33d24f7184b5e8fb45789d58 : SUCCESS in 11m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3b87b9e370444a997ca6cc8b2339a87 : SUCCESS in 18m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ba55c3aa2b9f48689deefc835023fad1 : FAILURE in 25m 40s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2d32894d7f44d6e9a14e2a20e1d6564 : FAILURE in 24m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/42ee5c728ace4d5f9a0fd75212a17b56 : FAILURE in 29m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3b1ef04e9d0442f9956371788b9f30b : FAILURE in 23m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45988a9c79fa46bf88edc9fe3a2ea35e : FAILURE in 22m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/edfcd522032e43959613c75c75cdbac9 : FAILURE in 1h 03m 03s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e3bed015c9b84f269573478b8e565bc0 : FAILURE in 58m 02s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9117e435070343eaaf9e5c63be9bd4df : FAILURE in 58m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/845ce02225a848d5ac302a7f7316978f : FAILURE in 1h 00m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/54906f1c20844568ba882bc5a2a30b6e : SUCCESS in 21m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/49998b9a96d048e4957c5dc758f741d6 : SUCCESS in 20m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2d9a2adc3034d0999b53220dc4dd633 : SUCCESS in 23m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1013f44aff14100a62a4f69606a696d : SUCCESS in 21m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/17a428279d82476ab0d01dee1b3f1b32 : SUCCESS in 22m 58s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8ea2395ae9f947269f67a12a54439107 : SUCCESS in 18m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e0537e7743d46aea7469acf792fec05 : SUCCESS in 29m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da7b1185c5874752b35be73a23d21c59 : SUCCESS in 29m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1b725e9bada94cec99477c54c8717a31 : SUCCESS in 19m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4274dc48152d410daa0ea7cba032efbd : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":18},{"id":"1444922960dcc22922cd3e7cf36ce1c5d3dacc5c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-16 08:38:19.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"6e107ea23e66e49d6cda5071cbcccf65ed7a395f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 09:24:02.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb3da6aec44c49f8bdc4cce1760b23ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c9535c4e25e401c8b45f6c83926e1e1 : SUCCESS in 7m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3ac34b25ae984b9f8ea79fe9310ecff4 : SUCCESS in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a29a1ce83192420b87e168579d5a6c5a : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eeb80f123a34d539ae4828186a46d73 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb68479b86ec43e0b1bb54a5b365a2c1 : FAILURE in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9cd26328d744bc8a8f703f91f575f2a : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb98e93b424244eda762c595256ecc1b : SUCCESS in 7m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1b25f27d6bf4ed5bff64074df41f72d : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b8aa387e7564982854b021b88aa133a : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3a2c2f963f7e470abec12a23691e51f9 : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0aafd08168f545159ef00fbeb52c1d87 : SUCCESS in 20m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/30571148a13e4109a65f833ea3f0ec7b : SUCCESS in 14m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8451d99961be4f90a2251ba644fb91cc : SUCCESS in 16m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ee64772d741b49629475925c59aa2049 : SUCCESS in 11m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8a53ba78bdc4b099fd1a9be7d3f1505 : SUCCESS in 16m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/88ef74cb064c4baabe38db45fe350f95 : FAILURE in 23m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/54246c6fa72b43deaa4ebcee342b3efa : FAILURE in 21m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9be0241cdb474169ab0b160eac2d7b27 : FAILURE in 38m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fb806f1e84c4635a53266529ad533aa : FAILURE in 21m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/614a1e2b0f5349c2bcc600ef4e0f5693 : FAILURE in 25m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/98d4c65d7dd1456aba8cfb070fb99c1b : FAILURE in 26m 20s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2da7c21a0d18409e94b4ecca7f1f6929 : FAILURE in 23m 31s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89ee41a2858a4b139067e18b1299dd4a : FAILURE in 44m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d9746a3aa9043e1b7b5a4c2e4a85924 : FAILURE in 21m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c0cea367c95d47b686dd9e4aaef60b29 : SUCCESS in 14m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d6637a25f4ef4bc88a748f97174a5b29 : SUCCESS in 18m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/caca6edc07564abab7a6be973c753b70 : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/20950c57898b4a0c8a4da5bcaabd66c5 : SUCCESS in 15m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b7496c006b7427d8c1b2d9e35ca0145 : SUCCESS in 42m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/eb19663b241f4717a1d3dddb4388eddd : SUCCESS in 19m 00s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61221c987d404413a4d99292fbb2bf39 : SUCCESS in 37m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10d5bccd05ce485dbf51da42c57e065e : SUCCESS in 19m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b8ef5b98ce744bd0ae44b4d7e6099705 : SUCCESS in 13m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2970642616a142f098bf64c13778b561 : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":19},{"id":"e4f898cbfb2518fd9d494d5d7ef6fde312e6e591","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-16 09:27:56.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"43710800bc42ae7f3471b0a69cf969309acd527f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 11:12:03.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca7f98463c4b4cc39103c8ba332e7d8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b32faec39d4060a446d5f9856c25e5 : SUCCESS in 9m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/921927f989e14820bc0117933367aea5 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d873bf376497483ead9652d42ed2bc3c : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b8f86a201d84c47a2b5b4e1bd32d95e : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a060e2b796d478a9bf3263b24c9e343 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dffca2dcc0b64845b3df8650517f3f74 : SUCCESS in 7m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32dd574a445042aaab411a04bb26255b : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cf103413262e41d4b29749a1ff4161aa : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/87200f6dfaff4ac8a6cf9a79e36af3e4 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0a0424cf97f646a593d9affe8fef711f : SUCCESS in 16m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d0c8864bf7a448adbf12884cb1ff5572 : SUCCESS in 16m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83d6fe65083d4fc78385cd28f002a3e7 : SUCCESS in 18m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/72803d4fc60d4b1b86084a77ea169e92 : SUCCESS in 10m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1256ca6274346f9a47483d1a53faedd : SUCCESS in 20m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/40b2664476314428aa0165c6e07adc94 : SUCCESS in 1h 13m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/418659be5568414482f57ff7f3720372 : SUCCESS in 1h 11m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4ec51eacb8ad4b699a84bef681231b9e : SUCCESS in 1h 09m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21d3941598c34c139c7ae9f30003bf4a : SUCCESS in 1h 04m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/25dcf83624ae4c139a4a33a5f514685a : SUCCESS in 1h 11m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b0beec19c43d436480e2b4c93ff6fbd0 : SUCCESS in 1h 35m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/96efe63289f94e9786892828daa481a8 : SUCCESS in 1h 32m 00s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2f714b2d13242ffbb110a86f88e1a85 : SUCCESS in 1h 31m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dda5bc3024a4eb7afc963fd3db65d82 : SUCCESS in 1h 42m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/59438c1fe86643078db17230b4c23013 : SUCCESS in 20m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/94f2cfbd40e8443b875db509ca8d0a8b : SUCCESS in 14m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce21889b892243e9ada14f975cea8fe6 : SUCCESS in 22m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/be457f41a4e24753beccd85c7f692404 : SUCCESS in 18m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/101c4c25235047da9c1aff97c582c59c : SUCCESS in 34m 28s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0b29cf61b2ce4287ac32d6854531472c : SUCCESS in 18m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f3fb67f65854ed6837471d24d4f34d9 : SUCCESS in 30m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00ba1c212c2c465e91f29ee5736a1b6f : SUCCESS in 27m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b2a20c6e4a9f4cfea98c6c5ceb863253 : SUCCESS in 13m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4604508902f47d2a3bea2239e10cc82 : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":20},{"id":"2cb1f1c8b0e935ca0348ecc62873cb9d851cc4b4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-16 11:13:31.000000000","message":"Patch Set 20: Backport-Candidate+1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":20},{"id":"8c87ab1468b9abe755c2c73001336848d6545887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-16 11:32:09.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"879a1e39d55e572c4c21b9d868ecdae55a974605","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 11:50:56.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"7a2b852ed2e04e5134da4a58c4c9480b6b630ebb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 11:53:30.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d450103f7624afb55e8822d4edda584c89fca493","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-16 12:32:10.000000000","message":"Patch Set 21: Backport-Candidate+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"7876957b95d646be78d1119b344f2ad69f54910a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 13:13:28.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a814cf09ad6a4dc7adc64f31979112f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5981ea01d322422095c811f47ba11f80 : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce123e99258540cc900b5082b90361c6 : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14b268aaceb44e30989ae92abe0ba625 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aff449259aa4e8b953ed6248c08bacd : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1cf10efb0534787b5ac3aec1b7f5dd9 : SUCCESS in 6m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8766bacd9db64b8d8cd6b71f004fde5b : SUCCESS in 4m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5463e395da649a98e935fc31f885b25 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab316b6bc7854aa886948dccbb528761 : SUCCESS in 3m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4eb6f02930c4a3c83737704024f7d41 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0943197a3e8c4c928aab094b3b4ebf2d : SUCCESS in 13m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9209bb5917244ca78add2576ee69454c : SUCCESS in 21m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76619a47608d4734bc3ca75323686d17 : SUCCESS in 16m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/21da5f609910446789ed57fa86404cae : SUCCESS in 15m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd759002e36f484bba10f0aef2160753 : SUCCESS in 18m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e206e43060004185952dd4d1bb47c11f : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ce2ecf97d9bc4bf1a8f705e1172f5ba2 : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/155c2635579c497cbd48c5dd222a93f4 : SUCCESS in 1h 04m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/857f4aa6fd6c42de89ef02f67dd4738c : SUCCESS in 59m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f041d369edf540899077964ab043d7f6 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7e116a4c031a4644b09ab47a2f40ae3f : SUCCESS in 1h 39m 18s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7eec245128a2457ea78d3ff0ea071f75 : SUCCESS in 1h 31m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aa3d850ef334e07a83b8af401ecb067 : SUCCESS in 1h 27m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8e4bb59424b42018bcfefb1ea9d6805 : SUCCESS in 1h 36m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fb2770067d164695baa700f4f6bf1844 : SUCCESS in 15m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6a6082b4a1c2473e976ab867149b683a : SUCCESS in 14m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff38d9663c8e4c53b3b61c788ac185be : SUCCESS in 18m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ed10efc94a9429bb0c413973fa5a6f4 : SUCCESS in 19m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c1032a469dfa4f7c85756b1763ba82fa : SUCCESS in 18m 22s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/189654d2c0894f45a9124d0580174be1 : SUCCESS in 25m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06eab2e129cc4bc38d4e35b60ac2e22c : SUCCESS in 18m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1955ccbc29445319ac9d00bb3d05b7b : SUCCESS in 18m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/466f0425ef3c46529acd1b3f2b5fe9a0 : SUCCESS in 14m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b36313a89e8f40c489d244cc139b799e : SUCCESS in 20m 06s","accounts_in_message":[],"_revision_number":21},{"id":"22d5186d076ae48ffb12d19a24138c032d0c4d97","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-17 08:25:24.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d4feb9153aca9874dbcf93fbe375633848d02c11","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-20 10:03:01.000000000","message":"Patch Set 21: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"201747da7b47ce718a6dd29039b2b0e8a8f5f8f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-30 14:49:46.000000000","message":"Uploaded patch set 22.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"905689f93c4e5db8dafc118a60ff12ca3c864999","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-30 14:51:07.000000000","message":"Patch Set 22: Backport-Candidate+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"cc8d592bff621545a406f483ffedcce743853b25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-30 14:53:32.000000000","message":"Uploaded patch set 23.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":23},{"id":"a0a4b9ea1c57eddc2eda985771f33a4f61fd4e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 15:31:56.000000000","message":"Uploaded patch set 24: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":24},{"id":"fbb8237b9386c02ca4d121c77415834ef773e671","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 15:34:31.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"f4b00a81772687c77e45dd2629ea663d66a4f8ff","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 16:03:14.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"b4c74042090b59554ebbfdd4149a46c9203cd754","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 16:17:57.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"21909ca1c0fd6dbf504b500edfcccaebd9ab3a09","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 16:41:31.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"7c360618213240edfe0b3b4450a7fa18ed6b779c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:10:07.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe818ed48e50436292f6245dee5659f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd98f43fbfd24833a714ee103e872871 : SUCCESS in 4m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5789a9707b14845ac3826ed4382a325 : SUCCESS in 3m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2130f4121131408ca96c94dd8f417872 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b672efe91ec441fa943e01e6f58ec5c8 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02b46f2abaaf4bdda8ada36f41cff64b : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8d1910c9d7f24250a59d7becd96f0538 : SUCCESS in 5m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/447969777ee941179d8acd5e1c7dc951 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c2ebbd4ddf9349c3a435d56ae35e11d3 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d385913b65f4306a9787cd9d45146a9 : SUCCESS in 14m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84659ce831014fe99a363442a5f15dbe : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/396e635d385247058cf0e7f153cd6da9 : SUCCESS in 14m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51c15dbec87a466aa20d74f3132e1f0b : SUCCESS in 21m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cb2f6424299b47ac98059209a3786071 : SUCCESS in 18m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6f5ee28d8a34211a9fdd679891206f6 : SUCCESS in 18m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/607465207cbf418485e58aea4107726a : SUCCESS in 1h 04m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0e735c0257fa4eecb7931e6e109bfcfe : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/65860f2c51524becb3f13727451f26f7 : SUCCESS in 1h 04m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3a8e665b06f4a67a2b155c03d8fef26 : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/579238dc9b534ece84bc99881a4bd38c : SUCCESS in 1h 02m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3860c5a7f5b14e9fa2e2d7aaebc4544c : SUCCESS in 1h 34m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/92c8c286c9904ca79bfe373f5a30c4e0 : SUCCESS in 1h 36m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/705a0840a6ff4cce80aa63a1884ef8a5 : SUCCESS in 1h 26m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/404504f3a7f54dafa5611214f3e63147 : SUCCESS in 1h 32m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/725f4efaab184073bfb8a468601fa22c : SUCCESS in 31m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f4f9a073cfa343c889e63ffc31e97444 : SUCCESS in 15m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4c413ad8049496dbe7e16c812dec932 : SUCCESS in 21m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3fcd150620e146719e242a05739e3e3b : SUCCESS in 23m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fed75e41403549bcbb4c6be8746d069a : SUCCESS in 29m 46s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3d5a273766a14677b163f1631d7e3470 : SUCCESS in 40m 57s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c6ff68d70b64175b4b39ba7308962f8 : SUCCESS in 28m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/423a51246d1d4ce3b35ccef7782013a3 : SUCCESS in 29m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ce8e0680d0c34e8084d5cc31ee2e8755 : SUCCESS in 21m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6af4ba866c8a421f860d2c17926c86b4 : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":24},{"id":"616f37401afb8be4d066fd02b47e4c1e7b50d42d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-30 17:25:07.000000000","message":"Patch Set 24: Backport-Candidate+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":24},{"id":"50834fd6cbe734fe0a5122c9138e2a79b3a06359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-30 17:29:52.000000000","message":"Uploaded patch set 25.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"4be1d84ecf4b9e5b3fb05b1bad82338d02b828d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 19:05:31.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/372452d95fa5472ba5ead4a4e10b8f90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d95f34939c7f42cd923788dd35bda3b7 : SUCCESS in 4m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/722fd177de844ed19910c3de23d6ef01 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52f22cdbc6e841f5931646857e85ec43 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451934bab33741d1810c9a0ba999c6d1 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0165161226f64458a9eb2ce5c5d09e78 : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec50b96917974ba6817945f464a97688 : SUCCESS in 7m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45838b8420154335b3eb33bf1cbb76a6 : SUCCESS in 4m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/48d7e166e085424ebe493371774195d1 : SUCCESS in 3m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ce52cd3a4cb4919a8b585db474cfdd5 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/03b3139c63014964ba51326804ce519c : SUCCESS in 7m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a3196db756d840a09ec71987cd06eff3 : SUCCESS in 18m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/449b52c144c748d5920338d4c872a18d : SUCCESS in 15m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9a369acd60174647b6157ea1a596777e : SUCCESS in 14m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac6fde5e2e8c49c594a542eb27651b3f : SUCCESS in 12m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c3ff66423a5141d6bf5a512eb2dd49af : SUCCESS in 57m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cc6fd6158cf3481b8f3c1016be3b8665 : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6552ae13e98f4c899fc959683b6d512a : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17dd4480e43840f8b90a3ec4d9dbbdec : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7da49924cf4746bab3647a848aed2502 : SUCCESS in 1h 01m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7587f0398e0543668a4b666840a6f086 : SUCCESS in 1h 33m 38s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/32a6fb61e23c49968e19befd46e60ba4 : SUCCESS in 1h 29m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab6477bd83ee463f9fcbaf6693ba1d07 : SUCCESS in 1h 24m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f97372de38b740cfb4d04ac7687b7ffc : SUCCESS in 1h 34m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b5e0d73f4ea44852ba6eefc13575c175 : SUCCESS in 14m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f1bde1fea79b4a0386aa629c3a967c65 : SUCCESS in 27m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9da6cec222054c2eb3d9c085ec1ccba8 : SUCCESS in 29m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/03816e7c009d4f00b2fa4787786fa0bd : SUCCESS in 12m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6e5efcb9acc44524aba7c4d35d078a13 : SUCCESS in 15m 53s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d747cac215714d83b94fd6153db9164b : SUCCESS in 24m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e186910a8ab449dac8569b7d4a3185d : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7358cec74e034749909282db96765769 : SUCCESS in 37m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/34d67286143f4adcb55f803388ab84df : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33aae66d3fd34bd1add0533f1714146b : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"fee95a4ecaab3d20b2d112ebdaf05e0353c52261","revisions":{"c24a57e5937085fe5579cadd4953af298c1eaa67":{"kind":"REWORK","_number":1,"created":"2022-04-30 20:19:00.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/1"}}},"commit":{"parents":[{"commit":"364a0d64b66cb5807dac627550568261862278aa","subject":"Merge \"kolla_passwords: add no_log for password overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/364a0d64b66cb5807dac627550568261862278aa"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-30 20:18:21.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c24a57e5937085fe5579cadd4953af298c1eaa67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c24a57e5937085fe5579cadd4953af298c1eaa67"}]},"branch":"refs/heads/master"},"5e351bd06fd138c213969a789a4446fa3a2542e6":{"kind":"REWORK","_number":2,"created":"2022-05-02 00:53:59.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/2"}}},"commit":{"parents":[{"commit":"364a0d64b66cb5807dac627550568261862278aa","subject":"Merge \"kolla_passwords: add no_log for password overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/364a0d64b66cb5807dac627550568261862278aa"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-02 00:52:55.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e351bd06fd138c213969a789a4446fa3a2542e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e351bd06fd138c213969a789a4446fa3a2542e6"}]},"branch":"refs/heads/master"},"b4ade28509a39eb46e635f01f53f0a0d5137850d":{"kind":"REWORK","_number":3,"created":"2022-05-02 08:19:26.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/3"}}},"commit":{"parents":[{"commit":"364a0d64b66cb5807dac627550568261862278aa","subject":"Merge \"kolla_passwords: add no_log for password overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/364a0d64b66cb5807dac627550568261862278aa"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-02 08:19:16.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4ade28509a39eb46e635f01f53f0a0d5137850d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4ade28509a39eb46e635f01f53f0a0d5137850d"}]},"branch":"refs/heads/master"},"97ee797d981ddb274a38f75d096c04b315b9cdaa":{"kind":"REWORK","_number":4,"created":"2022-05-03 11:20:14.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/4"}}},"commit":{"parents":[{"commit":"364a0d64b66cb5807dac627550568261862278aa","subject":"Merge \"kolla_passwords: add no_log for password overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/364a0d64b66cb5807dac627550568261862278aa"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-03 11:19:09.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97ee797d981ddb274a38f75d096c04b315b9cdaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97ee797d981ddb274a38f75d096c04b315b9cdaa"}]},"branch":"refs/heads/master"},"d8a51749b839ca61c34c9fc1401051c4baff595f":{"kind":"REWORK","_number":5,"created":"2022-05-03 14:30:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/5"}}},"commit":{"parents":[{"commit":"364a0d64b66cb5807dac627550568261862278aa","subject":"Merge \"kolla_passwords: add no_log for password overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/364a0d64b66cb5807dac627550568261862278aa"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-03 14:30:13.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8a51749b839ca61c34c9fc1401051c4baff595f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8a51749b839ca61c34c9fc1401051c4baff595f"}]},"branch":"refs/heads/master"},"617b969bc3b0bc243dc1dfbe39ea20d5cec92e6c":{"kind":"REWORK","_number":6,"created":"2022-05-03 14:46:00.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/6"}}},"commit":{"parents":[{"commit":"364a0d64b66cb5807dac627550568261862278aa","subject":"Merge \"kolla_passwords: add no_log for password overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/364a0d64b66cb5807dac627550568261862278aa"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-03 14:45:51.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/617b969bc3b0bc243dc1dfbe39ea20d5cec92e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/617b969bc3b0bc243dc1dfbe39ea20d5cec92e6c"}]},"branch":"refs/heads/master"},"1d39e732694c5cce6df23705737e28b45bb0fef3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-05-03 15:20:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/7"}}},"commit":{"parents":[{"commit":"43829dadc1f782d0abbf267925df3a4b0e4acdb3","subject":"Multiple DHCP ranges for Ironic Inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43829dadc1f782d0abbf267925df3a4b0e4acdb3"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-03 15:20:48.000000000","tz":0},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d39e732694c5cce6df23705737e28b45bb0fef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d39e732694c5cce6df23705737e28b45bb0fef3"}]},"branch":"refs/heads/master","description":"Rebase"},"cde71eb1cc0c70da086c26a5b7f86763b02b7c06":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-05-03 17:57:01.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/8"}}},"commit":{"parents":[{"commit":"339e86d43c2b2d775120f186b84ab71a5af4fda1","subject":"Merge \"Multiple DHCP ranges for Ironic Inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/339e86d43c2b2d775120f186b84ab71a5af4fda1"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-03 17:57:01.000000000","tz":0},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cde71eb1cc0c70da086c26a5b7f86763b02b7c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cde71eb1cc0c70da086c26a5b7f86763b02b7c06"}]},"branch":"refs/heads/master","description":"Rebase"},"a12d4e0725b9c1f631e71cd18a55d53776e641c0":{"kind":"REWORK","_number":9,"created":"2022-09-29 16:22:45.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/9"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-09-29 16:22:19.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a12d4e0725b9c1f631e71cd18a55d53776e641c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a12d4e0725b9c1f631e71cd18a55d53776e641c0"}]},"branch":"refs/heads/master"},"266d577cfe13847bf87fbf298f26b02332a92045":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-11-30 23:01:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/10"}}},"commit":{"parents":[{"commit":"68bb8d611244d1a85f8cc99b4b2f38989c054fbe","subject":"Merge \"Fix configuration dump with inline encrypted variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68bb8d611244d1a85f8cc99b4b2f38989c054fbe"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-30 23:00:54.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/266d577cfe13847bf87fbf298f26b02332a92045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/266d577cfe13847bf87fbf298f26b02332a92045"}]},"branch":"refs/heads/master"},"c8b9478e0c901477868db797d52b2f7bcc00d9af":{"kind":"REWORK","_number":11,"created":"2023-12-18 10:28:52.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/11"}}},"commit":{"parents":[{"commit":"bb2af07567152832700a2c2971cd72ff4e3b77d3","subject":"Merge \"Revert \"Use OpenStack 2023.2 release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb2af07567152832700a2c2971cd72ff4e3b77d3"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-12-18 10:20:55.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8b9478e0c901477868db797d52b2f7bcc00d9af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8b9478e0c901477868db797d52b2f7bcc00d9af"}]},"branch":"refs/heads/master"},"4aaa3e82c3334fe6c4c73896526f724045d57e15":{"kind":"REWORK","_number":12,"created":"2026-04-10 12:14:41.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/12"}}},"commit":{"parents":[{"commit":"323bf9c8df7c8536d36d8526b135700670056ce8","subject":"Merge \"Add custom_etc_hosts_entries variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323bf9c8df7c8536d36d8526b135700670056ce8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-10 12:13:06.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4aaa3e82c3334fe6c4c73896526f724045d57e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4aaa3e82c3334fe6c4c73896526f724045d57e15"}]},"branch":"refs/heads/master"},"f3365f3382e21f3a0dc623a7c2a0d6e8f781594f":{"kind":"REWORK","_number":13,"created":"2026-04-10 13:34:46.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/13"}}},"commit":{"parents":[{"commit":"323bf9c8df7c8536d36d8526b135700670056ce8","subject":"Merge \"Add custom_etc_hosts_entries variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323bf9c8df7c8536d36d8526b135700670056ce8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-03 22:17:57.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-10 13:34:07.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3365f3382e21f3a0dc623a7c2a0d6e8f781594f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3365f3382e21f3a0dc623a7c2a0d6e8f781594f"}]},"branch":"refs/heads/master"},"0a26aad82ff61230646d90af4232d7d72641f55d":{"kind":"REWORK","_number":14,"created":"2026-04-10 16:40:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/14"}}},"commit":{"parents":[{"commit":"323bf9c8df7c8536d36d8526b135700670056ce8","subject":"Merge \"Add custom_etc_hosts_entries variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323bf9c8df7c8536d36d8526b135700670056ce8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-10 16:39:16.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-10 16:39:16.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists after reboot because of old ifcfg-*\nfiles didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a26aad82ff61230646d90af4232d7d72641f55d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a26aad82ff61230646d90af4232d7d72641f55d"}]},"branch":"refs/heads/master"},"a1836b82343461c0aeb09e4bfcfcc613fb8d26c1":{"kind":"REWORK","_number":15,"created":"2026-04-10 18:40:14.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/15"}}},"commit":{"parents":[{"commit":"323bf9c8df7c8536d36d8526b135700670056ce8","subject":"Merge \"Add custom_etc_hosts_entries variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323bf9c8df7c8536d36d8526b135700670056ce8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-10 18:40:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-10 18:40:00.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1836b82343461c0aeb09e4bfcfcc613fb8d26c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1836b82343461c0aeb09e4bfcfcc613fb8d26c1"}]},"branch":"refs/heads/master"},"22998e5ca359c184768cf304472f53097cd94ccb":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-04-11 00:06:26.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/16"}}},"commit":{"parents":[{"commit":"6c5e1ace97620adbffa6e88776a5e41a0e4a0b83","subject":"Revert \"CI: Make all rocky upgrade jobs non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c5e1ace97620adbffa6e88776a5e41a0e4a0b83"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-10 18:40:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-11 00:06:26.000000000","tz":0},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22998e5ca359c184768cf304472f53097cd94ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22998e5ca359c184768cf304472f53097cd94ccb"}]},"branch":"refs/heads/master","description":"Rebase"},"cd560f2ea05006bdf875773811529ad15ee96e3e":{"kind":"REWORK","_number":17,"created":"2026-04-14 19:30:12.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/17"}}},"commit":{"parents":[{"commit":"6c5e1ace97620adbffa6e88776a5e41a0e4a0b83","subject":"Revert \"CI: Make all rocky upgrade jobs non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c5e1ace97620adbffa6e88776a5e41a0e4a0b83"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-10 18:40:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-14 19:26:49.000000000","tz":0},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd560f2ea05006bdf875773811529ad15ee96e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd560f2ea05006bdf875773811529ad15ee96e3e"}]},"branch":"refs/heads/master"},"fca0e6a9628982ea7f1e305ac9c3648ce40a555b":{"kind":"REWORK","_number":18,"created":"2026-04-15 21:34:22.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/18"}}},"commit":{"parents":[{"commit":"6c5e1ace97620adbffa6e88776a5e41a0e4a0b83","subject":"Revert \"CI: Make all rocky upgrade jobs non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c5e1ace97620adbffa6e88776a5e41a0e4a0b83"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-15 21:19:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-15 21:19:55.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fca0e6a9628982ea7f1e305ac9c3648ce40a555b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fca0e6a9628982ea7f1e305ac9c3648ce40a555b"}]},"branch":"refs/heads/master"},"f1c0197b22f9bb95d450e65befb3ec4d034a692b":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-04-16 08:38:19.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/19"}}},"commit":{"parents":[{"commit":"6ea0d331b2e3f59f3632635ce2de20632d6241fd","subject":"Merge \"Make Ironic Inspector config comments consistent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ea0d331b2e3f59f3632635ce2de20632d6241fd"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-15 21:19:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-16 08:38:00.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1c0197b22f9bb95d450e65befb3ec4d034a692b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1c0197b22f9bb95d450e65befb3ec4d034a692b"}]},"branch":"refs/heads/master"},"944b8f278442cfbc307dd8541c6aa906fcc8d16c":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-04-16 09:27:56.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/20"}}},"commit":{"parents":[{"commit":"9c9a8f97e09a73e722e43304ae0974052099ea38","subject":"[CI]: Fix various issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c9a8f97e09a73e722e43304ae0974052099ea38"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-15 21:19:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-16 09:27:56.000000000","tz":0},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/944b8f278442cfbc307dd8541c6aa906fcc8d16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/944b8f278442cfbc307dd8541c6aa906fcc8d16c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"f1c0197b22f9bb95d450e65befb3ec4d034a692b","theirs":"9c9a8f97e09a73e722e43304ae0974052099ea38","contains_conflicts":false}},"a197e2f5700853d71ee852f4295d6379cba95ec3":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-04-16 11:32:09.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/21"}}},"commit":{"parents":[{"commit":"ba91754fedbf9ed0b436c2d6721b4c378098648b","subject":"[CI] Fix various issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba91754fedbf9ed0b436c2d6721b4c378098648b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-15 21:19:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-16 11:32:09.000000000","tz":0},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a197e2f5700853d71ee852f4295d6379cba95ec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a197e2f5700853d71ee852f4295d6379cba95ec3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"944b8f278442cfbc307dd8541c6aa906fcc8d16c","theirs":"ba91754fedbf9ed0b436c2d6721b4c378098648b","contains_conflicts":false}},"cd89fa61f4667a771791c41298efcd0a323cb6b3":{"kind":"REWORK","_number":22,"created":"2026-04-30 14:49:46.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/22"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:48:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:48:32.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd89fa61f4667a771791c41298efcd0a323cb6b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd89fa61f4667a771791c41298efcd0a323cb6b3"}]},"branch":"refs/heads/master"},"d9aa8b252b3fd14c0fe509cf823bb3d4535d22cd":{"kind":"REWORK","_number":23,"created":"2026-04-30 14:53:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/23"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:52:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:52:55.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9aa8b252b3fd14c0fe509cf823bb3d4535d22cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9aa8b252b3fd14c0fe509cf823bb3d4535d22cd"}]},"branch":"refs/heads/master"},"52c2e0780c32b492e71bba6b377fab9db2a11887":{"kind":"NO_CODE_CHANGE","_number":24,"created":"2026-04-30 15:31:56.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/840033/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/24"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:52:55.000000000","tz":180},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 15:31:16.000000000","tz":60},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52c2e0780c32b492e71bba6b377fab9db2a11887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52c2e0780c32b492e71bba6b377fab9db2a11887"}]},"branch":"refs/heads/master"},"fee95a4ecaab3d20b2d112ebdaf05e0353c52261":{"kind":"REWORK","_number":25,"created":"2026-04-30 17:29:52.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/33/840033/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/840033/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/840033/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/840033/25"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 17:29:09.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 17:29:09.000000000","tz":180},"subject":"Cleanup orphan interfaces","message":"Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Idd9a67d3b3b403ac4d705d2b9ffd2ab87567a55b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fee95a4ecaab3d20b2d112ebdaf05e0353c52261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fee95a4ecaab3d20b2d112ebdaf05e0353c52261"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
