)]}'
{"id":"openstack%2Fkayobe~843323","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I9150e32579421e46782518948188e1363918fb39","project":"openstack/kayobe","branch":"stable/yoga","topic":"hacluster-build-issue","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-05-28 07:07:29.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-28 07:07:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9150e32579421e46782518948188e1363918fb39","subject":"Fix forgotten hacluster regexp for image build","status":"MERGED","created":"2022-05-25 18:01:19.000000000","updated":"2022-05-28 07:08:47.000000000","submitted":"2022-05-28 07:07:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843323-hacluster-build-issue","meta_rev_id":"b34e067183e439874c7e81a620231aa0c8ecb3a7","_number":843323,"virtual_id_number":843323,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-28 07:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-05-26 10:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-28 05:41:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-28 05:41:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-25 19:16:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-26 10:39:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-05-26 16:38:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-28 05:41:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"7186dd3578b827606ab6988cadfed09d42cab5aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 18:01:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b088cfa569b8bc0461cb3240d4bfac76b75af597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 19:16:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6d6c3001f6f4915a1980e801cf5dece : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16dbf62c0f364bf0bf71fc08fdcbe3ce : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a9676c8d25243909dd3d061460192fd : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59bd0eb9dc24484eb0034f3686e7231e : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d465b934cf6b46fe8b93e1db609814dd : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f791d911c0de44fca02d357fde0a0388 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a71ddab6193411ba2f99f1b309b0f40 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/04de5ff5b3aa4d65af07c762c3a22015 : SUCCESS in 13m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb27d0520c4a4f17826fbc9023a5956b : SUCCESS in 18m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/61ab06cc8c124d228126fed9c815b1e1 : FAILURE in 55m 30s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/441877677ded46d59ca360680c07d909 : FAILURE in 52m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2e58cd904d6434394a8e0653fc5cfb1 : SUCCESS in 53m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4ff113cdd7294709a447ef793015f639 : FAILURE in 56m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bbc16e0ab51741159b1d2614968cadff : SUCCESS in 22m 03s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/87220ac80e4d44c1bf61099b846957dd : SUCCESS in 22m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bf6994894e441fc9e8fc4b019d66c68 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aa53b732b35546e08da252bf931ab267 : FAILURE in 56m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4345f94f91f447ceae7839f4ff96ffe3 : FAILURE in 1h 11m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d55234582606495c95835d40c962b099 : SUCCESS in 19m 53s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b609028ab53e49ee8af9e7b721d96e03 : SUCCESS in 18m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9824016cf74942b0a8f1059c48749056 : SUCCESS in 16m 35s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/c150233510704b84adee245170286771 : SUCCESS in 45m 32s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f0d0c7167acb4eefba49f9faa79ec5ec : SUCCESS in 39m 38s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79d8fd18d53d4b079b7310ad9a5b855f : SUCCESS in 29m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1bc6dae2c3484768823535a794ec8db5 : SUCCESS in 15m 52s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bf3bc45097c54e10b29bcfab9688dc22 : SUCCESS in 17m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80452acd8dc8443c8a95733d0a0ad68e : SUCCESS in 13m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1e16137b09744d00ab5b3f90e65a9b5c : SUCCESS in 16m 50s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5a46d86cb81b4c49801a029fbab8f7d3 : SUCCESS in 16m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09daf23c9bcf4489b5b722192287d5d1 : SUCCESS in 14m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e563eca1a4160b429807568fcf813446a5634884","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 20:22:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0feeb27efe72730174c7101fd01dd752cc1438f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 21:30:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/804f6216e4624465a480f37033c90983 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c94a3c09464e79938ab6b772aa422d : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0920373d8b084101b8f1dfb7354bf8ae : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad5edef61f8d441a970b3eb70315c5b7 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938d23aeaae349daa0782b32ecc2e0b2 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79ada528d2644a36b4c2796e232b30ef : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a6de677a93f40bbb00b8166bb83a28d : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03b02b7287864b02a5a9ba702fa35be9 : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dbd604288bd441719d8f7f9a0ce36563 : SUCCESS in 18m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d40ce0fa20c942daaae075cbf2acfda5 : FAILURE in 49m 28s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/30fb7bd72459470f99914bfcb347b950 : FAILURE in 55m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecb8d9d7a4a84bb0b4fc32b1eb62177e : SUCCESS in 46m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/821e04e9098d47c8a7020f08f2dd9f58 : FAILURE in 41m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6c4ba075f87049e2b7f321057ada5e31 : SUCCESS in 24m 44s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/630052682c914a1bb1e717c6f27c667b : SUCCESS in 20m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dba9933bba1401db2da0bc84da0fd9f : SUCCESS in 11m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/82056984c9844e30891b360cf131ef52 : FAILURE in 57m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77d23b1d9b3e40d7ad1e9d34d63228fd : SUCCESS in 1h 04m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2be9972287594dd6a8c9a06b6d7ad5b3 : SUCCESS in 19m 28s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8f52267026154768b6674ed65ed0eac5 : SUCCESS in 22m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a63e5c59f3749cb9917ab2bee1a64ef : SUCCESS in 13m 06s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/5d49f2c587774d9bb2e6cacf96f752dd : FAILURE in 23m 42s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/22ab9a013df140eda1ebf38227ae3b5a : FAILURE in 23m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f74b90cf4b234e1e87c3301a1c425cec : SUCCESS in 25m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c12bd1f125af420d8070fc79b764d44c : SUCCESS in 19m 30s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/eb30ce24d2f44bab8328dd8a39d61a66 : SUCCESS in 14m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d868be82a5374ae3929f8c9291fb8a8a : SUCCESS in 10m 44s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/faa61a6277d848f1ab4a0db6db12a673 : SUCCESS in 18m 33s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8f516f72d8da4e10a005c3e720a2775d : SUCCESS in 16m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb4984a7caf44590808a74099052b530 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"aa9298ba2e4ef868c0a1db6979e178cbc02cc4ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 21:42:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1e834a7bd2e02e302d6ddd97585e0ea3b4ae3c3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 22:17:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"aab1b5772ca7fc1aea97c29eb09bc995ae52dba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 00:03:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6e09e8e874a44e4a1eb7bab1ca54d46 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9200a796b84e2caf39c4b94acc7969 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91bef6c47584427097b0512ebd1807dd : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd4a354ff9254469a42f0770be3f0774 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8c69329de749b2ae2be04aed333656 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a294f308d004e53863f2fb122cb6f26 : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f4ede8abda8472db58ab1804c9319e7 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91b350f5618a4b14b7fca37ca7348591 : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6310a161220a45fe81fd3748ef005a60 : SUCCESS in 17m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4e98b320d8de4b7a963030143bd2152a : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/0ca5f3dce1294673a0f0a6232d68dd16 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41db133d21d142aa900dcb7a204a0cfb : SUCCESS in 53m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/81dc0d07ade540bb900ecf3ee6d33609 : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/367f7e6c4f0a4c8cbe5d930d9f1e0333 : SUCCESS in 24m 16s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7e5860adcc9747179d78d939a5bd943b : SUCCESS in 21m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a45bafcd330b418495c8b4222f3cf7ab : SUCCESS in 17m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9690b4daba3841509293fa3c9363f7f8 : SUCCESS in 1h 43m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3516e044a11149f28a77fb72e60fb558 : SUCCESS in 1h 26m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cab259e753b84db29322c5d3c321f193 : SUCCESS in 22m 54s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/82219448fdd2485192e6ed15362eb5a8 : SUCCESS in 19m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3222107ab2d847699231240f27204d25 : SUCCESS in 17m 18s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/9fad7640bb564ae0b9e2e75d1d4ef9a8 : SUCCESS in 1h 04m 59s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/76000bc5817d42f1a8f05c4ab74f17f1 : FAILURE in 22m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4038a54ccc4495f9969831a81b8b5b1 : SUCCESS in 32m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fee8f7d42c124c8a88ebf868f458a9f2 : SUCCESS in 17m 59s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bf2f19539385494c83056864baa0201d : SUCCESS in 15m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/348beb37dfc9450ea44a90c4d8a32225 : SUCCESS in 13m 17s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e66ebb9df84b4f0e9809e63152523747 : SUCCESS in 21m 02s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c3e6c33760774c81a8018d2a64f9db9a : SUCCESS in 17m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9891b936df384816beed9f0aa2943568 : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f7a5e8811b203bb1243e435c39c6199bee6cfb08","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-26 00:05:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da4b07f61315c004c992890fb4edc92d36cb48ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 02:06:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4970bbeb4b73490db8c87f933c19bc1a : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c4f4eae8e534308a76c2ebd0e3cd2b8 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f3883db71524a7fbc099dbc48bd9e0a : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f04d262204624f8ba0892e1e634f859b : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465e87b1a2fb4c9386707fed3bcc0d6e : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af7ccd3cfb744497ab87f6687fa5ffc5 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92bca151d3714792b58a1c726283a2b8 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6307e9de1db84d29b6f3311de2e9482c : SUCCESS in 14m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f7807ad416644c6b96a08dbbdbba8bf : SUCCESS in 18m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e18a0e12e5ac4a24a81faa854f69c123 : FAILURE in 47m 57s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/161474588191494c90100c251836c3ec : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35314fe86aea4ee0a2687ec72b8ceb66 : FAILURE in 35m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d92a7470edc447ecb8efc321954efb6b : FAILURE in 45m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e380cbb92b3a4e8bb761126bca6d7e1a : SUCCESS in 22m 06s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/81a03a39e6e54b76bc4d57a529b55361 : SUCCESS in 21m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b02126e5362f43bdba93b7a13f0fe15b : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba8d3b37a7f143399fcef0a0ad9acd02 : SUCCESS in 1h 59m 15s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37e9b79f669e43cab4a1b3ca69df3703 : FAILURE in 51m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/264f5626737f46fcad0bdebac579ea25 : FAILURE in 20m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/78e04f43b71c45629974f27c24c97efb : SUCCESS in 19m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b52f207bd2584a588f79c19c18871bb4 : SUCCESS in 15m 19s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/4318211d43464e2ba7e1fce00215b585 : SUCCESS in 35m 59s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7a5f32b6cfaf4858b03e556d0b7e6768 : FAILURE in 41m 54s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70076b0d0884452dac452bd7def369d3 : SUCCESS in 22m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/afa9c24673e3423cbb855edddd6f955b : SUCCESS in 13m 05s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6f8ae62b9ea74ca2a89280fb1e980f33 : SUCCESS in 17m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/174f7d79ba7941308ac2a21d80513cb1 : SUCCESS in 11m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c5605dfeedc54979a8710be04da3f900 : SUCCESS in 14m 43s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d8ae00731a374c9a9748d3041932df42 : SUCCESS in 19m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8805179bf62c409a92978902581f3234 : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":3},{"id":"35cc70b33540b88425cb1f7b05aef8fcaa159a0f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-26 07:25:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d3ffbb09285214efc1f02c2484c278bd1e3863d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 09:19:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d25958aa7b24337bfc68d06358f1d7f : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8bf84837d324dfc9f5d65ce01288ed5 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e31ba6f12fa4ad29755cac78d77237d : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11465b4dea684587b6c08afce2803d4f : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2ab68f3fdb4dfb80d7a84ad5209f72 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aa77fd9810c40b285129a0209edb7e6 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2a363628b40b4eb4be9ef34ef826cc61 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ad367fdf52fe42678e0dfb3533843a07 : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/779bbbd1a6a44e31acf109c040dc00d4 : SUCCESS in 13m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/031259c8e99d4b91a1d3ce35514a6a37 : SUCCESS in 55m 36s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8db6296fe2594ae382599740563d046b : SUCCESS in 1h 10m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da8ab1988a1048dd8edab6b1d0a99b7c : SUCCESS in 44m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ba06d15fa7694a47b71d592df844cf89 : SUCCESS in 57m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/53942efcd9bd410daca7d5d994512fa1 : SUCCESS in 18m 37s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/fdb8dc15d1d846308f12f5e671601815 : SUCCESS in 20m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e10560fe21e14b828e35bcdd8f312464 : SUCCESS in 13m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/099e064b9f3f458a98e96c8935db626b : SUCCESS in 1h 50m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40fbb66577344020b2bcb6b046e01584 : SUCCESS in 1h 13m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0bda22f78c9d454c9da3df8df0a917ac : SUCCESS in 20m 22s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a139dc0bd23e49e1b7201b0c9ca10d3d : SUCCESS in 21m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78723d51896949048e5057726e984c96 : SUCCESS in 15m 46s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/31b226f5447b4be68be08d413fde729e : SUCCESS in 50m 27s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e538eb5ecd8d426bba88e82bd0b9b9c2 : SUCCESS in 41m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef989b3e3bd549aaa2eb1609afa6a70d : SUCCESS in 28m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/07de026b3b38496daafd9cd2f066a07f : SUCCESS in 15m 25s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1976634229cb4af3bf8be0b7fbdbb1ff : SUCCESS in 15m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c97837221bd4859af930ff04c8eb0c5 : SUCCESS in 12m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b6b2b34251c64d7ebe60f4e612d88142 : SUCCESS in 18m 29s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/18f9e87550af4c97b359ca6452e9a9b4 : SUCCESS in 17m 35s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fb8e6ff06814c20bf813b74fc8e763c : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":3},{"id":"200d244cd5b8e35c4da6fb5b04cfc80497f700de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-26 10:25:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b5977c12ad77ad473fa3ccf0503dbea5f8f4a107","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-26 10:39:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f18785ac1ee712cefa2789b744290e68aea794e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 11:53:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5ed9de6e89d448399a202fb0b641998 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a79b72c6ec4c60ba11134011e2f8ea : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad0923e626bb4e8dbf42152d956360bb : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d6a567b636641ffa95476c78d0fc7b0 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e7c4dfa4f834ca1a466a66368f7d5a6 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/141c4d3238db45b099e6e373572aa04e : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/572359c5e106469aa774c0adb9a60547 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6af0d043a4c246ffaafd2c35946b4aa1 : SUCCESS in 15m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/520db058a26542b7b0f03b3b683fbaac : SUCCESS in 18m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/698ce4fe0ace4a29bbb0fbaff2521cb4 : FAILURE in 55m 52s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/573e88f29cbb4c598cc555cb5ebb735c : FAILURE in 54m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a63cdb23a32f4a88980df98efcc4fead : SUCCESS in 52m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8285b148040242fba4dade3787583361 : FAILURE in 56m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2e0bb40dea1f42af8ef0b68f7500d28c : SUCCESS in 20m 53s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/df814dea91044757bed96f3f25f16396 : SUCCESS in 19m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a28d1388c8c4682be5886f9ca5662e9 : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e7502ae8e06d470ca402010f8131687d : FAILURE in 1h 00m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fedfeaeaed114dc38f1543ea3ee20bfa : SUCCESS in 1h 21m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c96ad81bead442afad41469fab74c054 : SUCCESS in 20m 58s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4181df21212047f296b020ab2353caf3 : SUCCESS in 19m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94dc7671c542470f82810aecd3b06642 : SUCCESS in 16m 14s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/2971eb0ab6d44273abee8dc96a97ec8f : SUCCESS in 45m 13s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ffe31ad3ebd94bc3b353fa36a561047d : SUCCESS in 40m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4cd21f6c5734445b1c1e617d130508c : SUCCESS in 30m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4d4a2cf641044832bb8a495a67d2694b : SUCCESS in 17m 34s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ea0b9858b6874ff4a34ccda5bf4e7b5e : SUCCESS in 14m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/585ed788d84e4ec5bb24bf6e9f031d10 : SUCCESS in 12m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/75636cac0a9d420c91c97a2a018870d4 : SUCCESS in 19m 19s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6b87a02c61cc44bf8e876e5dda907508 : SUCCESS in 17m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8eed6ec4194e4260a7bf6fd521d69403 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":4},{"id":"fedbc44d355c2255e7c463e40882d5863e7fa414","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-26 13:38:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"54a2f8927e2c69da776b416182467444f6b6b858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 15:28:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/963c020eea104ef1befbc9c85fb4b885 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b5faccd3f144501b19c749356459279 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d231567c56d4857a9f05bdc6af1a226 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d5b4d6d194d49ea9084b37e1c037a6a : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bf5079157754329859b5cde977b7ba0 : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77b91c250c5448a38eaa25610e7266b8 : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/905cc4952de544369f9804732c5cce04 : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d012ad5d6baa41099a988c0e69e76b78 : SUCCESS in 14m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aef4a1e6144c402bb20d671f3bdfbf4c : SUCCESS in 17m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7a84a6724a104fb0875bb21fa9c5c175 : SUCCESS in 1h 08m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a9647c0eb31841c5bc13d761525a98ec : SUCCESS in 1h 05m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c87a219ac3540f7b43796be71a931ac : SUCCESS in 55m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cb1adb5e59aa4939baeac2bfc995a2f8 : SUCCESS in 1h 33m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bd7c1437c5cd413dad8ff56840f08a03 : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/aeeba3cb1f4b4894b6cae105599bd7b1 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21fa0cd623e6451888df6bdddce4c047 : SUCCESS in 15m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3fabc241607d40e9bd085b5b2a251f49 : SUCCESS in 1h 34m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd8394f9e94743768c761163929fb50a : SUCCESS in 1h 42m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a4df0427d2ba482f9c6860b90cd1d9a5 : SUCCESS in 21m 21s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b1fc956d94a24f44a54f66bfca737f7a : SUCCESS in 19m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cf50d98445d4c8ba2a2c39dc7c9585a : SUCCESS in 14m 06s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/bb543258b2e24ca09ad5c456ded9222f : SUCCESS in 51m 09s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/001ce68762f04694a8fa8da64a2748cf : SUCCESS in 32m 48s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/148e185bac5243eb88401de96cae1c32 : SUCCESS in 35m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dc0ca8cd7e9040c69d800ee177018d05 : SUCCESS in 14m 45s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/980fce8be6424b469b16603c907f09c0 : SUCCESS in 16m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c999f369692641bb81611d3251725618 : SUCCESS in 11m 33s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/eee5b6bf20a44c229eebd049832e47c9 : SUCCESS in 17m 36s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/756276c4806948f2912d7b54192f7910 : SUCCESS in 17m 42s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f6ee00db9744847b7b9ddafcb33a6b1 : SUCCESS in 11m 37s","accounts_in_message":[],"_revision_number":4},{"id":"a4a4b55382add3fdfee516b633ea541bb8382673","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-28 05:41:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4d832f58849b5c710e71b861c9e34fb031076ba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 05:41:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"891acea8485305449568d7ab6ec1dfdb3d5cd0b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:07:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb5c424974954d988b22e93ae257c059 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b978784826fc44f59b1a33dae4c8b241 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96add97199c74105ac6ba44ad6cba269 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71c225e0052f4f91be5e5705cf71bad1 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6672c73d966418288413b317ac50ed1 : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6bdeeaf6f8d541ef9ce66763008bc4f0 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5cd3955d05f4d0486ac1f6028a15a95 : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8b9d4f0d76a4834b9c37dc01cc74b53 : SUCCESS in 14m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3210022f9bf349e19b6d9c01b7544ab2 : SUCCESS in 53m 40s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/50c541a7f3754a1091f59baaa312c949 : SUCCESS in 1h 12m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29e8e19d0aa34d4f9da8179579ec67c1 : SUCCESS in 44m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d088f50aa84d4c03861be1f966fd11d6 : SUCCESS in 56m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/be1e340a3eba431bb4aca77ffdfc01b9 : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e32a5323724d4612aff4f73039ec4cb7 : SUCCESS in 20m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f33afcc4802e42baacef090a3ddb5828 : SUCCESS in 11m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8b834ef130714683b993a984e8f3f720 : SUCCESS in 1h 25m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68a9c39e25ab43c7b15ed280875a8b8e : SUCCESS in 52m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/18dce78a652e4ad1abcaae1b480e5aac : SUCCESS in 18m 05s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/939e3448192b4cff88e78a55c6d541cb : SUCCESS in 18m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c6af7ec32ad4a9cacb2a54332f6a5d6 : SUCCESS in 15m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5f789e9f084c4a4fbccd9e7118d4d7e6 : SUCCESS in 29m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1e2527ec0cb4daca37906b804525007 : SUCCESS in 25m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2ca1d1c35e27412a91b6ba8e5241242d : SUCCESS in 13m 33s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/922f593e544e4266a385b9f1dcfbfb68 : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13cdce2bafbe41c3a632dcb15f6fd545 : SUCCESS in 12m 45s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/811f33d5fdf64a15a409167e83d5ea32 : SUCCESS in 16m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5d937466a1a54762b4db753da7ee907e : SUCCESS in 15m 41s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed9b9944baf448a4beff26598f6d7b98 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":4},{"id":"d14fb390c75ec5b498d9d80334d59a44de534e55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:07:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b34e067183e439874c7e81a620231aa0c8ecb3a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:08:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b66e3f158442d6be4a1c74b97f2c92 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c77ac811b946ac8847cfc95e14d0f6 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2f3a8f80d1329e90076b307e18045a72c842e280","revisions":{"8cdf4ee027a4e513b97a214172b3a2ddcce3413e":{"kind":"REWORK","_number":1,"created":"2022-05-25 18:01:19.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/23/843323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/843323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/843323/1"}}},"commit":{"parents":[{"commit":"64c0c1e50eaef3190d8b5e48fe7800eadc0d1aaa","subject":"Merge \"kolla_passwords: add no_log for password overrides\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64c0c1e50eaef3190d8b5e48fe7800eadc0d1aaa"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-19 22:04:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-25 18:00:43.000000000","tz":180},"subject":"Fix forgotten hacluster regexp for image build","message":"Fix forgotten hacluster regexp for image build\n\nThe \u0027overcloud container image build\u0027 command didn\u0027t build all the\nhacluster images when hacluster is enabled.\n\nTrivialFix\n\nChange-Id: I9150e32579421e46782518948188e1363918fb39\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 3e7b01ad829e0c14396045cf147e719ad0a89ed2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8cdf4ee027a4e513b97a214172b3a2ddcce3413e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8cdf4ee027a4e513b97a214172b3a2ddcce3413e"}]},"branch":"refs/heads/stable/yoga"},"12a5ac9458dae0c439d5ffd2ddc6f55a80b4bcc8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-25 21:42:24.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/23/843323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/843323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/843323/2"}}},"commit":{"parents":[{"commit":"9819f7854f78811c612474374d9bddf0709b3d9d","subject":"Use yoga upper constraints to avoid Python version conflict","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9819f7854f78811c612474374d9bddf0709b3d9d"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-19 22:04:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-25 21:42:24.000000000","tz":0},"subject":"Fix forgotten hacluster regexp for image build","message":"Fix forgotten hacluster regexp for image build\n\nThe \u0027overcloud container image build\u0027 command didn\u0027t build all the\nhacluster images when hacluster is enabled.\n\nTrivialFix\n\nChange-Id: I9150e32579421e46782518948188e1363918fb39\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 3e7b01ad829e0c14396045cf147e719ad0a89ed2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12a5ac9458dae0c439d5ffd2ddc6f55a80b4bcc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12a5ac9458dae0c439d5ffd2ddc6f55a80b4bcc8"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"de6327119d0e4893f896452c1a2fd6c0768b5b31":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-25 22:17:38.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/23/843323/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/843323/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/843323/3"}}},"commit":{"parents":[{"commit":"c04f315fbad54c00c0510c6451582603238d5f0f","subject":"Use Yoga upper constraints to avoid Python version conflict","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c04f315fbad54c00c0510c6451582603238d5f0f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-19 22:04:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-25 22:17:38.000000000","tz":0},"subject":"Fix forgotten hacluster regexp for image build","message":"Fix forgotten hacluster regexp for image build\n\nThe \u0027overcloud container image build\u0027 command didn\u0027t build all the\nhacluster images when hacluster is enabled.\n\nTrivialFix\n\nChange-Id: I9150e32579421e46782518948188e1363918fb39\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 3e7b01ad829e0c14396045cf147e719ad0a89ed2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de6327119d0e4893f896452c1a2fd6c0768b5b31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de6327119d0e4893f896452c1a2fd6c0768b5b31"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"2f3a8f80d1329e90076b307e18045a72c842e280":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-26 10:25:12.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/23/843323/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/843323/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/843323/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/843323/4"}}},"commit":{"parents":[{"commit":"64c0c1e50eaef3190d8b5e48fe7800eadc0d1aaa","subject":"Merge \"kolla_passwords: add no_log for password overrides\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64c0c1e50eaef3190d8b5e48fe7800eadc0d1aaa"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-19 22:04:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-26 10:25:12.000000000","tz":0},"subject":"Fix forgotten hacluster regexp for image build","message":"Fix forgotten hacluster regexp for image build\n\nThe \u0027overcloud container image build\u0027 command didn\u0027t build all the\nhacluster images when hacluster is enabled.\n\nTrivialFix\n\nChange-Id: I9150e32579421e46782518948188e1363918fb39\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 3e7b01ad829e0c14396045cf147e719ad0a89ed2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f3a8f80d1329e90076b307e18045a72c842e280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f3a8f80d1329e90076b307e18045a72c842e280"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
