)]}'
{"id":"openstack%2Fkayobe~843347","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I40c6b898963c2218d41d37bd73d40ce8dcf22b87","project":"openstack/kayobe","branch":"stable/yoga","topic":"yoga-uc","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-05-26 10:20:59.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-05-26 10:15:00.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-05-26 10:20:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I40c6b898963c2218d41d37bd73d40ce8dcf22b87","subject":"Use Yoga upper constraints to avoid Python version conflict","status":"ABANDONED","created":"2022-05-25 21:41:16.000000000","updated":"2022-05-26 10:20:59.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"40b3af887b79cd3d4eed3fe0ada18846da33df6c","_number":843347,"virtual_id_number":843347,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-26 00:04:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2022-05-26 09:56:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-25 21:41:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-05-26 00:04:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-26 09:23:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-05-26 09:56:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"1c35358740d8e7a275298aaa54ffba9cafa96e43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 21:41:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3785e97f46c6a1ea5a2f9601bb2de3061da78d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 22:16:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"111744988ed06ac23c3e31e16b7acd792d0152f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 00:04:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70b2232021d84db09ece7f0b6fd38a91 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91504170fe414d8eb70d4d7075e9d311 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81243a7efa284673870aca3257366074 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e68b10979c846839d69e67add74c62e : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cee1ea468e314ad7bdd90879bd74c69c : SUCCESS in 10m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19df69355092494abe35733badf47274 : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f1ae6753d2049c5872f82d1c7763d9a : SUCCESS in 15m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4c18a823b0640c3bd4005d448549a16 : SUCCESS in 17m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8526a8696c8c4bb88782cb73dfc1764a : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/faccb0ce6d004030bc9a6facb28da4ab : SUCCESS in 1h 14m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c1f28e6a74c494eb860f7fd0baf8f15 : SUCCESS in 47m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b0b99447695d49058f9789861074a403 : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/80813dc7c6a4429ba73f03737ac087a1 : SUCCESS in 23m 07s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b193e2ee35da4be08daf920b1e7af221 : SUCCESS in 22m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a8f0104f50845f2aa48525608a4eb63 : SUCCESS in 12m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bc8c2b05fe484d748f31f99b8cf62f97 : SUCCESS in 1h 45m 30s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b75ace72dfd141be9c38ca5efdfcb7d0 : SUCCESS in 1h 16m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/60722a4fa50d4a5fa9259977dc413c98 : SUCCESS in 20m 11s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2d88ee7dc8714f509f99645e5eeff9aa : SUCCESS in 20m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b74e321dd1c4ba6b9d000941390aff3 : SUCCESS in 16m 06s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/2452c32a32204ba9a160c588a00ce77f : SUCCESS in 56m 32s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/30a0219deb034c8698af75c90ac1d666 : SUCCESS in 47m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa671ba6d5a34504bdb141aa155eba4f : SUCCESS in 26m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7dd8f75214764e1e9e726ba4f8cda4c1 : SUCCESS in 17m 21s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/749d3a7e705e4fb7ab9d706b4aeb10ba : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b24560fce5af4221b77e3847c8a07daa : SUCCESS in 10m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/742dcd4e841146f5a12a5c3795f11316 : SUCCESS in 16m 53s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/809c7750704a4e25a90f3ca35fd4ecf5 : SUCCESS in 17m 27s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bb47d24323b437d9d1278706e50b66a : SUCCESS in 14m 59s","accounts_in_message":[],"_revision_number":2},{"id":"86ac15acf5e3ca0ec96303612404099f82d33207","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-26 09:56:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4eaeaf9106e3b545bb9e875649cd53f132de507","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-26 10:15:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b682f576628b7ec68eb898c92af17f6d3f0ce53","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-26 10:20:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40b3af887b79cd3d4eed3fe0ada18846da33df6c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-26 10:20:59.000000000","message":"Abandoned\n\nin favour of https://review.opendev.org/c/openstack/tenks/+/843424","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c04f315fbad54c00c0510c6451582603238d5f0f","revisions":{"9819f7854f78811c612474374d9bddf0709b3d9d":{"kind":"REWORK","_number":1,"created":"2022-05-25 21:41:16.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/47/843347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/843347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/843347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/843347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/843347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/843347/1"}}},"commit":{"parents":[{"commit":"64c0c1e50eaef3190d8b5e48fe7800eadc0d1aaa","subject":"Merge \"kolla_passwords: add no_log for password overrides\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64c0c1e50eaef3190d8b5e48fe7800eadc0d1aaa"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-05-25 07:41:01.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-25 21:38:36.000000000","tz":180},"subject":"Use yoga upper constraints to avoid Python version conflict","message":"Use yoga upper constraints to avoid Python version conflict\n\nRequirements upper constraints bumped python-novaclient to version\n18.0.0 [1], which requires Python 3.8 [2]. This results in failures when\ninstalling python-openstackclient on CentOS and Rocky with Python 3.6.\n\n    ERROR: Cannot install python-openstackclient\u003d\u003d5.8.0 because these package versions have conflicting dependencies.\n\n    The conflict is caused by:\n        python-openstackclient 5.8.0 depends on python-novaclient\u003e\u003d17.0.0\n        The user requested (constraint) python-novaclient\u003d\u003d\u003d18.0.0\n\nWork around this issue by using yoga upper constraints until we upgrade\nto CentOS Stream 9 and Rocky Linux 9.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/842808\n[2] https://review.opendev.org/c/openstack/python-novaclient/+/838944\n\nThis is partially cherry picked change to solve only one issue.\n\nChange-Id: I40c6b898963c2218d41d37bd73d40ce8dcf22b87\n(cherry picked from commit d2e0d64eb00d4cea8a4f8ff6a963b1ec0c3660ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9819f7854f78811c612474374d9bddf0709b3d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9819f7854f78811c612474374d9bddf0709b3d9d"}]},"branch":"refs/heads/stable/yoga"},"c04f315fbad54c00c0510c6451582603238d5f0f":{"kind":"REWORK","_number":2,"created":"2022-05-25 22:16:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/47/843347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/843347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/843347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/843347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/843347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/843347/2"}}},"commit":{"parents":[{"commit":"64c0c1e50eaef3190d8b5e48fe7800eadc0d1aaa","subject":"Merge \"kolla_passwords: add no_log for password overrides\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64c0c1e50eaef3190d8b5e48fe7800eadc0d1aaa"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-05-25 07:41:01.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-25 22:14:37.000000000","tz":180},"subject":"Use Yoga upper constraints to avoid Python version conflict","message":"Use Yoga upper constraints to avoid Python version conflict\n\nRequirements upper constraints bumped python-novaclient to version\n18.0.0 [1], which requires Python 3.8 [2]. This results in failures when\ninstalling python-openstackclient on CentOS and Rocky with Python 3.6.\n\n    ERROR: Cannot install python-openstackclient\u003d\u003d5.8.0 because these package versions have conflicting dependencies.\n\n    The conflict is caused by:\n        python-openstackclient 5.8.0 depends on python-novaclient\u003e\u003d17.0.0\n        The user requested (constraint) python-novaclient\u003d\u003d\u003d18.0.0\n\nWork around this issue by using yoga upper constraints until we upgrade\nto CentOS Stream 9 and Rocky Linux 9.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/842808\n[2] https://review.opendev.org/c/openstack/python-novaclient/+/838944\n\nThis is partially cherry picked change to solve only one issue with\nTenks deployment where upper constraints from master used.\n\nChange-Id: I40c6b898963c2218d41d37bd73d40ce8dcf22b87\n(cherry picked from commit d2e0d64eb00d4cea8a4f8ff6a963b1ec0c3660ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c04f315fbad54c00c0510c6451582603238d5f0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c04f315fbad54c00c0510c6451582603238d5f0f"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
