)]}'
{"id":"openstack%2Fkayobe~843350","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I40c6b898963c2218d41d37bd73d40ce8dcf22b87","project":"openstack/kayobe","branch":"stable/xena","topic":"yoga-uc","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-05-26 10:22:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I40c6b898963c2218d41d37bd73d40ce8dcf22b87","subject":"Use Yoga upper constraints to avoid Python version conflict","status":"ABANDONED","created":"2022-05-25 23:41:54.000000000","updated":"2022-05-26 10:22:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a643713260388c3541e49f1c5cc2a3790bd47b5e","_number":843350,"virtual_id_number":843350,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-26 01:38:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-25 23:41:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-05-26 01:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-26 09:23:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"511cb5539fed8883dd1fc19e7d5d4708a07cd3c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 23:41:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99b988b9d11982173cb96aa9dc2718df0f796068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 01:38:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9df58fc5fb64489e9c3fc686fea05ff8 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12c4244eaeea4b579957c233626ef2cc : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6f5e36f09534a0b80f9f912f8de0667 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9fdb8ab968046d29482e0a34ba57c2c : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c5eee82e445473ba9dee03897555730 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2181e55e9cb4480eb8b42b7b75eb43cf : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6dcf9339ca2d4ca7a84779cebaedb040 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc8674825ac1492b93b96bd140008dd2 : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/badb391c26ab49fd91c71d7427c0afec : SUCCESS in 16m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0cf4d75280654bb984faff1fbcaa6b37 : SUCCESS in 1h 00m 06s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8872e169456e4a39a3928a87cdc833c1 : SUCCESS in 1h 13m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba177a557b5a43c0808643990b49eb37 : SUCCESS in 48m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1337a65868f6437091aae90de65781d0 : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/aec0f9b84b164715872c20e6ca56114a : SUCCESS in 20m 39s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/35100014888540dc91990c15a34efe0c : SUCCESS in 23m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63ae8a92c64e4e9b9deeea432ad5f5f4 : SUCCESS in 11m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3298438a280c471faffbfe8a4fa15863 : SUCCESS in 1h 52m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9f9f88ec1bb43daa4764a82e48f28b4 : SUCCESS in 1h 01m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7d67569fbb1444739e7ede1e84498aca : SUCCESS in 17m 32s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/29f3a456b57040a195b511c395d14fc1 : SUCCESS in 19m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4018a87c41104f87aee648d025ddd2e9 : SUCCESS in 13m 53s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/2c2b67781fc84dc990788e3d94254114 : SUCCESS in 52m 07s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/287955adb28c4f7387c9de371d19230b : SUCCESS in 24m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/851ba7b8aed24679a3f7c6a2cff081bb : SUCCESS in 19m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0e3c0d00d9d94948b4ff394589da5c07 : SUCCESS in 18m 22s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/eea1e7129bde44049f829547d95c167e : SUCCESS in 17m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e022a950913420784af4ecaa32c5dda : SUCCESS in 9m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/44da0df3266c4943bb793b768cc6f0ab : SUCCESS in 15m 04s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/05fbaad454204063ae7faaf2c855b924 : SUCCESS in 19m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd874b8dba544a819c87f743e812494c : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a643713260388c3541e49f1c5cc2a3790bd47b5e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-26 10:22:18.000000000","message":"Abandoned\n\nin favour of https://review.opendev.org/c/openstack/tenks/+/843424","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f4c872bbda49f934ded4b82973fcf4c4fc23c52","revisions":{"3f4c872bbda49f934ded4b82973fcf4c4fc23c52":{"kind":"REWORK","_number":1,"created":"2022-05-25 23:41:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/50/843350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/843350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/843350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/843350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/843350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/843350/1"}}},"commit":{"parents":[{"commit":"c65e3cceeb5062f513d34c791c2d3e3464e6dc4c","subject":"Merge \"Fix Bifrost inspection through DHCP-relay\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c65e3cceeb5062f513d34c791c2d3e3464e6dc4c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-05-25 07:41:01.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-25 23:41:35.000000000","tz":180},"subject":"Use Yoga upper constraints to avoid Python version conflict","message":"Use Yoga upper constraints to avoid Python version conflict\n\nRequirements upper constraints bumped python-novaclient to version\n18.0.0 [1], which requires Python 3.8 [2]. This results in failures when\ninstalling python-openstackclient on CentOS and Rocky with Python 3.6.\n\n    ERROR: Cannot install python-openstackclient\u003d\u003d5.8.0 because these package versions have conflicting dependencies.\n\n    The conflict is caused by:\n        python-openstackclient 5.8.0 depends on python-novaclient\u003e\u003d17.0.0\n        The user requested (constraint) python-novaclient\u003d\u003d\u003d18.0.0\n\nWork around this issue by using yoga upper constraints until we upgrade\nto CentOS Stream 9 and Rocky Linux 9.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/842808\n[2] https://review.opendev.org/c/openstack/python-novaclient/+/838944\n\nThis is partially cherry picked change to solve only one issue with\nTenks deployment where upper constraints from master used.\n\nChange-Id: I40c6b898963c2218d41d37bd73d40ce8dcf22b87\n(cherry picked from commit d2e0d64eb00d4cea8a4f8ff6a963b1ec0c3660ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f4c872bbda49f934ded4b82973fcf4c4fc23c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f4c872bbda49f934ded4b82973fcf4c4fc23c52"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
