)]}'
{"id":"openstack%2Fkayobe~843664","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Id9418e4e88c52056412daa22462aa611bfcc59ae","project":"openstack/kayobe","branch":"stable/xena","topic":"ironic-mtu","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-05-30 21:45:52.000000000","reason":"Maksim Malchuk replied on the change"}},"hashtags":[],"change_id":"Id9418e4e88c52056412daa22462aa611bfcc59ae","subject":"ironic: Set MTU on provisioning and cleaning Neutron networks","status":"MERGED","created":"2022-05-27 17:52:12.000000000","updated":"2022-05-31 01:40:28.000000000","submitted":"2022-05-31 01:39:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843664-ironic-mtu","meta_rev_id":"d0417930f57813d71fd49cfbda4bdcba22635d4a","_number":843664,"virtual_id_number":843664,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-31 01:39:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-05-30 10:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-30 07:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-05-30 10:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-27 17:52:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-05-27 19:54:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 07:08:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-05-30 10:39:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"53a3bee5ce780c8acf8056f71b0752af8cb8820d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-27 17:52:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d23a921e76bad9605e62bcb7c1d02eeb53f54733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 19:54:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/496db549eeba467ab33f7584d012e826 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/140ffef6ba624f31b6890e9046bb7d3f : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1185b42b930843919de190ea1e3e10f2 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/692774391bb940f69e123608f83b375c : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/841a4942cdb54669b8da77689cb5491c : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f03077762e4c21a5fc6d2d17c4530c : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5a623b56f134104920fd3cd2a369497 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec3cd6d344ab4d00bca5dbd4cb57a5dc : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/293fe97443c94efa913a6921769451d9 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b81bcf5cd5204cbda22ce2b6607d5b95 : SUCCESS in 19m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b94580c701db40b58476b487efcd959c : SUCCESS in 1h 11m 01s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/dba53c7b6ba741c6af59fff6538b6fd9 : SUCCESS in 1h 16m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/327fbb794a914a23b0f215c061b8ec1f : SUCCESS in 1h 00m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/00e2d8b5cb324b1581d23c00ce97b5a4 : SUCCESS in 1h 15m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d6858c3854c84a93afd4e501f3f38568 : SUCCESS in 26m 52s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c37a4da207c74c1fbdef517d327de2b3 : SUCCESS in 23m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8a85753619142e7a74752ded5191c19 : SUCCESS in 16m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d02766f9698b41939e15ed7b4f737f8a : SUCCESS in 1h 57m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ac97a11a58443348f06bc1a4ad8052b : SUCCESS in 1h 18m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f66db2703ecc4c058f53bb652e0aadba : SUCCESS in 20m 47s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/add9c240eec04f4498f819c136572afb : SUCCESS in 18m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0537d917fd5464185020fc6ae15a73c : SUCCESS in 16m 35s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/324a0e90d43c4141a532ca4793c661ee : SUCCESS in 50m 02s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2e5a17fced9d47e98afc9ef21280faa6 : SUCCESS in 33m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20a4ea3f84024be1a189b4b6d5824afb : SUCCESS in 27m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/174071ed6e6749d1ac599f649b24baf4 : SUCCESS in 17m 50s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5a6e11d15ab84c8d80fa6ba4a1e88b28 : SUCCESS in 18m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b317aee7dad480eaf0fb1802d07e89f : SUCCESS in 12m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c5978b1e5c5e4fdd81f2146edc7fc2f4 : SUCCESS in 16m 14s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2a7b108feddf4562bc363e1eb2bbf527 : SUCCESS in 20m 09s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/878a5e2a83554b74a66aca634e9d99ef : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b082ccca9b4173fb7e7dc3378bdf0acf36a68e45","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-30 07:08:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff74fdc75d9c6e62ea7d0213e1c84e4038efa2d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-30 10:39:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64350be821d54e6ca74706e7ebdcddcaf29a605b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 10:40:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec611990e4a2124e5f7d7891aa33518b1f491025","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 14:59:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e1bfd4104440b3b904ed38159a61c3 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4596b337d29941f891683b67e4512ea4 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db6f3461cedd4ff382eb52b8eef2ff21 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a1cca40d9d4f809de80fda5a7e1984 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cf9a740679441d89334ba944bdea860 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8a9e49290d74b6ca2bdc86a8113cf28 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9547af0282ab4649935e081fd369a363 : SUCCESS in 15m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/529916bd870b4356b009ae27071afbcd : SUCCESS in 19m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/47f64129387444b1832837896c185033 : SUCCESS in 1h 17m 48s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7524dc591af64a3bbde57898ac45d633 : SUCCESS in 1h 11m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/956388f54d7543a98b3d61407bf721c2 : SUCCESS in 1h 06m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f10dbe79dfee4b5fa341f9e3a414caa8 : SUCCESS in 1h 15m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0cb77c4e988f4f5d93bfd636df5ba62c : SUCCESS in 26m 16s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/bce09b97f1c7434180b441bac1e83dd4 : SUCCESS in 24m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c7085afdf774ea084ae00fe6ead34d2 : SUCCESS in 17m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4e2a7bf7f9864bd78ec6e115f7ef79fd : TIMED_OUT in 2h 35m 47s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1126bee8b22a415cb8945c7280195f55 : SUCCESS in 1h 23m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/432fe903a15149b1bcf0bbeb46ff2188 : SUCCESS in 22m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1c4f7438bb8142a0924e4302f40c9c58 : SUCCESS in 21m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfa1cb5e044148419e998c317e713733 : SUCCESS in 16m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1ec7833de3b74aac9fd291f08f2d1665 : SUCCESS in 28m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faac8ad156a9450b958a2d7d3a9aa6c0 : SUCCESS in 19m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2fd8c2bce06b4aaa86ddd6b75128eb1f : SUCCESS in 17m 24s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bc9cb81d9256480d861041f81b48ef60 : SUCCESS in 16m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7898cf72263e441eb140cc99fa0b0525 : SUCCESS in 12m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/77dccaf7b67e416d82e1e61166427b05 : SUCCESS in 15m 50s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3888c59ab13d4d1b9e8bcd7d533562a5 : SUCCESS in 17m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e1e918c7c3c4fec8a5484ceea07b411 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"710497e95f8e76dc2bec816786b99bb70beaefa2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-30 21:45:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5e2fe4dc00e0d98ce24d2ce65d8968e83467ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 23:41:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35a9eb86af8842f18ca06a0b00813c8b : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e64b23be5c46f3b0b824ca7f8c941f : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fcb3cf4d36345ab9f98e3eb1321d077 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7eba04b297ce4b069e59d43dfe738739 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b49f638f017a4df29e77e12e5cea29a0 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92365e4db8b24f4eb0b3d31dd3f8d76d : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5097dd68bdd4bcf955ebf68cbba0d71 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b631c62ab6e842f59fba1bb9f18ac8fd : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36fcf017d36f4a268f3234256bebb09c : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/947ddc56383f40f49aa643929e164326 : SUCCESS in 18m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/309a826791784e01b4473ab8e0b4af1d : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/716bcda47bd4493098556d242b07bf0a : SUCCESS in 1h 13m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f7473233bb74a5bad5094817428ad87 : SUCCESS in 50m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e296ba41c0c541b5920303f9574aa931 : SUCCESS in 1h 03m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d844b0e83ef94e0db0d3d633628b797a : SUCCESS in 23m 55s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/488089dd14d3435f91b4a96bd76f6afa : SUCCESS in 23m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/013012f2cb414953a12a35cc52c64905 : SUCCESS in 15m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a291e1f428e04cd7bfb47ac6348d79d8 : SUCCESS in 1h 47m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ce265f20b60442388db7ca4827e06ae : SUCCESS in 1h 13m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/43453d791be04045a7ead39a0d7c43af : SUCCESS in 19m 49s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2eb5ff0b922d472e8a32b50dcd5a82f2 : SUCCESS in 22m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ce8c313f4844aa3a1d1acca4ff40043 : SUCCESS in 15m 15s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/66175850a53945db92f7fbf280fd23ec : SUCCESS in 44m 51s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/72d71981ae7b45fb924c8ba57a07a9c4 : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5058cb409ec54d0bb8a974f11f2830bd : SUCCESS in 21m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c6c40b226e5c4e4c9eed4b3e36bdbaae : SUCCESS in 16m 22s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/901dab3afe324c5b97acd5357aedf75a : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd1070e4a1ad4ba589a22dc9a23665cc : SUCCESS in 11m 29s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9ba16f94a9364873959d4cb65d8fc97e : SUCCESS in 20m 08s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/38f2afc25dc442b88f72f4b9686a388b : SUCCESS in 16m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/592dd46d9e1745a3bf1f2ef4d9eeddd1 : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7214faacef4d0f62ce6968de4cc7cacbd660de15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 23:41:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2867fc592e65bbabfe2aef9b4279d6146853c683","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 01:39:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daafbe010d1a46ebab1a8727398f71cc : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9aef976fac04e39a385fd022c7c8c73 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13f43eb4d5a8494194ee6c1a81005bcc : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d687377cc0f04dd8b681ae964e439e40 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3c38103a4f94cb1a49a2e2b62e7febf : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a2df9e3431340768aa28e341ba1aed3 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0df06dcdc54d4c978b129c2905ff559e : SUCCESS in 13m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/27da2fd9413640c3a8b21a77a07a20a2 : SUCCESS in 16m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5f82c13908984bbc9a0f9827adbd3a71 : SUCCESS in 56m 16s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f6a5c6fa57044de8bd1ea49950633274 : SUCCESS in 1h 09m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2aa9d0df9bc24056b3306bb2fa7e6dd9 : SUCCESS in 46m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a4f7001662b74dd892dbd8090e957e87 : SUCCESS in 1h 09m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3ad210e90f0c458b81c8718d6a4b3b35 : SUCCESS in 18m 53s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f51083f3e02d451e953444f627047112 : SUCCESS in 24m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/345cc3a2e58a4d8a9d5836e63e2e977c : SUCCESS in 12m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/640fc14ae1884fa29f411049127a9382 : SUCCESS in 1h 56m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b09e620dac59465f9ceecc4d86c044cf : SUCCESS in 57m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7297ab9305b44b488b763c7c8ff01b54 : SUCCESS in 20m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6384d196e5224b41bd39515d95af7fe6 : SUCCESS in 19m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee5ae8b4d646435ca2d1e675d0019824 : SUCCESS in 12m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8004ad40244f4b50afda2ffe8a59a042 : SUCCESS in 21m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0f189c9a7d047eea1d527b64d466ca0 : SUCCESS in 24m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f09d3f16defd48d5ba54bbd21644afda : SUCCESS in 13m 07s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/58de53b47c6d40798a9887c1125ed369 : SUCCESS in 14m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9e38bb11951472b9657b600aad62a67 : SUCCESS in 10m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3bea1b030fa14c3aa4a353a568cbca51 : SUCCESS in 17m 13s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5f2e314494384d918943d4890a374deb : SUCCESS in 16m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28d08da9c816407397b15f87a093b656 : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"945b1c52defe376412d557f969d1697bb4925036","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 01:39:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d0417930f57813d71fd49cfbda4bdcba22635d4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 01:40:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3dc00eaaac54f69bc8252344009b833 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ec68aad2c3f40c5af01a4221f250867 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0467484a6b343c204c136fe0f6cd3d3d1e8ed946","revisions":{"0467484a6b343c204c136fe0f6cd3d3d1e8ed946":{"kind":"REWORK","_number":1,"created":"2022-05-27 17:52:12.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/64/843664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/843664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/843664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/843664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/843664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/843664/1"}}},"commit":{"parents":[{"commit":"c65e3cceeb5062f513d34c791c2d3e3464e6dc4c","subject":"Merge \"Fix Bifrost inspection through DHCP-relay\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c65e3cceeb5062f513d34c791c2d3e3464e6dc4c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-29 16:09:47.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-27 17:52:01.000000000","tz":180},"subject":"ironic: Set MTU on provisioning and cleaning Neutron networks","message":"ironic: Set MTU on provisioning and cleaning Neutron networks\n\nPreviously we were not applying an MTU defined in Kayobe networks.yml to\nthe provisioning and cleaning networks in Neutron. This could lead to\nissues when nodes communicate with the Ironic and Inspector APIs.\n\nChange-Id: Id9418e4e88c52056412daa22462aa611bfcc59ae\n(cherry picked from commit cd24b6d98e1b82ab9c255e3e5ddbaed99c49067f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0467484a6b343c204c136fe0f6cd3d3d1e8ed946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0467484a6b343c204c136fe0f6cd3d3d1e8ed946"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
