)]}'
{"id":"openstack%2Fkayobe~843665","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Id9418e4e88c52056412daa22462aa611bfcc59ae","project":"openstack/kayobe","branch":"stable/wallaby","topic":"ironic-mtu","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-05-31 21:46:40.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-05-31 21:46:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9418e4e88c52056412daa22462aa611bfcc59ae","subject":"ironic: Set MTU on provisioning and cleaning Neutron networks","status":"MERGED","created":"2022-05-27 17:52:36.000000000","updated":"2022-05-31 21:48:26.000000000","submitted":"2022-05-31 21:46:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843665-ironic-mtu","meta_rev_id":"df8733a4992e53af94fa21504cdc84eb4ecb962e","_number":843665,"virtual_id_number":843665,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-31 21:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-05-31 21:48:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-31 19:56:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-31 19:56:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-27 17:52:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-05-27 19:48:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 07:08:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-05-30 10:39:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-05-31 16:38:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"69fbacb3a65d7aee72eb89f4126e8773ddcb75d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-27 17:52:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cbda2cd62d9c41a1e1602ce777a5e69401b6f52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 19:48:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/180ffdea50424aa89f75a579bd371e20 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39bb439f68294d42bad20acc804f6d3d : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d64ffb310e2a422fb5355dfc8532f876 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed7f4f04a4af43818922b6912061b763 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc9fec11c7664079975268b44da81e59 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e5dacb457a047ec9929a27b376bdcfa : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b8da22829fd44d292146fccc78986cc : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dcc85411265b485dbba547ff65d153ea : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4d147f90f684d9e827e8f0995bf9cff : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/875297c314264edd81d123fc5a65fd4e : SUCCESS in 19m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b4fbde72de4640feb31aaa904488f4d0 : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96d291fcc0c4435aab2eae87867992a9 : SUCCESS in 48m 58s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a1ca3f931871409bbae9323a2c5be5c3 : SUCCESS in 1h 17m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f3bc3a89f4e64d95b537432d44750f0c : SUCCESS in 17m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ffc9a534ee24a4dbeeeb5ae9898a0e9 : SUCCESS in 15m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fffa40be6ed94e22800a808e69f5407d : SUCCESS in 1h 51m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/53630a8a717a49a0b99321ace6857d21 : SUCCESS in 21m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7a8ad8f591a4755b42812d8a792e277 : SUCCESS in 14m 50s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/76507f29baec49a5a6a12db7dfb775ce : SUCCESS in 47m 45s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/61eb8c007cf948c49fcdab4343913d52 : SUCCESS in 27m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/763117d22dc44ef181e51ba2adae6cbc : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70bddfdc72b74e828b9974ef299033be : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6fbcc0f63db4a5f1a7f98520869c4cd2b974d560","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-30 07:08:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a492a3a844964007314a5817ce0a9a8129227dd5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-30 10:39:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d92e8bceb20a80313c290a4e7cd6a84e986b767","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 10:40:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"996cd1dc033f8c6c1d5af3695da4e8e7d0bc8be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 12:18:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d39edddba06d43c7b221dd69362accec : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ef81d51d3b9472f8c4670f82a9f7c16 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22791044b97b4a6bb1f32f3a851709f2 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d3c55a80a64b5ca216c5c37de563f2 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/357cbbe47c4c4a148328e0ce8c7de3e1 : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c988830947d4386898583b32f2111a3 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eaab8ab328694711b5c6cae3c11956c2 : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d82ff8547297419aadca312d95acad59 : SUCCESS in 19m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/74a724c5765648ac94d5f44ab938039c : SUCCESS in 1h 09m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/877a73d6d65a4785ba5f05a095460cc8 : SUCCESS in 1h 11m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5a49da4f3ee2422f976147a4a764b1ab : SUCCESS in 1h 18m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2730514709b94e12bf0fda7f8991057f : SUCCESS in 20m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b002538c02ec47da8afec78a5737d8da : SUCCESS in 35m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6049dc4ea50c4ff6b1bba54d48e6b5f6 : FAILURE in 1h 30m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e84ea497d20b4437bdd62d0c22b29284 : SUCCESS in 42m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3ad149d829c425bb5cea878bcd1c6c3 : SUCCESS in 36m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bb1f049ff8f941159caf7c38a3c7f3f8 : SUCCESS in 27m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f3eae8095aa14a2dbbc1c6d13c6ea97f : SUCCESS in 15m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e62aa81fac54cd8a9604fcfff16e1aa : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1ffa4267d5baa7dd4ad8cb0c10c0e8cbef8a9fde","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-30 21:46:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6988aafd67b6178a4ae6f375825575ac017fcf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 23:07:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa15ecd6b4f24e41a10dffe976db704e : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4a5d3f2ef534a9fa33ba4a21e3eff79 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49e2bedf29ed429db515ab73c22620da : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc45e5e957b64ce7b694add415299eab : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0eb317f912da49729df13d7e489f82e7 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86104a0161994dc6afbc13336cba00a4 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48726f470784425d99a4b035d47296d8 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8008328b87f0464eb775ed4b60251ee1 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/504b0288a7c349ecb41b4212fe48d767 : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/752a0c0b600944648af0ffcd529b7a97 : SUCCESS in 16m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dada474e035d421493559ed15fead633 : SUCCESS in 1h 11m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e126192f26004f83b55ae683d7b2de7e : SUCCESS in 52m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b428627b953848619a838f8ff0a889fa : SUCCESS in 1h 13m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/edad270a420d49c6828d298b0956a3ac : SUCCESS in 20m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af4c574479614892aa96ead265a3a09c : SUCCESS in 14m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8c741a395e15409eb35a0cadd27b3ec9 : FAILURE in 1h 17m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/53520bfc6e7d4471b7d0c4d547fb8a0b : SUCCESS in 19m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c3c835a23c141568d674647f4c4d4e2 : SUCCESS in 14m 16s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/3a72c6925d3e429dadc89e37b75a3248 : SUCCESS in 46m 40s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/67bb079e800d469b8a6bdbc20c09bbb1 : SUCCESS in 25m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/693550ae2f3c40dabc32b4bab27deffd : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eb7aa79523b40c1b0f52e0f25f0f9ed : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f50447ca3cac7d1b8ce4d7c89a9ee33725312804","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-31 09:16:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fda5f2832ce6e49598f7f43b28dc5656f5d58a8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-31 09:16:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"114ed5606ef9cb0a45f34b61376a2f5ee0048cbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 10:56:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5ad13d725514bd3a83ab6d070338836 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca3c4bc571a481291cf963694a0e63d : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fcd46796e9c4a6c908cd3a225c5e768 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b1538523ee144dead62112610449d6b : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/142b78b038a145fe8e0df29302fc0d93 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a7ab4311f94041a6fae932698d50f0 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7779cae8da554b11a3cad14068c883fa : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c18f3f48f4464f588e108a7d92922b30 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48180557299649e8a731dc5c39c17ce8 : SUCCESS in 9m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a59f1ad45eec46f8bba21ecc27f232cc : SUCCESS in 19m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b1775741d7ef462f9edf5d68d6d88739 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbd182088984458bbf91a22b0bb343e1 : SUCCESS in 58m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/099b3c44ad4f4e2ea9345c5b641169f8 : SUCCESS in 1h 11m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f2298bd847764eedb1a3e71e078ecdfc : SUCCESS in 23m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e3ea3d21cce4a6fa6a36e17ab10f393 : SUCCESS in 21m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c1b61ae25917446d83de1f5da837c817 : SUCCESS in 1h 37m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5d7bd6d02e334dbe9dfabe26c77b33b6 : SUCCESS in 21m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee70b39c510c4e188d576e0487418d96 : SUCCESS in 15m 23s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/3ba48c6586ca4c47a6a02044f99cf76d : SUCCESS in 52m 24s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/05b515bfc49c497fb151ec894185fc72 : SUCCESS in 29m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a8f0bf09edf44dac95578d6f45cb0671 : SUCCESS in 17m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90ba699b6f0a4d38b02800c12d56a0f6 : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8f60b9a1653ec9c811c0d21d87ebb93f3044302d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 10:57:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75d559b2a82e8684416d1e37db52a26963b2c86f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 12:33:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdc5978b742e454490a50176994822b9 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7896db9f4a6a419ca0b79c15313baf1d : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbc3014935b74b0f8dd6606c492872a2 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a104cf09efe44c208875ea5b74b0e295 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83165af2a5a0439bb44bca8bc438e106 : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0dbca4f4ac34fbdb43fa4690a87d58b : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50afd38497764222986fc0bb973e0435 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41c2a9aa89df4915a2da9b556799106e : SUCCESS in 18m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/72cd4367167b4a469aeec7d3644e73f4 : SUCCESS in 1h 09m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de7b617908724d2b8684fad153dea05c : SUCCESS in 59m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8bb4303a2c974c5595f2cd407be4f229 : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b2a291db6b34428eb7c6eed1e12c4d5c : SUCCESS in 22m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ba6bb654c784830b5b38b340e2d7d57 : SUCCESS in 12m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4a36bf75f8004c50806835a4f413427c : FAILURE in 1h 31m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c414fa598f4e4dd6b1f86b6fd4df510e : SUCCESS in 21m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8087d596e4ca4dda8c5c7895ab451c76 : SUCCESS in 13m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/19f54e65daa54d908911d1b2adc44154 : SUCCESS in 27m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a1d25a55fd324eb18bbfcf5e394676ee : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0f116df50b84fa5891f4838d966a054 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"85be64af4e156a5203029467c0323960a73d0218","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-31 12:41:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a141e916f6b0d7a38d8d116c9a9748a6a707c3a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-31 15:55:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9fe78927b8f774cf22f13a6745c882297fc25106","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-31 16:08:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e20e5ce879d4cdb61ba1e2b38afc7557c1866216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-31 16:34:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2dead780f31d63a9501a38c7edb8c3a5ae7a3c8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 18:25:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bcbfe809ff84dbe95751255e08530a8 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89bc402d221d4ac28edb44f920a1f23e : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ef113222bc94e6686885752f0e1c02a : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58eda5f9494d48e1abbe1f487ded9aef : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a89a801d10d471facbe32ffb52dca4d : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85abc1ed5aca4129be5d071f26e5aaff : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d76845f8995142258d82d26097547d29 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3efd1d90ea6f41c89cfaaee7afc31356 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/feca279d6a1641719b162e5728cc885f : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5d2956d47024097bdbaadafe15fff43 : SUCCESS in 16m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7c3247f5781f46929803a6774c21b8ed : SUCCESS in 1h 13m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18d45f433b7347b0af912339acb2111e : SUCCESS in 48m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ad6e57b187344df7a8a963070cbf1386 : SUCCESS in 1h 16m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b4811a26a900435e922a96858b056399 : SUCCESS in 20m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d11db35dbcc745aa90245b1f4fea5d51 : SUCCESS in 16m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bcacc9fdd83f42e88addf472a6e7b7e6 : SUCCESS in 1h 48m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5ef1f5d4cbe64a9c8e956c0d4dfa20a5 : SUCCESS in 19m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/658176b6405b4998a733ad88ff7c6f77 : SUCCESS in 15m 26s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1bc9f5d5d27444f0b71dc883dac1a59b : SUCCESS in 54m 07s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3c1e43a7b2904a37822e58bbd89931e0 : SUCCESS in 28m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ed06192f0c8d40cb8fd475e44188ece6 : SUCCESS in 30m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffc3269f6e654d998bff05c4fd1095d9 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b93bd912174dbb5475e17e987449fcc0b3609548","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-31 19:56:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28dfaf7022ad8c505fb21247152e6e680b6c2e3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 19:58:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5f3ca2e89fa4b71dd3b7ca89941ae64d2a5a2b2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 21:46:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080e240cf1d2415ca22c86fa016ed77a : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f16bca8c3ff349eaa250ffc733f81311 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b60770ef25854d4995366f87babbbdeb : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c395ad0bd4b47e68763270beb5b17a9 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/520bcec2d2d44e6d8739537bb9aefd24 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2335081c52324c0a862f3c995030e102 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d0bc5ea107d447c9b6f28ddabd32ea8 : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c01fd595f4c347f0bb034f863c417efa : SUCCESS in 20m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fddca8161e6843e4a9a2b39e9ef89c40 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a4dbae0bf1b42af981eab3f97be7725 : SUCCESS in 55m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dac2ef2b56cc44bf920b399bc43aea63 : SUCCESS in 1h 13m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ef5adcdede354ba8abc79a541e813ff5 : SUCCESS in 23m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce9062654df84dc89e1ad876c1df55d4 : SUCCESS in 15m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3c70e3fc905b44299ae515104946d019 : SUCCESS in 1h 47m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0c4dfea427564753b6095d254f2e33be : SUCCESS in 19m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b09df7cea5a4001a264f7252d81d5bf : SUCCESS in 19m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2ef06012ee354af192b0e2866e9b52e8 : SUCCESS in 25m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dc733b94d8fc4d85807a7b72dea082f7 : SUCCESS in 16m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4b971360648496bbc3c9fd1263ee60a : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":2},{"id":"5979070d1c94467b9d11ba9bc37e53bb3941fafd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 21:46:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4bcdf72c4beb4963195361b2f100fa031fa413f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 21:48:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb67589c864d4568a28eb90eba989ce0 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e99d12de668413d803e4db252cc58e5 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"313a98512ab387e3857760fa0a73afb09f598061","revisions":{"3d14abe207e1e53b6491af374aa91e70591686cb":{"kind":"REWORK","_number":1,"created":"2022-05-27 17:52:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/65/843665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/843665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/843665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/843665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/843665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/843665/1"}}},"commit":{"parents":[{"commit":"c231848e3887e35ef897e81f8fb9370f9ecd42a7","subject":"CI: separate image builds into a non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c231848e3887e35ef897e81f8fb9370f9ecd42a7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-29 16:09:47.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-27 17:52:20.000000000","tz":180},"subject":"ironic: Set MTU on provisioning and cleaning Neutron networks","message":"ironic: Set MTU on provisioning and cleaning Neutron networks\n\nPreviously we were not applying an MTU defined in Kayobe networks.yml to\nthe provisioning and cleaning networks in Neutron. This could lead to\nissues when nodes communicate with the Ironic and Inspector APIs.\n\nChange-Id: Id9418e4e88c52056412daa22462aa611bfcc59ae\n(cherry picked from commit cd24b6d98e1b82ab9c255e3e5ddbaed99c49067f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d14abe207e1e53b6491af374aa91e70591686cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d14abe207e1e53b6491af374aa91e70591686cb"}]},"branch":"refs/heads/stable/wallaby"},"313a98512ab387e3857760fa0a73afb09f598061":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-31 16:34:08.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/65/843665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/843665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/843665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/843665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/843665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/843665/2"}}},"commit":{"parents":[{"commit":"4d68a5dd5e94fe6ebc879485bf379e3a80a53f0a","subject":"CI: Disable horizon in upgrade jobs to save disk space","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d68a5dd5e94fe6ebc879485bf379e3a80a53f0a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-29 16:09:47.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-31 16:34:08.000000000","tz":0},"subject":"ironic: Set MTU on provisioning and cleaning Neutron networks","message":"ironic: Set MTU on provisioning and cleaning Neutron networks\n\nPreviously we were not applying an MTU defined in Kayobe networks.yml to\nthe provisioning and cleaning networks in Neutron. This could lead to\nissues when nodes communicate with the Ironic and Inspector APIs.\n\nChange-Id: Id9418e4e88c52056412daa22462aa611bfcc59ae\n(cherry picked from commit cd24b6d98e1b82ab9c255e3e5ddbaed99c49067f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/313a98512ab387e3857760fa0a73afb09f598061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/313a98512ab387e3857760fa0a73afb09f598061"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
