)]}'
{"id":"openstack%2Fkayobe~844046","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Ia57a82ced121777e06af228d54fd14a892f0ecf0","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-05-31 18:04:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia57a82ced121777e06af228d54fd14a892f0ecf0","subject":"CI: Disable horizon in upgrade jobs to save disk space","status":"MERGED","created":"2022-05-31 12:41:00.000000000","updated":"2022-05-31 18:06:15.000000000","submitted":"2022-05-31 18:04:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844046","meta_rev_id":"078abb032a25f8cc063cbb3fa2a3b7603a8150e0","_number":844046,"virtual_id_number":844046,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-31 18:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2022-05-31 16:34:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-05-31 16:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-31 16:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-31 16:09:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-31 14:35:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 16:09:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-05-31 16:34:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"fde66886d69f1eff5b517c0779f7c4cbe7b7c5e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-31 12:41:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d46b82346f8c339cf9c6e92bcde6d6c37e75a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 14:35:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0919524d42034a368e75df3a66539209 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6881678809ce4674aaeb6a9a500e0570 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90d348a217e44ff99c637d562b11220 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6590baea210c4a52a1d4e4ee438dd434 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b7c3248896c4393afe1b06e7a545d6c : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1cc4d1dec6446a595d65bd2eb943054 : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d9c7a4d47fe4cb98ba4c18d906ce877 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e281b747fa7440cbb696ebff68d09122 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9cc83fb04edd4fb7ab48e0d1da19af53 : SUCCESS in 19m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f40fd29a62a44fd2ac4c2ee9ce8bc6a9 : SUCCESS in 1h 41m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba9c717174134fde94bd0c84d5017849 : SUCCESS in 57m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0dc7341ba55b400a8fc1a2d5cc259381 : SUCCESS in 1h 43m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3a832683ab23470eb946b3ba137d999e : SUCCESS in 22m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd2be3737ff84cebae9f37e09adf3e8b : SUCCESS in 16m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0400c925e03d42e88769ce9df3b4ff5f : SUCCESS in 1h 51m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d8e9eca4a6b14ec992482c8adc067ff5 : SUCCESS in 18m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/588f56d5e4e44453a93890114c51a575 : SUCCESS in 15m 52s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/2586efd4e3fe4a5db7d249d674a9d8bb : SUCCESS in 51m 10s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0f5b56b93aa644f49829a18cde576c9c : SUCCESS in 27m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4d4f1bd7951d46ae886b5919c1e562ee : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/201d1714ac4242be99dea4746f6c9469 : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fc420695ebcbc4c0e20a5089b0a3e1714c8ae050","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-31 16:09:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bfd9109f8b638822f1fa3aa33a8ad85ed8d712f4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-31 16:09:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e88c5be9828c73a55be4db5052ca82d58c4bc50d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 16:10:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2384f0e79c22011a0abdbc9b236522e903d5191","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-31 16:34:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e7bb0f6ee833aad3440b3efb20b98941ae42bb78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 18:04:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af162cabde784631854c62029dbafa6d : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04349d6f4f204687a0cfac3ddc378b04 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1605ed246e048a8a57670aa4c5efb50 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6240879536524ed6bd7a6e405ca71a1f : SUCCESS in 8m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/39579da1ba80466197fc3c39dadb3e1a : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8934fa5b8e5f43e882ef51a75d5884d6 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d050cec28c274d1a9a671b5a7263767b : SUCCESS in 18m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7967640551f04899bab0e6d33ef11ebe : SUCCESS in 1h 11m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c694b386c6543c6951e5101ee72e923 : SUCCESS in 58m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f9063bf869de49c38c268d94515c03de : SUCCESS in 1h 14m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/35014b79e747408eb9e7a066b5228581 : SUCCESS in 17m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/245651d2846d40d8979ca6038715ba64 : SUCCESS in 15m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e946570d663c404d8cae819864187345 : SUCCESS in 1h 53m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/381768c7eb974cfeb668270d7c3aa9b0 : SUCCESS in 19m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc590eeb3ad044c4af053ba682af75ee : SUCCESS in 14m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2189909e83f64ff790c77f6a49c2cf7b : SUCCESS in 25m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/02d139e1e19e4a4d9754925df7d4aa72 : SUCCESS in 12m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/371045bf9e174547b5e670bc025c0d22 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"dc7b963a32cd603fdb87e7a7a44e8df960adb81d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 18:04:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"078abb032a25f8cc063cbb3fa2a3b7603a8150e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 18:06:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db08485df6b14df3842d66738d9bf1c5 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d68a5dd5e94fe6ebc879485bf379e3a80a53f0a","revisions":{"4d68a5dd5e94fe6ebc879485bf379e3a80a53f0a":{"kind":"REWORK","_number":1,"created":"2022-05-31 12:41:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/844046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/844046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/844046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/844046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/844046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/844046/1"}}},"commit":{"parents":[{"commit":"8888e406510f1e22c1062119346a5a9eae654c32","subject":"Fix forgotten hacluster regexp for image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8888e406510f1e22c1062119346a5a9eae654c32"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-05-31 12:40:25.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-05-31 12:40:25.000000000","tz":120},"subject":"CI: Disable horizon in upgrade jobs to save disk space","message":"CI: Disable horizon in upgrade jobs to save disk space\n\nWe see frequent failures of kayobe-overcloud-upgrade-centos8s due to\nlack of disk space. Disable horizon in an attempt to free up extra\nspace, since we do not need it for our testing.\n\nChange-Id: Ia57a82ced121777e06af228d54fd14a892f0ecf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d68a5dd5e94fe6ebc879485bf379e3a80a53f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d68a5dd5e94fe6ebc879485bf379e3a80a53f0a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
