)]}'
{"id":"openstack%2Fkayobe~844387","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I16d721539e5956f0d86503dd5b01ea608b129873","project":"openstack/kayobe","branch":"stable/wallaby","topic":"bifrost-netmask-filter-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-06-06 11:04:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I16d721539e5956f0d86503dd5b01ea608b129873","subject":"Use net_mask filter instead of ansible\u0027s ipaddr","status":"MERGED","created":"2022-06-02 13:12:09.000000000","updated":"2022-06-06 11:05:36.000000000","submitted":"2022-06-06 11:04:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844387-bifrost-netmask-filter-stable/wallaby","cherry_pick_of_change":813265,"cherry_pick_of_patch_set":1,"meta_rev_id":"c252e76c104661e2fbe8d6c7c7470a2cb5eb541b","_number":844387,"virtual_id_number":844387,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-06 11:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2022-06-02 14:11:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-06-06 09:09:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-03 12:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-06-06 09:09:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 13:12:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2022-06-02 14:11:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-06-02 15:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-06 09:09:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"38f66cad2724d28d006e6811ea04557611e8f716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-02 13:12:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ba6d29a8178b5a22d6a24fc57e23ee653a3c00c7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-02 14:11:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"74d7b27611bfa919fb589825abfc35478f5f82f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 15:57:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/522c4ad3870c48fb97db61fdad0d73cb : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6eacd1220a6414ebe19e2ed0f5a374a : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df9b6d75efb145b9b03f5861b3cfe65f : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e59ead311c5545e8b85d075bdb89a5ee : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eccb4a3456844919a2580be9fb07a97 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17af50d0a494d3dad3c8328df867bcd : SUCCESS in 8m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b4822441fc74158a3bf77006cbad719 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68ae96e7d5474d16b5217da962d77005 : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e18a439db77e484b8742190d6503b1f3 : SUCCESS in 19m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bbcbc7879f474dcbb26bb1d7efb8dcde : TIMED_OUT in 2h 03m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94ebdf7526f1467bba353e7054213bb5 : TIMED_OUT in 2h 04m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d293ab568b6f44318b8580a91be671a0 : SUCCESS in 1h 14m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/311af12298304dfdb3140db846ffd42f : SUCCESS in 22m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74fd86e3cdf44afab15c2b7d1441403b : SUCCESS in 14m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/37b5aa11abc9476a810ba198f8cd1f94 : TIMED_OUT in 2h 34m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f69b0a9d2eb748e08dc7f4c16577aa69 : SUCCESS in 43m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccd636a78c7c4b39a0d0365ddec00660 : SUCCESS in 14m 45s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/7dc857795051496aa1e997deb8355971 : FAILURE in 53m 55s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3087f914a0fa4d9f963abdec2b7d25e7 : SUCCESS in 1h 06m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d42d6ff2c4c54b22b78d84702a8a710d : SUCCESS in 17m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a76aaa0813b140debeb837d2f6ccfe33 : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0f77d5bf11158b726f2a1d77e16189fa37ee190d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-03 07:39:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c3c0a4ad3be0b329275f344dd84ad5f8680b8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 09:06:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c175fa7f9eb40fdb001d55824d755cb : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f7de08b7e24d9394feb3f45f268ef0 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/071fd84833db4ba28f812715c7c026cb : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06c4bbd8a5234aa3bab4ee59b4473271 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65609e476ec64a929b8b5f7406cd84f8 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d6c7e4d7fe434e84ab8e1f222fe1a8 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fd9cf503bbcb4c1db5d38689ba8ddd21 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/83a7a917f1444dcbaae9755b76869cbd : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2dc39cca154c43528b14068f262e6faf : SUCCESS in 15m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d2854854d8644a51aea90d0bf77cc8e2 : SUCCESS in 1h 10m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b8e2538560f4362afbdee19175dd82a : SUCCESS in 53m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5e55fc80fbd8410fa7856c4a3438811f : SUCCESS in 1h 08m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7ecfb5c32fa342eeb610807f608164fe : SUCCESS in 19m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc19509fab7f42a09b78ea7b66f713d4 : SUCCESS in 12m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/32ea91fbe02c41a996abe127d44cf2cb : FAILURE in 1h 23m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d50497ba7f08465d96579d343d593856 : SUCCESS in 20m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d5bd9cfc58144518b53b300a4af2c67 : SUCCESS in 15m 00s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/f62ecd7ce22440c28e735065d4cdd04a : SUCCESS in 44m 08s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/666276eb6ca34804b458354554366a49 : SUCCESS in 26m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ac940a6e95b04bf8b4650c3620c42b6e : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a4997eff10b4333ba947bf43c48534e : SUCCESS in 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"871783bce92e7f1eb8e5ab798b5ab73848e7d1c9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-03 09:17:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31cc798e85d9bac905c833ac5d22ecca1b276a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 10:38:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef2ad791e6354063a077705b06090a01 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8df555e298e4edeaabc7dd182105b0f : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c0249a35ab14648ba5eaff9322dd249 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/010ef29aafba46ad94f772aa38dd2997 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/287d22b337864c40912b8523c213360a : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10380aff19f41ffbf9d52b9a7684d03 : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b57c496077dc4760978c15e48d560798 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a86ddbc9d98430d9671d4600ae84931 : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df49d39162f64d55859419ac8ecef62d : SUCCESS in 17m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/23ab1bf362174b20939cf399da31e624 : SUCCESS in 1h 15m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1713c0cedfb496a954297a2579f66d0 : SUCCESS in 59m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1e211e972e70477f8d1be8590f50b5ac : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/df8f60d9cd7b488089dcd43bca005593 : SUCCESS in 21m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bb7c26c1ad2477e9b53579026bf9a95 : SUCCESS in 15m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1c12f39773634d4cb037802ba03a285a : FAILURE in 1h 16m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/211d6a5e17b54b7da084f8baddaf9876 : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7235788290348aea4e3d700854cae04 : SUCCESS in 13m 38s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/eeb3986b254c478c88dff2f12db84da3 : SUCCESS in 50m 18s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fa560681705f4721a7b26e20eae9df04 : SUCCESS in 23m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c949d57c625f447587e90e16327b6b50 : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06a321f2dfbd4416bf26f9cfaf51f7d5 : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3b4fd6c3b8bbe0bca519f64e55f5ab6382699270","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-03 10:52:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d9e30cfad04f9227a0993997701ab9dfd86c368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 12:23:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c73636901fb4ff5966c33f01b3637c7 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d33d56af544e609327b32a093d786a : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39a8747b55d6466cb66c005ce649ffb6 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e301877c0974bea85ccdbea5185931e : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a44aea02edb48f686ca2b965e9d3ca3 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e597f78c4324f00b084cf583dc8accb : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/66d8f4c8554540f7b33952b92c575ee5 : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e56db70d420f4e3282baee7d187d616a : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be20db3952d84808a9ea87838ea9c009 : SUCCESS in 17m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fbf8f56bd80b4161a4dd3ae3940c41e5 : SUCCESS in 1h 11m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7575973c34d4058aeaa81e81a2b45b2 : SUCCESS in 59m 30s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4712570e77d34f609e6e87199c118b94 : SUCCESS in 1h 04m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a2bbe6740672452b93f973b8e78833d7 : SUCCESS in 19m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bebb0608fe144f9b82399c0b834a947 : SUCCESS in 11m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/68a94e90c54d4c6c9a99fa8ea20dc852 : SUCCESS in 1h 27m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6c670fd0755a4f329a2df00211c6afc5 : SUCCESS in 20m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f7440b5861e4eb29a39a8e7d40349fd : SUCCESS in 12m 49s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/37f415ef56734cbb830df0579a858925 : SUCCESS in 46m 01s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/175d85cffbe244148e8a56580a51c74f : SUCCESS in 26m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f701852818ab447185f372e1e26a2137 : SUCCESS in 14m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0137fc3977af4f79910e7b242d23fb22 : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ce468030cf00cdfec163b2674d58db4664ba75ef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-03 12:25:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d6a633ba1193014b9c72ee180cbc40613d57ed6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-06 09:09:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4e662f417d23e86255c7e241490ebc02e57f93c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 09:09:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e67f7126f6d2026d034a5ef95e00f1048d599007","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 11:04:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/441b7e6c29104e00b26605b6eb37a4ef : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ee3a682d115402ea40acd31db6be369 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/878de7eb8b89466b956e6f170b2291fd : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead8466ce2a546c58ae5279cc128b559 : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b213da014385406fada9170a39bead92 : SUCCESS in 6m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f63329323c53427084bbd440959e4ede : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81234a4eb8bf4b7dbd36e88c5cff6d9e : SUCCESS in 18m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/874483f0d9884e0ba23789374f8e284b : SUCCESS in 1h 14m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0dfbf46e4444ea5aff36b58a2c1d5ba : SUCCESS in 57m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/295a9dcb038447e0bfb711927d31cb6e : SUCCESS in 1h 20m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f15c0b1923014dcdb69f438b11068232 : SUCCESS in 21m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24204e60124841ab98289180ce445d3d : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba27a7210a0c4507a8366db2421f8d04 : SUCCESS in 1h 52m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b4e575e4ce294dacbee45fe89a50a13b : SUCCESS in 21m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9972ee1a5d1941c584aef1a11db697dc : SUCCESS in 17m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/812628ce16ca4adc996eeab96de5962d : SUCCESS in 27m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fd03a256877e4033b930c6793e1e2d1d : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e22507195c144bb834a2c6fde0e6569 : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8474d6e649c78b81278e6f97a1a10d8417f57496","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 11:04:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c252e76c104661e2fbe8d6c7c7470a2cb5eb541b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 11:05:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8159eaabfdd84112b18e0c78833f28f5 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a093b59d23ea882ee6caa8e072abebe5a326cc8","revisions":{"9a093b59d23ea882ee6caa8e072abebe5a326cc8":{"kind":"REWORK","_number":1,"created":"2022-06-02 13:12:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/844387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/844387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/844387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/844387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/844387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/844387/1"}}},"commit":{"parents":[{"commit":"313a98512ab387e3857760fa0a73afb09f598061","subject":"ironic: Set MTU on provisioning and cleaning Neutron networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/313a98512ab387e3857760fa0a73afb09f598061"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 22:39:15.000000000","tz":180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-02 13:12:09.000000000","tz":0},"subject":"Use net_mask filter instead of ansible\u0027s ipaddr","message":"Use net_mask filter instead of ansible\u0027s ipaddr\n\nFor a long time [1] we have net_mask filter, so no need to use ugly\ncombination of the net_cidr filter together with ansible\u0027s ipaddr\nfilter.\n\n1. https://opendev.org/openstack/kayobe/commit/dc45036c92cb28fd4e9f2d9c84a03438bf7ad26f\n\nTrivialFix\n\nChange-Id: I16d721539e5956f0d86503dd5b01ea608b129873\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 0751a316f4ef093221ae3bab2d6514689cd980de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a093b59d23ea882ee6caa8e072abebe5a326cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a093b59d23ea882ee6caa8e072abebe5a326cc8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
