)]}'
{"id":"openstack%2Fkayobe~844795","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I0658c0059867468d6032cb1bfa3d05ae0d01c422","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2022-06-06 20:02:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0658c0059867468d6032cb1bfa3d05ae0d01c422","subject":"Add missing declaration of kolla_enable_hacluster","status":"MERGED","created":"2022-06-06 12:22:15.000000000","updated":"2022-06-06 20:03:30.000000000","submitted":"2022-06-06 20:02:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844795","cherry_pick_of_change":844591,"cherry_pick_of_patch_set":4,"meta_rev_id":"00686b987d27fd10edc26ad1c9a9a7358293c6f9","_number":844795,"virtual_id_number":844795,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-06 20:02:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-06-06 12:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-06 13:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-06 13:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-06 12:59:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-06-06 13:10:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-06-06 15:01:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d6422e9233f22231361c0aff918d9a353b5f6ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-06-06 12:22:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6f2243c97accac9f2260a7c74f6ebf971335b5b8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-06 12:59:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c840462ee2da831aa26189e51490248cff32c6c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-06 13:10:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dcfd1e109fec73d41c7224fa0cae65b2b969f7d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 15:01:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e94e6e6c57534b41824e98d6b0e46ca1 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abed3985a5bc468db3b013bf89a6068a : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d63cf6a6f3c44250b8ccd1a51f507077 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79d2dcf73e624c0598f975f7902d9cca : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abe530ecd7fa4b0eabd167f7c6be5a05 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2af350e12c64bafb4780b5f6228f271 : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10623a46148844308f5151e9b39b61f7 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15affba89bab489d9844acbd43f50024 : SUCCESS in 6m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1947fcf07b5f4923b115cf8119dffefc : SUCCESS in 14m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c972813639bf4760b9340fb983d89681 : SUCCESS in 18m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3b7544835b304a1b9d4a176c92423ce8 : SUCCESS in 2h 05m 21s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/6ef8faccfd1340fb819b975e2aed9ad8 : SUCCESS in 2h 02m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b8ee1a81f1b4db7bef776e5d1f6bf7a : SUCCESS in 1h 20m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/67f2b51c196d455e9dda71f2690385d1 : SUCCESS in 1h 55m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/74973bc09810473f9fb8ddd2523a423f : SUCCESS in 24m 08s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2b545c9a904b4e939d2becab866efc89 : SUCCESS in 24m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65f5ea6c579f471eb6178800340da94c : SUCCESS in 15m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/558fd15e1d2b4e1f981d75db9b672a2c : TIMED_OUT in 2h 33m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/110ecde0f2a54b18be2cdc1a9ee1fd95 : SUCCESS in 1h 07m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1ca6af1cc30e4f1d86e0677575cf9c40 : SUCCESS in 19m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/71b44766ed2a4408bbecef2109b7ca6d : SUCCESS in 20m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8efcbd7e20e845639d92574248c062a8 : SUCCESS in 14m 52s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/3a538888271349ddb33aa672e156e0be : SUCCESS in 44m 11s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/114c23c265564b3a9eca27271c2d77e4 : SUCCESS in 27m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f933bd9c09c541b0bc0969cd76a13ed4 : SUCCESS in 21m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6d832377b7cc4274951ceb9fbc94734a : SUCCESS in 12m 52s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6d3b6534d0f84403ba33ce5ee6c8dc24 : SUCCESS in 16m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49e620167a244441985825f7023371ac : SUCCESS in 13m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c61a762f0ab8478baf01d0e1255bf1a4 : SUCCESS in 19m 08s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/512c1ed695da43a9a0495f5595eddd3b : SUCCESS in 17m 27s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4834da2d29644723810e9b2c61f3737a : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"fca0710918869a468b7f7abc171e24ebe0eab023","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-06 15:43:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f3d29689696d6539ce5272783b59606ec45c934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 17:39:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94df91e9ffe94d859dab239f10e1e51b : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/526b1a98510c457fb03ff783293b0d3f : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d72d54290d048408b586343e7c2137c : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77adea06e555429ebb940bddd78326c5 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/899e0e38ac7c4aa7a6e63be2eb44f121 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e8f57b82b143edb649461c49842e20 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b96887e0b336461bb4877af2fc99eed0 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d10feaa088de4b98a1f3f2b78c5830db : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1bee36c7cc7f4fce80d9ed9a08e44762 : SUCCESS in 12m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/19ff25b8644540d5b2c1d0f29ebfbd11 : SUCCESS in 18m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7fb74292f90049d7bdc19d226f5e14d1 : SUCCESS in 1h 15m 56s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/cdcac2170110434aa53ebecd082c4cd5 : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd26c1518a1b473fa439891b8eda741f : SUCCESS in 57m 58s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9a23d4eb5bc94a608109727e8b705ed4 : SUCCESS in 1h 18m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d23ab964945d45cfaaacb4245cb04091 : SUCCESS in 24m 33s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/8bf571ca628443c18e923b40bb1f9a34 : SUCCESS in 23m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6d64c37523c426ca4cf0e8a163612be : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/400712311d374f93b129a02437cd3093 : SUCCESS in 1h 54m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c41fa8fdb2fe41d3a239bd0d3fcf5675 : SUCCESS in 1h 10m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9991da1fd68b4442aa7c21713629393d : SUCCESS in 19m 35s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/bdb5be3478e94488a76ccb18a768bb49 : SUCCESS in 18m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bde451ddb684753bd3baa01fe40c2db : SUCCESS in 15m 28s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/f6dcfad156bb4193aa403b770e2a6ba4 : SUCCESS in 44m 45s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9f4813919dd44433be9f630cf5a9d329 : SUCCESS in 28m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70c6e778b47c4623aa0edffdd9e8cf12 : SUCCESS in 25m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b6c6dc7b999e4f969aa7c91e4139de83 : SUCCESS in 16m 08s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/58c249a46c734a9a88afdf5fd5b3529c : SUCCESS in 14m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/405272d7722f4d3e8ffbdf370104c2bc : SUCCESS in 11m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/54177a4a0dd9496291e389533ff07599 : SUCCESS in 18m 25s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/93dfeaecc5804ca89b557e910e163d58 : SUCCESS in 17m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5979f1c75464fa8b88cfd9e9a8f596e : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cc4ac738c89db4814b07311a1ad67c5118cc9eea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 17:39:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b7da0c5f1211f099dc1eebc301b37be6abf89df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 20:02:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf82d19a32f44f01b638c3047d40f96a : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b65d5265a2343e2bb5185c68509b5ab : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1b88a250e084bf497f8cdf95efd0636 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65bbb78e608d4bc4af4f96e8e0b155bb : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19ea54157750454b926d2096d008993d : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76fdf6d2c4cd4e7f901453967bc46971 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a582aa6c09d84888a1e0b20c22a8ee83 : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a9efd1fbdb449088fc8630237381bb8 : SUCCESS in 16m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8b04ed4209b94184a22fbbda7ee82a44 : SUCCESS in 1h 08m 54s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f10a4ad2b66d4df1a8532308607b8128 : SUCCESS in 1h 16m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7089c87e5e2a4cb0a194469e73d59164 : SUCCESS in 51m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8d035f931a8e4d2c8a6dc5d0b820a797 : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3e6e21c7000f4f60bb25b1a6fb057a42 : SUCCESS in 26m 08s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3fdffd04ad494990bf911a89427a6c6d : SUCCESS in 24m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dff33b8c567445659e318723fbc27a57 : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2ab3085b620e48b99f88c964b5d9c3e5 : SUCCESS in 2h 18m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc6163dc4c15401e93d671a3eab11f35 : SUCCESS in 1h 01m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/90fb1035f1eb4b4382b85f03826fa5b7 : SUCCESS in 21m 09s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e0f93b7a023d474bb400402a1ca7f537 : SUCCESS in 20m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61b2bbd38a7a4608be58aaffe03d71ac : SUCCESS in 11m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/163b4c4b680d47268da1f3806aed7b51 : SUCCESS in 23m 04s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25e21d001e1b41a0954022370a8cadca : SUCCESS in 22m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bf71fff674144c4399cf5791de84a335 : SUCCESS in 14m 01s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9cbf46e569d44a66a22402b0a43d5d1f : SUCCESS in 15m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84b9da3c054a43ff89f6c42fed241c58 : SUCCESS in 9m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/96e7f90e83504901a55d77fbfc514cd9 : SUCCESS in 14m 56s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a1a8d5c4b5744aeb98fec9eda2e94b5b : SUCCESS in 18m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dd97ff3cfce4596af32da63861266fd : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fc1975266a2d1fa69c4c1306ee4bf22159b42f5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 20:02:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"00686b987d27fd10edc26ad1c9a9a7358293c6f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 20:03:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d0076704464491cb9d19a38f5a7001a : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/617be158ae8a4574ad87e536398d4fc7 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"582f05020b382cab693d4eff58ecfdeeeb104397","revisions":{"582f05020b382cab693d4eff58ecfdeeeb104397":{"kind":"REWORK","_number":1,"created":"2022-06-06 12:22:15.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/95/844795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/844795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/844795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/844795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/844795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/844795/1"}}},"commit":{"parents":[{"commit":"8c9d8af727ce192ee37f86b1f2af4002d95a1d5e","subject":"Merge \"ironic: Set MTU on provisioning and cleaning Neutron networks\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c9d8af727ce192ee37f86b1f2af4002d95a1d5e"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-06-03 13:28:05.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-06-06 12:22:15.000000000","tz":0},"subject":"Add missing declaration of kolla_enable_hacluster","message":"Add missing declaration of kolla_enable_hacluster\n\nStory: 2010069\nTask: 45540\nChange-Id: I0658c0059867468d6032cb1bfa3d05ae0d01c422\n(cherry picked from commit b77033970a79587b3be5c04a5b81e8b88e21dadf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/582f05020b382cab693d4eff58ecfdeeeb104397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/582f05020b382cab693d4eff58ecfdeeeb104397"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
