)]}'
{"id":"openstack%2Fkayobe~845278","triplet_id":"openstack%2Fkayobe~master~I28f2b43ddb8bad6097e690e766f0348b1b8a296b","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"33307":{"account":{"_account_id":33307,"name":"Kyle Dean","email":"k.s-dean@outlook.com","username":"k-s-dean"},"last_update":"2023-02-21 17:44:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28f2b43ddb8bad6097e690e766f0348b1b8a296b","subject":"docs: add information about enabling TLS to automated guide","status":"MERGED","created":"2022-06-09 14:32:29.000000000","updated":"2023-02-21 17:46:59.000000000","submitted":"2023-02-21 17:44:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845278","meta_rev_id":"13993272763ff7397de88cf446bdec86970499fd","_number":845278,"virtual_id_number":845278,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 17:44:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-02-21 16:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-02-21 16:49:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33307,"name":"Kyle Dean","email":"k.s-dean@outlook.com","username":"k-s-dean"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33307,"name":"Kyle Dean","email":"k.s-dean@outlook.com","username":"k-s-dean"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-09 14:47:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-16 09:09:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":33307,"name":"Kyle Dean","email":"k.s-dean@outlook.com","username":"k-s-dean"},"state":"REVIEWER"}],"messages":[{"id":"8554df16682890fd8f634deae8f8b188efed0b30","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-09 14:32:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4be4e65f17f62ecf42c10e4e6bcee7e632c12c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 14:47:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c82666245abe438f9727a5bd5eceb107 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37da68ba514f4249884320b682c7e40f : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8e4081d61fe769874333be37213b299a9ec376b8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-10 09:15:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10d9f4180d40280adf93a1d7b9d0ae7560acf64b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 09:25:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017075239bb340ca83ae6e1e5154b8ad : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8027a690f1c4c399ffa7f97bf5be37d : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":2},{"id":"678fe7bf2eb782e0f01e59fd04ee68367260f336","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-16 09:09:01.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"76fbfdd6c784c095369b903cfb75b23ddd86d27e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 16:49:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"129637fc1487cdd8f2be43912d471247c62dd2b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:50:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00a890bd0f3a2fa9d41a4c0be5b23625309737f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:44:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a6b6324d67e45db8f936512669fcf56\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5a3d54d106d4b52807cc2925abfbfa4 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15628cdfd8da42339860613f59ea6edf : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a71b7c5f2323350a3d256e61506d8ff08144c322","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:44:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"13993272763ff7397de88cf446bdec86970499fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:46:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd4ba55cd3f149e886d21cdbcc7143c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff49acd510c94442b0023b2330b047ea : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"804c1cb669756abd512477263269def90ba3e077","revisions":{"e495b3dae3203495062f0c5dcf01be8abe2abe43":{"kind":"REWORK","_number":1,"created":"2022-06-09 14:32:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/845278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/845278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/845278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/845278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/845278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/845278/1"}}},"commit":{"parents":[{"commit":"3e7b01ad829e0c14396045cf147e719ad0a89ed2","subject":"Fix forgotten hacluster regexp for image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e7b01ad829e0c14396045cf147e719ad0a89ed2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-09 14:31:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-09 14:31:47.000000000","tz":60},"subject":"docs: add information about enabling TLS to automated guide","message":"docs: add information about enabling TLS to automated guide\n\nIt only covers the internal API VIP, since we don\u0027t have a separate\nexternal API in the development environment.\n\nChange-Id: I28f2b43ddb8bad6097e690e766f0348b1b8a296b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e495b3dae3203495062f0c5dcf01be8abe2abe43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e495b3dae3203495062f0c5dcf01be8abe2abe43"}]},"branch":"refs/heads/master"},"804c1cb669756abd512477263269def90ba3e077":{"kind":"REWORK","_number":2,"created":"2022-06-10 09:15:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/845278/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/845278/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/845278/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/845278/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/845278/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/845278/2"}}},"commit":{"parents":[{"commit":"3e7b01ad829e0c14396045cf147e719ad0a89ed2","subject":"Fix forgotten hacluster regexp for image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e7b01ad829e0c14396045cf147e719ad0a89ed2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-09 14:31:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-10 09:15:51.000000000","tz":60},"subject":"docs: add information about enabling TLS to automated guide","message":"docs: add information about enabling TLS to automated guide\n\nIt only covers the internal API VIP, since we don\u0027t have a separate\nexternal API in the development environment.\n\nChange-Id: I28f2b43ddb8bad6097e690e766f0348b1b8a296b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/804c1cb669756abd512477263269def90ba3e077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/804c1cb669756abd512477263269def90ba3e077"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
