)]}'
{"id":"openstack%2Fkayobe~845338","triplet_id":"openstack%2Fkayobe~master~I0d7ab0d8ff5b16ac5de8e50e63400bd455996555","project":"openstack/kayobe","branch":"master","topic":"user-provided-repos-yaml","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-20 08:53:11.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-04-20 08:53:11.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-20 08:53:11.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-04-20 08:53:11.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-06-14 14:52:26.000000000","reason":"Reviewer/Cc was removed"}},"hashtags":[],"change_id":"I0d7ab0d8ff5b16ac5de8e50e63400bd455996555","subject":"Allow to use own repos.y[a]ml file to build containers","status":"MERGED","created":"2022-06-10 09:56:00.000000000","updated":"2023-04-20 08:54:15.000000000","submitted":"2023-04-20 08:53:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845338-user-provided-repos-yaml","meta_rev_id":"8085d11e09aa2db4e81f08974aac0f8414eb32b0","_number":845338,"virtual_id_number":845338,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 08:53:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2023-04-20 07:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-04-19 16:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2023-04-20 07:02:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2023-04-20 07:02:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-10 10:16:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-06-10 10:16:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-06-10 11:45:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-10 11:56:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-06-13 08:46:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-06-14 14:52:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"},{"updated":"2022-06-25 19:16:57.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2022-11-03 07:09:59.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2023-02-15 14:33:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"4a4582595e77e4a5a0e2c65029448b233ba69681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-10 09:56:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba3a22dfd694e07f9ff3be12236f0be1ef5211a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 11:45:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/907dd5e4a72840df96027fb694326f3c : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2efd5767039d46cb9e7ed49d842f01f9 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e60c20f292d430b9b72a05f03d936de : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8bd043be8b34267b112388018845e46 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7abed6e0cd214243a9c32962f3eb3cdc : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd93777c5b34705b905200a9b191702 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43ced19b79bc432a9c4915614766fe72 : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6256638537034de394faab3027fd8904 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0583a105e8a3482298ee16febe1e1705 : SUCCESS in 14m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ba711b27f314ee1ad6f5dbffca356c5 : SUCCESS in 18m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/57cffc79cbd1491395f0a60c69ff7636 : SUCCESS in 1h 12m 10s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/39111cfb1b674353be894e3cd12d938c : SUCCESS in 1h 02m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7fdb0d033fc4aa18bbb2cb94876a472 : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6ed9bcf644b24f9fa32fe79eff1166dd : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0143a66f99fc456ab027e977c95d920f : SUCCESS in 23m 45s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/03f6bc1ecdc04e78ab64a0534235c718 : SUCCESS in 19m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d92ec1e537de485aa01ee396e755131d : SUCCESS in 21m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/24be61e6bcee461ab541a0082353abe4 : SUCCESS in 1h 47m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa354e33c3b44ed983d6f126cbe3937c : SUCCESS in 1h 11m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/92c836cdd42a4bf18d0d708a2c07e43d : SUCCESS in 21m 58s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c31b3e8d27964e819b1b6f2362c1dd8e : SUCCESS in 18m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79a08e57bd3d4c768f4ff806e8de4708 : SUCCESS in 16m 52s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/665f721d8d984ba7908968d3564a0d48 : FAILURE in 30m 13s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2a53aef013ab482a82ceb303c5fc438b : SUCCESS in 48m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62a6b6b59f14407f94ca158967037cf0 : SUCCESS in 32m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/210e129ce7c64c00934ae5ae179759aa : SUCCESS in 19m 04s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1bde6a4b023a46d496761f5f3c9110f3 : SUCCESS in 16m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac2640bba45a4f91bc4c25010b8316d6 : SUCCESS in 13m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/62419c63572443208929ed529e50250b : SUCCESS in 18m 31s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2f463d22ec8e436a889f2836d8376bc4 : SUCCESS in 17m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5aaa1e9625c44cd5b55dae992c21d04a : SUCCESS in 15m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3a6f66075cedf98d14e2de51dc61136aa25e99ae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-13 08:46:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"936b7780ce958be7b87cb366d4b1b1fdb8ec8578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-14 11:24:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6d3c9406e7e7217de17c8ec86c36adbc2f9509e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-14 11:26:44.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"77dbb675a0fe85a81d10a26aa2c07b0b3b3f0eb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-14 12:04:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f89bbced56ef2d519f442b88a0da0cd539e4ceaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-14 12:41:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a7027e6cac41cb271e5e42581470665e28a54213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 14:34:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fae3d2b100e45e8a82eafd34f1755d1 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bab6cd26b78a45ffac7cff313e01176f : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18b8d9d5df06475d9aeafd5065dd35b8 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f2779eaf79d4334aea4cbdb04b75649 : SUCCESS in 7m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bd9040770c643e1a629862f0ca191a4 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1ca27191ab47368130dc0db821c21e : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eebbcd6290204ce69ad33f095101ce52 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5e937bc9b334432ea79040f67a5ec78a : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/760f9271a34546db9661502cbf51ad20 : SUCCESS in 15m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7124f53ebbcf4c04a3fd745fb5487d9e : SUCCESS in 19m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5d131e83e5224bdabd4aa15a69321741 : SUCCESS in 1h 14m 02s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/1f870532eeed447693ff3a1070084920 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9274720611ca4c05b281f6e80c8f8dde : SUCCESS in 54m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6d45e684fe364faaaff15e9b1d8822af : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/71ee67bb9bc241ccaa06dce94645ff90 : SUCCESS in 25m 59s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/00042bf2c0a945578f6fea05736be69b : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cde9568c867a4159a6805eb7553bb4d4 : SUCCESS in 16m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/67cf26f8dc6c44cebb6a29a37d4bda26 : SUCCESS in 1h 49m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce277bb9a198421ca91e738b87895a22 : SUCCESS in 1h 12m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7d2769271c0a42778b988df5766e7e87 : SUCCESS in 22m 05s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3cacec66c80a4dc4b45f8139e00b0383 : SUCCESS in 21m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ef2f109bab0456cada4867a211dc0ea : SUCCESS in 17m 46s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6fd2e6479d8d46e39e67f1e4f1ca8d10 : FAILURE in 31m 11s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5af4294d2c5c43e7a53c5909b9b14500 : SUCCESS in 38m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef5f838d93f04fc387b80a92fda64ef5 : SUCCESS in 33m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/82949519ac9545e9829b583c8088c693 : SUCCESS in 17m 52s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/243ca1379f814f84ad857ea8d465aaed : SUCCESS in 17m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/449002941ae44a728d6185306a655623 : SUCCESS in 14m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4a895d131f484d8d889f7bc35b873ae5 : SUCCESS in 30m 54s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/242aca61682c42fba663f260af9af909 : SUCCESS in 17m 30s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2074adc139249aca014a92bc7d22744 : SUCCESS in 14m 21s","accounts_in_message":[],"_revision_number":4},{"id":"40ec65a801b44b7ba60db7b98235f2f2b5938e4c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-14 14:52:26.000000000","message":"Removed reviewer Radosław Piliszek.","accounts_in_message":[],"_revision_number":4},{"id":"5e0844c0b3f11d002b0d38a659e2dfff7919ec8e","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2022-06-25 19:16:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"eea3229c5c88a47813cfc4c789242108573bdc4b","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2022-11-03 07:09:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b30db2aadfc1c477325165922cba7f3f54cb657d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-12-07 14:52:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b66fc7fd5c8bdf6353868c785a41395945ce0b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 16:04:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fed95b06750b48d5b3d15ef20d591141\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3d9d47500b8466b9b0928fc08a20486 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b6c474ba2742ecbe0d586ea1e6b75a : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7e3a89a6eb6458a8b90d205d4b0a714 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16e1ae9901564bf6b1e6d5dc56e5b04d : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b4a3a2361e041f5bb74100fd147ea06 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aae1a29e401456cbf4a3316b4276d00 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/650c28a2f4954b28b35087c947e9004d : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/658a298730f541149e191914192c354e : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4719320294cd4cb8b776dfbb5150446e : SUCCESS in 15m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b92ca7d170a843508149d04c90391d9e : SUCCESS in 23m 43s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f6431d494b041b4b99a06b05299486f : SUCCESS in 50m 16s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e962a82a60e40a0a36c9df941412eda : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed1137d5a72f4eefbd03cd193e051cd6 : SUCCESS in 1h 07m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acaa310c0fe14353befa18dc97599323 : SUCCESS in 16m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41496f4708c9459c99b6019fc5e29abd : SUCCESS in 28m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9e1d4968c034c46a3d526b53f3df574 : SUCCESS in 10m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3901e7db04846829ee06eca040e8295 : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":5},{"id":"5765e00214f8fc1afdb86bc16936e3688f931e39","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-15 14:33:12.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"a587455067d3b8ec8ba7de55cfbca40170a91243","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-22 15:46:56.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a55ad492b99bedc5593a02d7df7843ac1cb73acd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-02-27 11:28:21.000000000","message":"Patch Set 5: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"48affa4dcf307704e8ff0098dfda543d245d29f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-01 17:28:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a2beee3b516bd47d24d7fbe4a94d24dbcec56bc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-01 19:29:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6264552993cb4d13b957f1b26f47a130\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/326fc4c1ddd74881ab2d7e53e1dd7ab5 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e337ae79fc4835b3bf593af53a700e : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a29a05df99bb4d2d97916365d29a6520 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20fd7104fa614b10ad4e86dc3e78efba : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0ad680e95f4344a31953f7cc0989c1 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4f685783a894c15954ff0edf49b75ce : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36fa643c3fb64f81be0c620704b2d506 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc99e7809d5f416887cea5424126db67 : SUCCESS in 7m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a5f828228adc4a248cc5c1c49c3887ff : SUCCESS in 15m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2d137f690a83499a84ca22b23dbedf9e : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/379c1f15438a4d788c70825aa97a61ae : SUCCESS in 44m 47s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c3702d6af53431ea79cb0312f0ca0a3 : SUCCESS in 37m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/949482b3736c4f8482135e0e9eb51c3d : SUCCESS in 1h 02m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/988670208fc6480483614f796106abd5 : SUCCESS in 16m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/491f891786314812a98921f131d2f335 : SUCCESS in 13m 42s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d14168cc11e944589362f9a9d3a0bc52 : SUCCESS in 15m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5610495d624544f0852e604fb71b8c99 : SUCCESS in 1h 55m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96b9a862af374d92bc812cba5e24ba29 : SUCCESS in 52m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1eeb377869b9407094e96ac1b43c4323 : SUCCESS in 17m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/beddbf1584c343c59a68078ab7306447 : SUCCESS in 15m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f3e9ce61456c42578f79f82ce72fbd85 : FAILURE in 29m 24s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ae8a16744e6c4586bae0f96731e0d120 : SUCCESS in 27m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d0a6e9c50d244c9af057cd6776f587b : SUCCESS in 16m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec775093e5b9468d8cefe43b2d685f0c : SUCCESS in 13m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cc164bca6984c0b9071a67aa91c0eab : SUCCESS in 10m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/42ef707fdcc541beb61b195771fa8f6e : SUCCESS in 17m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef1c44388eb6483ba197e2fd61f957ef : SUCCESS in 10m 30s","accounts_in_message":[],"_revision_number":6},{"id":"fbf3ca8754bd3d2ee015795618a772c8940b37e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 16:06:48.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"80771916f9833d8122d41762d09ccf67abbb3adf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 16:06:54.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"522bd640a81743ea1b9fbcd6079872303c5c91c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:58:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f181b7ca01374799a453fe811b303733\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e2f6b7a9eb44a14a10b57549c8530fc : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f11e183b9c5542a3afd42a033de61b78 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3d78663caf945af8843cc1e0461d5a9 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce9bfd37a0ff42948680953647f7cbb9 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1959ad9b6fea429894700d33af24d8ed : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/595c0e9fa6644901be2fa659aa944836 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ffd3b699232a433686d9f652d5d4a098 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/248d8e86d9054a4c9c788344d433fe14 : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ed13dcdc4dc54cdca4b3e39b067972db : SUCCESS in 16m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/27edb733a0d947659a4c294ec547492b : SUCCESS in 44m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c6d87557ebcc4058aaec685368d76e22 : SUCCESS in 55m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3303d6110c514c87b68474fee26b6586 : SUCCESS in 53m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/54e8914eefcb4376ac94cd21c1c75a27 : SUCCESS in 50m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4a0c02ceb3a84d2f9f37c16bfa008296 : SUCCESS in 15m 39s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0e1ea613598a458ca6bc5e73e91527d2 : SUCCESS in 13m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b03879e51d7d497cb22c0a870263e86f : SUCCESS in 13m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/59fb48730f5840d1979c808631234328 : SUCCESS in 1h 45m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4296f2d799e7453bb69ba084f70e4e3a : SUCCESS in 1h 08m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f34ca14072614c8c9f026970801d957b : SUCCESS in 18m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3829e6a8d3394aadab09d763d9a76280 : SUCCESS in 13m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fb560d2e43a941439f4096728ba3af66 : FAILURE in 30m 55s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c580cc96f7cf4b21848a9117e799b628 : SUCCESS in 25m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5e68f407f404755bb9279c7f00be7ef : SUCCESS in 23m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/234cdea6e77743469e40317e01f0033e : SUCCESS in 13m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5b6c8a2f824498aa7a114408d866d57 : SUCCESS in 10m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4ee0ae6c29904cb789bede7aeb5361f7 : SUCCESS in 17m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66b65427f272460b8b1812c48ee0941e : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":7},{"id":"36c0146ae6960657042e10c3c8abdd9346056482","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-19 18:07:01.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d7f3620d859bb2980bb2140b37214a3b34de2d2f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-20 07:02:15.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1 Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"9315c67646edf5251df1c2702209ccbba9594374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 07:02:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"89079fab649957e4dfb600644fdf31176099dd6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 08:53:11.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57e7f2a25f7642e6bba7d221b40528da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd1fbd5e67f4ef99521911353d3ec27 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1afcbaceaf04c2d89c4d64b06a0ac46 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a8e1f87b2f14778ba5fd1ff08af08c5 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065dc8ddc5e740abb2b22a43cdec5f41 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29cfd43d47f244b0bd1506aa1dccf93a : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/71881eedaf864257a2ea5b61ce643d59 : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/882a66d3cd714ccf94c92916427311a4 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/887af8033adc41c8b6d92c00de4145a0 : SUCCESS in 14m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/56a38cf5a0f6430387324e704e338ed5 : SUCCESS in 52m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32fc8c584732470db54724104ad2cc67 : SUCCESS in 50m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c02b10c575a24c43987f6363b45343cb : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7850bca5b9484713ad9c0dfa6a0b0dd8 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9015427ad0b34dca9c83996bd7bd6a5b : SUCCESS in 12m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b9513eda21824696aefb5dd82c19f594 : SUCCESS in 1h 44m 25s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a461f7f83a54c9fb1174059143b8ab0 : SUCCESS in 1h 26m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e83f590ba5ea46be8a364db3c32fb16a : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/728712b074864e24b86a546d0aabed5d : SUCCESS in 15m 23s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4665c816c4a148388c90c6214cf6b528 : SUCCESS in 27m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/848dae8d664948f8bb8f94195f988f95 : SUCCESS in 23m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/345f7357ba0c471aa76e1bac3cb78b90 : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a19dc4d98db40c48ab2fd9afc9f337b : SUCCESS in 11m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3f71602767dd4541959cf82689f53400 : SUCCESS in 15m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48e6574b3d3b4faa853a48a4c4b098b5 : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":7},{"id":"d40e95da78020ad21e18cd7b9281fff8fad16bd6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 08:53:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"8085d11e09aa2db4e81f08974aac0f8414eb32b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 08:54:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bd550d187b94746af161c88c90dcf28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ec5a2d99b741d1b22fb1da45e9536a : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adfb17b49b0442bb98470abece33a220 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"070621a45df11fd2487ddc38f715c1075524ee93","revisions":{"2bebc533e48ef53477209be6856959aa56ce9dd1":{"kind":"REWORK","_number":1,"created":"2022-06-10 09:56:00.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/845338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/845338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/845338/1"}}},"commit":{"parents":[{"commit":"b77033970a79587b3be5c04a5b81e8b88e21dadf","subject":"Add missing declaration of kolla_enable_hacluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b77033970a79587b3be5c04a5b81e8b88e21dadf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 20:10:52.000000000","tz":180},"subject":"Allow to use own repos.yaml file to build containers","message":"Allow to use own repos.yaml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.yaml file to build containers.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bebc533e48ef53477209be6856959aa56ce9dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bebc533e48ef53477209be6856959aa56ce9dd1"}]},"branch":"refs/heads/master"},"80e9dc89021e7ca34fd6c8468cf209da43813763":{"kind":"REWORK","_number":2,"created":"2022-06-14 11:24:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/845338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/845338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/845338/2"}}},"commit":{"parents":[{"commit":"b77033970a79587b3be5c04a5b81e8b88e21dadf","subject":"Add missing declaration of kolla_enable_hacluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b77033970a79587b3be5c04a5b81e8b88e21dadf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-14 11:24:14.000000000","tz":180},"subject":"Allow to use own repos.yaml file to build containers","message":"Allow to use own repos.yaml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.yaml file to build containers.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80e9dc89021e7ca34fd6c8468cf209da43813763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80e9dc89021e7ca34fd6c8468cf209da43813763"}]},"branch":"refs/heads/master"},"27b5d93f92393e9dcbd3831a60251f4e887eafe1":{"kind":"REWORK","_number":3,"created":"2022-06-14 12:04:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/845338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/845338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/845338/3"}}},"commit":{"parents":[{"commit":"b77033970a79587b3be5c04a5b81e8b88e21dadf","subject":"Add missing declaration of kolla_enable_hacluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b77033970a79587b3be5c04a5b81e8b88e21dadf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-14 12:04:17.000000000","tz":180},"subject":"Allow to use own repos.yaml file to build containers","message":"Allow to use own repos.yaml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.yaml file to build containers.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27b5d93f92393e9dcbd3831a60251f4e887eafe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27b5d93f92393e9dcbd3831a60251f4e887eafe1"}]},"branch":"refs/heads/master"},"0d5cf2646567c64a5392b34e8447bded21fcdf18":{"kind":"REWORK","_number":4,"created":"2022-06-14 12:41:15.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/845338/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/845338/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/845338/4"}}},"commit":{"parents":[{"commit":"b77033970a79587b3be5c04a5b81e8b88e21dadf","subject":"Add missing declaration of kolla_enable_hacluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b77033970a79587b3be5c04a5b81e8b88e21dadf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-14 12:41:03.000000000","tz":180},"subject":"Allow to use own repos.yaml file to build containers","message":"Allow to use own repos.yaml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.yaml file to build containers.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d5cf2646567c64a5392b34e8447bded21fcdf18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d5cf2646567c64a5392b34e8447bded21fcdf18"}]},"branch":"refs/heads/master"},"019f25918c014e3dda2c7f9b7982861f2309e165":{"kind":"REWORK","_number":5,"created":"2022-12-07 14:52:46.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/845338/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/845338/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/845338/5"}}},"commit":{"parents":[{"commit":"a50fdb6b7f859aab20bcdb668b0de77bed3d0010","subject":"Merge \"Remove kolla_install_type variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a50fdb6b7f859aab20bcdb668b0de77bed3d0010"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-12-07 14:52:14.000000000","tz":180},"subject":"Allow to use own repos.yaml file to build containers","message":"Allow to use own repos.yaml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.yaml file to build containers.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/019f25918c014e3dda2c7f9b7982861f2309e165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/019f25918c014e3dda2c7f9b7982861f2309e165"}]},"branch":"refs/heads/master"},"f8bdebb797b0e8079ca293607cde4f684905f394":{"kind":"REWORK","_number":6,"created":"2023-04-01 17:28:44.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/845338/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/845338/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/845338/6"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-01 15:54:21.000000000","tz":180},"subject":"Allow to use own repos.y[a]ml file to build containers","message":"Allow to use own repos.y[a]ml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.y[a]ml file to build containers.\nMultiple Environments supported.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8bdebb797b0e8079ca293607cde4f684905f394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8bdebb797b0e8079ca293607cde4f684905f394"}]},"branch":"refs/heads/master"},"070621a45df11fd2487ddc38f715c1075524ee93":{"kind":"REWORK","_number":7,"created":"2023-04-19 16:06:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/38/845338/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/845338/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/845338/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/845338/7"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-19 16:06:46.000000000","tz":0},"subject":"Allow to use own repos.y[a]ml file to build containers","message":"Allow to use own repos.y[a]ml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.y[a]ml file to build containers.\nMultiple Environments supported.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/070621a45df11fd2487ddc38f715c1075524ee93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/070621a45df11fd2487ddc38f715c1075524ee93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
