)]}'
{"id":"openstack%2Fkayobe~845882","triplet_id":"openstack%2Fkayobe~master~I1bd306698a0b2afd1b824eda3389fcc3f919a1fb","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"33436":{"account":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"last_update":"2022-12-16 09:00:20.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-12-16 09:00:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1bd306698a0b2afd1b824eda3389fcc3f919a1fb","subject":"Update python testing as per zed cycle testing runtime","status":"MERGED","created":"2022-06-15 06:52:52.000000000","updated":"2022-12-16 09:01:27.000000000","submitted":"2022-12-16 09:00:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845882","meta_rev_id":"fb948a5ccbeec9e2367d695ec612d58528edd620","_number":845882,"virtual_id_number":845882,"owner":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-16 09:00:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-16 08:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-15 20:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-16 08:39:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-15 07:05:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-15 07:38:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-06-21 07:37:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-12-15 15:28:40.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2022-12-16 08:39:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1053ea729f173d0fa93e9b53b578899821769d55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-06-15 06:52:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adc1e15da66e486fc3cde95414c9d6a9be1c2b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 07:05:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eb5b7da5328c47cb86bf9a6e8f9c48a3 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cf9f8aa348447d69458df974241c7b6 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce776ebd50041559756aa4ce6b2c806 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c01819bd1f441d8a1ad16a0400b3903 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2df9dc2594724f3fb863cdb352a115b2 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b336ca343d6a432fb1769a1fa154bebc : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c2089e843574fa8a7c1ada1c8b94310 : SUCCESS in 10m 19s","accounts_in_message":[],"_revision_number":1},{"id":"adc2ab9c1e073698d0dc41658f8cfdb35e6545d0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-15 07:38:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da5aea749f07e430370892ec96b95d669eaffff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-06-16 02:28:06.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1fb6b45c8f4d5280d18f61c8ef78e158a7f38f2d","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-06-16 02:30:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0bd075f188a3a55f9b4ab224b74be02048b52013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 02:48:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1ea62de89870432b852fcdeaec4762c6 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81b7a32a5ae74f0396c969d6c7bc40fa : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923f6b2e4f33421d84c811d30c31b385 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea5baff293554791b54afa78002756ef : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1403db868fca495aa98cafab04f02ce1 : SUCCESS in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/716693f3143a4c77a8fa9db1ac311ad0 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380b34491c0a4688beda84f30a53ad24 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":2},{"id":"35a89315a02035bcedddb0eb91a99324f3f1eacb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-21 07:37:34.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"15ca5df2e30cf8ea40344295d0af4417cc9f6b65","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-06-21 10:01:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36fb91ce9bbb039d192e13de83939ab0bd361354","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 15:26:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cc1d00a5a34d47a306a06f0f71d37ba705ffb3de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 15:26:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a19857b675ff14e17d113d777f494d8b1a92eb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 15:40:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/129cea74924d4c02b337a25832ca996c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/74d9178ba2604e6fadaa03e49bd941c4 : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f7b1520417542f0b743755ba36a30bc : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98ca5924dbda493b910358e54b0f34c4 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24e7530390844f0ea4333847e0e26926 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be557e687def48c687a89a5dcdb75e48 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6355df6d7f734f8db56d79ed1dff559e : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c7e2f27d394a2f8990adf1dadfe4a1 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":3},{"id":"8c34a80166ff4c86c014e749d62565243ebe97bd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 20:43:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d256ec91186b8d657924b71cde56bebd27358ded","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-16 08:39:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"986fa476de3fcc42dd82a5f75b1a71571c7c33eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 08:39:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"29e22d2a9c6adb185107e11cc66d006e7a5b3e6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 09:00:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ab4d18a71042a0a5a3e622e1f3dc3b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5bf116046a8645c2bfa941a342c61a16 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484a4bc58eea40189d04cb9e3170a68b : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0d135a4d9084a7cb0b88721d2c6c441 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bcc7dd1901a46edb161325927c5ece1 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23670bdf68d5437c8e5329bb8693b3d7 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":3},{"id":"379e5317cd420406189f0fef5b29bb2a18304c36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 09:00:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fb948a5ccbeec9e2367d695ec612d58528edd620","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 09:01:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad2e6a48f6324b1c8d926af74be62b8c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83187c672e9441358dc7776c2c317caa : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c84a7f9a9c725dcedb33d8c50ce2477d25bf5735","revisions":{"9492a6cf9dff576f8754d9dbb6a703918cfac7f1":{"kind":"REWORK","_number":1,"created":"2022-06-15 06:52:52.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/82/845882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/845882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/845882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/845882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/845882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/845882/1"}}},"commit":{"parents":[{"commit":"c8b89504e94a78cf4673152d8387d0aa0c03544e","subject":"Merge \"Add support for dellos10 switch configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8b89504e94a78cf4673152d8387d0aa0c03544e"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-15 06:52:18.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-15 06:52:33.000000000","tz":480},"subject":"Update python testing as per zed cycle teting runtime","message":"Update python testing as per zed cycle teting runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Removing the py36 centos8 job as well as\nupdating the python classifier also to reflect the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I1bd306698a0b2afd1b824eda3389fcc3f919a1fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9492a6cf9dff576f8754d9dbb6a703918cfac7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9492a6cf9dff576f8754d9dbb6a703918cfac7f1"}]},"branch":"refs/heads/master"},"62c4baa2e42030d8a0ce43476412cd36964849d7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-16 02:28:06.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/82/845882/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/845882/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/845882/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/845882/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/845882/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/845882/2"}}},"commit":{"parents":[{"commit":"c8b89504e94a78cf4673152d8387d0aa0c03544e","subject":"Merge \"Add support for dellos10 switch configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8b89504e94a78cf4673152d8387d0aa0c03544e"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-15 06:52:18.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-16 02:27:27.000000000","tz":480},"subject":"Update python testing as per zed cycle testing runtime","message":"Update python testing as per zed cycle testing runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Removing the py36 centos8 job as well as\nupdating the python classifier also to reflect the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I1bd306698a0b2afd1b824eda3389fcc3f919a1fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62c4baa2e42030d8a0ce43476412cd36964849d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62c4baa2e42030d8a0ce43476412cd36964849d7"}]},"branch":"refs/heads/master"},"c84a7f9a9c725dcedb33d8c50ce2477d25bf5735":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-12-15 15:26:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/845882/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/845882/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/845882/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/845882/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/845882/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/845882/3"}}},"commit":{"parents":[{"commit":"63e8a47206d6faaf69c18e6f33553f304ae42c2c","subject":"[release] Clean up release notes for Zed release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/63e8a47206d6faaf69c18e6f33553f304ae42c2c"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-15 06:52:18.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-15 15:26:35.000000000","tz":0},"subject":"Update python testing as per zed cycle testing runtime","message":"Update python testing as per zed cycle testing runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Removing the py36 centos8 job as well as\nupdating the python classifier also to reflect the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I1bd306698a0b2afd1b824eda3389fcc3f919a1fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c84a7f9a9c725dcedb33d8c50ce2477d25bf5735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c84a7f9a9c725dcedb33d8c50ce2477d25bf5735"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
