)]}'
{"id":"openstack%2Fkayobe~846270","triplet_id":"openstack%2Fkayobe~master~I25fd8d4e4bcef4b015a1634ace927e37aa2e80a9","project":"openstack/kayobe","branch":"master","topic":"remove-timezone","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-08-26 16:45:30.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-26 16:45:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I25fd8d4e4bcef4b015a1634ace927e37aa2e80a9","subject":"Remove timezone playbook","status":"MERGED","created":"2022-06-17 05:39:29.000000000","updated":"2022-08-26 16:46:35.000000000","submitted":"2022-08-26 16:45:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"846270-remove-timezone","meta_rev_id":"078d94223edf060061654b6ead473bf320d6f160","_number":846270,"virtual_id_number":846270,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 16:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-08-26 14:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-06-17 08:44:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-08-26 14:26:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 08:05:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 08:44:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-06-21 07:38:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2022-08-26 14:26:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"daa408d9eb175764050b1602ca2c0e93df9a95e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-17 05:39:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ce4f22f225a3e531e21f9e26d4858039eca42b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 08:05:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d520225f73044b1ae9e5d03706c5605 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e325a57e012341fa8d9f4b4acde6adf6 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f05662309e024ea5aa08267edb2d35c0 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/984aa78731b34793aece82ae828791a4 : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5704a846c1c4b10bacbac08d25d357f : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7f4dfd89bb4fdfb1e819988c1ed6eb : SUCCESS in 9m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78bb4669f8f54649b214d8aae6cec30c : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ce027afd44a4203b1c3c9b7acb4cf68 : SUCCESS in 13m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c1ab0e9e66c42d683cb64100453b9f1 : SUCCESS in 14m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/00f627efa7304bc8a12bca52435615e8 : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/0aade3bc01834b7b8ee7784bce93a768 : TIMED_OUT in 2h 05m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4149eceadcdd434bb594c21d836b22a1 : SUCCESS in 45m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/21b6ad18306b4b519069c909b08504f4 : TIMED_OUT in 2h 05m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a95c8ccd68ba4de9a9cf682371945841 : SUCCESS in 25m 26s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/6e491795587841658995cdb886e59112 : SUCCESS in 20m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9445a736cac74bea80c22a1328844908 : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6fe4108a76454b09a96b1198c7b1b6ea : SUCCESS in 1h 49m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c86ac787394b48ed8aee84216ab3d7cb : SUCCESS in 2h 20m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4bc090135e6e4781b1c3bd86ef06b900 : SUCCESS in 1h 00m 58s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9009d18a75634a3d8c25c1aad4d218ff : SUCCESS in 40m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8c097e29c2348c9953a8a47362ea428 : SUCCESS in 15m 54s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/39ef6e6a2a7142e6ac7d41e27aad9316 : FAILURE in 1h 00m 13s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/977f5f9913c147338474888e0a489de9 : SUCCESS in 1h 28m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcc71fe66c4740109bc60de4d8bb036b : SUCCESS in 30m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e93b73bfe4e145519096f4d867dc9304 : SUCCESS in 16m 57s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9b437856a1984c179e7aa84f5a1d1a9c : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6523c0bce80649948ad052fe6ed870c8 : SUCCESS in 15m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3a72fea92fe049d8bc755033f8a2a0c5 : SUCCESS in 18m 46s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b03e8ff84ea3419abd6b853db56591bb : SUCCESS in 18m 09s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90a8cfd2142d48d7a73c97573438e0c9 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8f420e473ec0c49b3e7204ddc690a3967789b2dd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-17 08:11:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db066ad26f549a912d39ad5f0cdf3aeec2f95ee5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-17 08:44:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"245e77dbc90a32b63ee5f148805bc472b28d12e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 10:10:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5eaf6547ec54bab8c6c819ec273896a : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75957ef1572b4d4eb0233bd95d172ccf : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4e5cc388417404b8d5a76516c9b71b3 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94b84b1a6fe94a29997d711ed4385137 : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d30bad61782a4447adc38e13b9c019df : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5b42c3c846e4fd1821df11a9596aa32 : SUCCESS in 9m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41bd6c2f3d6d46eabdf21520c8e37125 : SUCCESS in 7m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0bee27bb94ff4fb7a078dad325ab8c40 : SUCCESS in 14m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17dffad900e84ab28ed76ba84fe07904 : SUCCESS in 20m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3b32e16a1261483c91cc741b9041d23e : SUCCESS in 56m 20s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/0c1a3a3d10c94ce58b50b877fb8c1020 : SUCCESS in 1h 10m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a671e6621a64187880449fbe2f0d1ef : SUCCESS in 53m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/716d6bb517a441bab06445042b06d4b0 : SUCCESS in 57m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4bb93d9290ac4936a722ac4409194d0e : SUCCESS in 25m 10s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/31442ec1581e4fcb9d7d60d1b5eb2bf6 : SUCCESS in 25m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/594f89d727bd4c8cb20a676585f411aa : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7528a94cd5f747498528aacee2b96346 : SUCCESS in 1h 41m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e5f682a4d334107ac7a57329d173b01 : SUCCESS in 1h 21m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/baed13cc938a4fa19314427912aca383 : SUCCESS in 21m 58s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d48c827159d44f70b4e16e2d6139efd6 : SUCCESS in 21m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2fee6566a0c491595ca4a62f3f9c531 : SUCCESS in 16m 26s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/727999bdfa7548e8acd685418850b58c : SUCCESS in 52m 51s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/53a320fed6764643a7cc4da217f73e3e : SUCCESS in 35m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6db3bc3ccc6c40e9a33bf498bb2d4807 : SUCCESS in 29m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dc0c5aa1639e4f999a9dfa72e30f6d81 : SUCCESS in 21m 10s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/41b209b3d1a04275aa8fe87d8f110199 : SUCCESS in 19m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86368863cfc84856aa5d43659e1877c2 : SUCCESS in 12m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/46f048ce27844d88bcfbdca4bb524867 : SUCCESS in 17m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/31892840d4074d5bb930f68f63663499 : SUCCESS in 19m 48s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65e06945791e4708b915d10bc80118a6 : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3cac372a0deb139f320af0d57c8d597315d1ef99","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-26 14:26:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17639f2c8318193511edbe13539bd78c6e896e0f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-26 14:26:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7319bac5e5725264056a349cece552860a21823","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 14:27:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"417ddb6ed854c8f4dc98d5b0ea386de45702961b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:45:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be5c047c9204f3893ef8cb6d69f31a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44bc642ef3cd43a4871b59302794e97f : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2036ccf3a237420e8785645bb3f903c4 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/013f4839b70241ec8a744aee5f2a31f0 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42e2aaf0bf042f2a92d45260314617d : SUCCESS in 10m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1bdd9948c1549a89609aedf7fd41bb6 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6178b85f4bc047c8a0635f36ebedaf51 : SUCCESS in 16m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef90ae16c12e4fec97db63da67b38249 : SUCCESS in 20m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9a032750be142f79ad8dbd8d301ddf8 : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a1833659e0864f34a0d79de39b0b1fc2 : SUCCESS in 1h 06m 58s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3586c0fce3374d21b04519ae5b297f01 : SUCCESS in 53m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bd5cd2bbce554adbb021b86288316a87 : SUCCESS in 1h 06m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4c205b05a61b485a82d47c092ef30a1f : SUCCESS in 24m 55s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2bface864f674924b7fd533bea4b6608 : SUCCESS in 22m 34s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d25b61379cc84d71aee2c888b734ccd7 : SUCCESS in 17m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/de29f8aa70af409b97c81d6829499f3c : SUCCESS in 2h 11m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eac7aad115244d368e88758415b0a70e : SUCCESS in 1h 15m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1a58e633e96547c99060d173b827c0d3 : SUCCESS in 24m 19s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f2717a3a775549e580013ce25750fd1a : SUCCESS in 21m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e223690cf4b438bad01f7a4135b20f4 : SUCCESS in 16m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/616f49cc72dc4b8e845e24cebf2d1dab : SUCCESS in 43m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee5d916c038a416cbc1186ca8120d87d : SUCCESS in 30m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/35586a166ee146428f2ab07e076e58e1 : SUCCESS in 19m 07s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2ef0dff05c7340938c1daa268ec42dca : SUCCESS in 17m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/419074fa1b9a4ad7b6276571b9e256a7 : SUCCESS in 14m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e40b47c36f094ddcbde5b564d3427211 : SUCCESS in 21m 15s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6f9c8c83a26b486aaf8c916366d29b47 : SUCCESS in 18m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aad4ad50a9684366b27a71de448abde9 : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"eccfa9455527ad2e2cc40385184d512e1450ffbe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:45:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"078d94223edf060061654b6ead473bf320d6f160","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:46:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bf5f0e5e9fd4bfbba3a27fc86be4839\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e190bfa4f534407588421a4bdfe93340 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66f763ea06e6719509270fcbbae7ccc0215c135d","revisions":{"66f763ea06e6719509270fcbbae7ccc0215c135d":{"kind":"REWORK","_number":1,"created":"2022-06-17 05:39:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/70/846270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/846270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/846270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/846270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/846270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/846270/1"}}},"commit":{"parents":[{"commit":"258dd730d197a3d343123b21a13fe7704ec19376","subject":"Merge \"Force use of yoga upper constraints in ipa-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/258dd730d197a3d343123b21a13fe7704ec19376"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-17 05:38:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-17 05:38:53.000000000","tz":120},"subject":"Remove timezone playbook","message":"Remove timezone playbook\n\nIt is time.\n\nChange-Id: I25fd8d4e4bcef4b015a1634ace927e37aa2e80a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66f763ea06e6719509270fcbbae7ccc0215c135d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66f763ea06e6719509270fcbbae7ccc0215c135d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
