)]}'
{"id":"openstack%2Fkayobe~847572","triplet_id":"openstack%2Fkayobe~master~Ie4bba0f1bdc82657572e0f06a71db1140c0bd3a4","project":"openstack/kayobe","branch":"master","topic":"kolla-ansible-playbook","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-08-26 16:45:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie4bba0f1bdc82657572e0f06a71db1140c0bd3a4","subject":"Support --kolla-playbook argument","status":"MERGED","created":"2022-06-24 11:27:21.000000000","updated":"2022-08-26 16:45:36.000000000","submitted":"2022-08-26 16:45:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847572-kolla-ansible-playbook","meta_rev_id":"22f4c96df6a72b3f5ad848cdb94b142e259a7fc5","_number":847572,"virtual_id_number":847572,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 16:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-08-26 14:37:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-05 10:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-08-26 14:37:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 13:36:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 15:42:53.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-08-05 10:19:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-08-26 14:37:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"421452aae74cb5cff5eddf059dfcecf4efee066a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-24 11:27:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e9467a3d9a2072ec82912b1cdcfef9d6c55d670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 13:36:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bb85af4705849d6b58af96aacc22600 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb16c4a8bae444f6bdb0e689389c0241 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16f0891505f5450aab307fa3c9c6017b : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9d887588c22493c8f60a4774dce6b2d : SUCCESS in 6m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b9e2b5304554fa1936fff0c0cd3db76 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba18fd1b79c74d30bd0b92711c6b582b : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1750272d4e6644e5918e7855445f86a6 : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/530cb670014248fd8778cbab0f2d3388 : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00d4c64e13be4c3eb1115e4df2a57608 : SUCCESS in 15m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af63cf411f8d4875baa733ad62d54ba9 : SUCCESS in 18m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/63809bb90e1c43e8a868dc663452934d : SUCCESS in 1h 16m 46s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8865d5f930c54d3bb89c043cd61550be : SUCCESS in 1h 10m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5dc1fe2accf454f8a104f8934b7b28b : SUCCESS in 54m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fd78b79c98a84eec95d9d0cbae8ccd45 : SUCCESS in 1h 18m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3573fe2348d442229080b4d646d61e60 : SUCCESS in 23m 32s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/01e9ec9b10c34d55beea158ba5a6fc1c : SUCCESS in 20m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0855bdecd75d49ce9618aac4328fc418 : SUCCESS in 15m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7f1ba9ada8874d41ab924b1c289bf483 : SUCCESS in 2h 02m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08b261636c2844c4842ba0a0161f54af : SUCCESS in 1h 18m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/13fec04101234613bd7cb0b7e6b353cd : SUCCESS in 22m 09s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/7ab7dce9396f4de2b8058f60fd5f19f6 : SUCCESS in 20m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3f7b7420db2433e97bcf94c3216e251 : SUCCESS in 17m 47s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/4e57e904c1484e1c81cf9bf127038c46 : FAILURE in 31m 27s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7692eaeaacc84a2581ed174b4f77eb88 : SUCCESS in 43m 37s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96d012f1cc744a1d9d1dd7cba79368ce : SUCCESS in 29m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/999d25592f46441b9434b18ea9093c13 : SUCCESS in 17m 38s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/29a1fdc5ef624580b0e9931b3eac57b4 : SUCCESS in 15m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50b228e97a4242598e085a62af349c78 : SUCCESS in 15m 41s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b87f02b551d64982b94e2facd4adfab8 : SUCCESS in 20m 23s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fc5c881fe3b74a3aab8964b9def13691 : SUCCESS in 19m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/011ab182619b4e8686768ad86571083d : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b00fcbfd048d435f27734c2274f06582f4f433cf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-19 15:42:53.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2635ef3552b7f2c472feb2e4c5a5cc1c6a522222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-07-26 08:27:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da9bc44afd9a48d734d738e6620457f64c4d6c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 10:16:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0311ea2dda24e4fa78c7cf4bb0abb56\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19b2238fadbd41a7a05d1cb7a1682303 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3840bcd3b496497f8a2678a49b0cf478 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e101b06267c453687888dcb7f4f5c1f : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee5c6178eaf249dd93be282acdf91162 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faf62c1cac534f94a8d727ae2ced9985 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b4314fa0ce4e63a15f376da5b5991a : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/067f9e0eb24847818aa30c65ff51b5d2 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1d4c67e90490423f81035d5012789a8b : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/209b9ff938b04c3d8f0395e35750574a : SUCCESS in 13m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e58d6a81f6a4d8a871d74a6f683a177 : SUCCESS in 17m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/424f141600cc4d6fa93dccbf49d526aa : SUCCESS in 1h 09m 48s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7eead4d36e054253a918b4030187fb10 : SUCCESS in 1h 11m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a336a8402c0542998b75f01fd14ea5ca : FAILURE in 41m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d41150c8cfdb4655ad6d62e93fd58d6a : SUCCESS in 1h 10m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f2c7d02e8bb44fcc937eb3aeaae9ca76 : SUCCESS in 26m 33s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5062990250b3440eabb56ff39ef084ce : SUCCESS in 22m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2224cc8dadeb4b46baadceeda223702d : SUCCESS in 16m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/22bcaae11f404bef9e010683352afbfc : SUCCESS in 1h 46m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c279a75c38f240b7b70a0f512165f3d2 : SUCCESS in 1h 03m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d58ed13237df4cacbac38edf41c37c9a : SUCCESS in 21m 43s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6e9229df873a414b988725fd3f1c5424 : SUCCESS in 24m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34fddc3718824b609bf282b18dcc8e29 : FAILURE in 18m 06s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/776ca6d6cfc144d382647b2ee1736226 : FAILURE in 31m 05s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e5d7eb5611ca4bbbb1b6f7ec43865a13 : SUCCESS in 45m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de0d6e32fff442b58f1295504ae8dbf2 : FAILURE in 25m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/df3c27dde3fd4809a4a9d46c2143411a : SUCCESS in 18m 22s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/96d85adf5e2f4bedb5aaaf905b35faf3 : SUCCESS in 19m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e72d94fb92d45cba4eb2d96edee974c : SUCCESS in 11m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/311b7060a3c04be6a3a15326fb4743dd : SUCCESS in 16m 41s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5e9862aa10124eadb9a7fbf6a5e1d5fc : SUCCESS in 23m 36s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/792c55d3a3474b83a62c2c9771536e6f : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e43c300f0f51a1a0cf93a76269955dc09fee8ac1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 10:19:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6e263384b951d3f5382179160dfb8c0685416379","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 10:21:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41d898f9fdf313b8f9def2c81eb76356d9d62c8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 12:06:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb231df5d0ec46099c06fb803b9662be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c0701e999234aaba8cfbc2fa3458b0a : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3943c9decce43028c25f7f0d2596414 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de1170a5f6aa4c498fdc05dccc14015c : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f8dff6b6aaa496dae0779a407860d3a : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50a417ef28fc477588e14ef0f946be0f : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3847caf58a2c4d37b68719b229aa61a7 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b2ff8b2fda44530a73e91ef53962085 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d321f6b0d46b48b8aaa3be5589cbc4b0 : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f87e5e4991e4f25adb061e62fa5ab2a : SUCCESS in 13m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b18438dfb754521b16f9eb4c2d7852a : SUCCESS in 17m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e313675bff4f4f72b8e4645556086ca1 : SUCCESS in 1h 08m 55s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/50055ce9fa574cae9885b1da833b4a19 : SUCCESS in 1h 09m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6ee3fa61f7149669027fe36bad1831d : SUCCESS in 54m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1876c725391f4bea80c2222765db7a05 : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1921f8cde78b41ab98d19e471bea97ee : SUCCESS in 25m 32s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/01e70baee3084cc9ae4e85777879f26d : SUCCESS in 22m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8d4b3c646d746aea884faf053a9653a : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9258861ba8d04145bfb6a14958c2c8c9 : SUCCESS in 1h 37m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3280c71361444c3aa83d37529becd610 : SUCCESS in 1h 37m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e519e4341df440a2825362b43b5fdb41 : SUCCESS in 22m 19s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4c8014471c854e8093b7b2515907324b : SUCCESS in 21m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9143c7f752d046f7b9a330a771af32de : SUCCESS in 14m 24s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1bee75b032fc48ffaea3637916caa0b1 : FAILURE in 31m 00s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc9b374c4a394486a754e42d2f3329f8 : SUCCESS in 39m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ab0c84416ca4402a6fd2e64c9fd6b0c : SUCCESS in 30m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3367c5350791461ab4b7e2bd97b299d0 : SUCCESS in 13m 08s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e3493dc236024d339c78d03a8d24451d : SUCCESS in 16m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b9fca7e581d4d64b7863bcc99039941 : SUCCESS in 14m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8b39f06715e04ab1a1dad09f22762605 : SUCCESS in 22m 24s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/70220125e916417aa99ae716d03e8ef7 : SUCCESS in 17m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6c8f43f8e184f37913ae97cf1b3aa3f : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":2},{"id":"651d840e2bc381b070423ee4d0f475e14dc37da7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-26 14:37:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d18c707ada67a5a98f13db49f6371c844b9a1e17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 14:38:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b61a1f8f64af0d5116f4c8fd185bbbd93ff67c46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:45:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ca11ab6774e41c5afa90803f99c84cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b81f793995ec4eba8079d3ed20b7ec82 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be3db15847a84ee9ab2016152d84bfca : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/940ddf857c3849938bdacd784c992d21 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/368b0fb3b9ab49dca88225d9424d321c : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6206d8b78c14ff1bcc2014e4a9f16af : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46f89550164e40918bc18df0a9a2066c : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec264982361c4558b708708354f06ad6 : SUCCESS in 15m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/43f3c25a19e64f4d924cd2142b57e8a0 : SUCCESS in 16m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0b3525a72dd54913b99d2ece2c08cdf9 : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/701b187aa2cd4c6ca5574b8909384400 : SUCCESS in 1h 09m 09s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e587b0b930b2432cb1e5c0e083499879 : SUCCESS in 50m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/066dd023700d477fa1dacb0f4c33170d : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/797835d0773d46fcb33b75db038169da : SUCCESS in 22m 58s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/57a1fccde7b54218b2b81ceb960dc75b : SUCCESS in 23m 06s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d655142f16b54f848cd18432f2f29bf7 : SUCCESS in 17m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/567c516b313b498da71f6d006b77cdff : SUCCESS in 1h 40m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e94d110cae494a2db102ef1ca9f1c919 : SUCCESS in 1h 16m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/21ca9b2cc2204295beaf86e2b9ca5985 : SUCCESS in 25m 47s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9f49357e5c4f4c72858790f9b46b4ddc : SUCCESS in 20m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e8e7598bcbf4ebf8255ae9115f7f7b5 : SUCCESS in 14m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b3bb978f118042358a07b800cdea0ac5 : SUCCESS in 45m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6b847c39f2c4716b3797520d55ad6b3 : SUCCESS in 30m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/afdef829fa5f48378ec81a9a303a199b : SUCCESS in 21m 50s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3fa3a646ea8c4e12b014159ac3410abc : SUCCESS in 17m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a344778e0c744275a1a6c00dff2296ce : SUCCESS in 12m 28s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b08d645515e1495d803f3983e9734d14 : SUCCESS in 20m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/87e4dce919e24984be83ac0ce4a92606 : SUCCESS in 19m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22ed1f82df144d1e8f301ec39a1cce0a : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":2},{"id":"22f4c96df6a72b3f5ad848cdb94b142e259a7fc5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:45:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79a8610030b214dd013f167b66464b79be937f67","revisions":{"386ba893953b8fa1db99868d16e67609de22f366":{"kind":"REWORK","_number":1,"created":"2022-06-24 11:27:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/847572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/847572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/847572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/847572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/847572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/847572/1"}}},"commit":{"parents":[{"commit":"0c9912ece163d90eee6e0990dabc8fa59033bc38","subject":"Merge \"Use openstack-ci-mirrors in overcloud and ipa image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c9912ece163d90eee6e0990dabc8fa59033bc38"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-24 11:20:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-24 11:24:20.000000000","tz":60},"subject":"Support --kolla-playbook argument","message":"Support --kolla-playbook argument\n\nAdds support for specifying a custom playbook when running Kolla Ansible\ncommands via a \u0027--kolla-playbook\u0027 argument.\n\nChange-Id: Ie4bba0f1bdc82657572e0f06a71db1140c0bd3a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/386ba893953b8fa1db99868d16e67609de22f366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/386ba893953b8fa1db99868d16e67609de22f366"}]},"branch":"refs/heads/master"},"79a8610030b214dd013f167b66464b79be937f67":{"kind":"REWORK","_number":2,"created":"2022-07-26 08:27:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/847572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/847572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/847572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/847572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/847572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/847572/2"}}},"commit":{"parents":[{"commit":"0c9912ece163d90eee6e0990dabc8fa59033bc38","subject":"Merge \"Use openstack-ci-mirrors in overcloud and ipa image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c9912ece163d90eee6e0990dabc8fa59033bc38"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-24 11:20:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-07-26 08:27:46.000000000","tz":60},"subject":"Support --kolla-playbook argument","message":"Support --kolla-playbook argument\n\nAdds support for specifying a custom playbook when running Kolla Ansible\ncommands via a \u0027--kolla-playbook\u0027 argument.\n\nChange-Id: Ie4bba0f1bdc82657572e0f06a71db1140c0bd3a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79a8610030b214dd013f167b66464b79be937f67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79a8610030b214dd013f167b66464b79be937f67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
