)]}'
{"id":"openstack%2Fkayobe~847573","triplet_id":"openstack%2Fkayobe~master~I6bb6c379fe149916e4a50e3e32228816d77f5dbd","project":"openstack/kayobe","branch":"master","topic":"kolla-ansible-playbook","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-06-24 13:24:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6bb6c379fe149916e4a50e3e32228816d77f5dbd","subject":"DNM: Test --kolla-playbook","status":"ABANDONED","created":"2022-06-24 11:27:21.000000000","updated":"2022-06-24 13:24:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4bddabe0b128b56e1a221d0da02ebed969bcf7fb","_number":847573,"virtual_id_number":847573,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-24 11:55:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 11:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30e86ced11310bd669c88fea407fddd23794df1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-24 11:27:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c08d6762f397ce1f8121306e7c0dad2ef81b8adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 11:55:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/614f6a788082469b869c1074a9af436b : FAILURE in 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4bddabe0b128b56e1a221d0da02ebed969bcf7fb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-24 13:24:12.000000000","message":"Abandoned\n\nWorked as expected","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"257022348147a9e2a20c7605ded5f3ca6c1eb7c1","revisions":{"257022348147a9e2a20c7605ded5f3ca6c1eb7c1":{"kind":"REWORK","_number":1,"created":"2022-06-24 11:27:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/847573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/847573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/847573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/847573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/847573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/847573/1"}}},"commit":{"parents":[{"commit":"386ba893953b8fa1db99868d16e67609de22f366","subject":"Support --kolla-playbook argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/386ba893953b8fa1db99868d16e67609de22f366"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-24 11:27:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-24 11:27:15.000000000","tz":60},"subject":"DNM: Test --kolla-playbook","message":"DNM: Test --kolla-playbook\n\nChange-Id: I6bb6c379fe149916e4a50e3e32228816d77f5dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/257022348147a9e2a20c7605ded5f3ca6c1eb7c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/257022348147a9e2a20c7605ded5f3ca6c1eb7c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
