)]}'
{"id":"openstack%2Fkayobe~848649","triplet_id":"openstack%2Fkayobe~master~I86c92a1b0b8cf5e4dc534e04709e38bd90f7660e","project":"openstack/kayobe","branch":"master","topic":"rocky-image","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-08-05 19:15:57.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-05 15:27:11.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I86c92a1b0b8cf5e4dc534e04709e38bd90f7660e","subject":"Update to latest Rocky Linux cloud image","status":"MERGED","created":"2022-07-04 16:24:56.000000000","updated":"2022-08-05 19:16:59.000000000","submitted":"2022-08-05 19:15:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848649-rocky-image","meta_rev_id":"b9882f17d7e0f6959ec0d80a2f08e78522120dd9","_number":848649,"virtual_id_number":848649,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-05 19:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-07-29 14:54:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-29 12:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-07-29 14:54:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-29 12:50:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 18:14:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-05 08:05:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-07-19 08:57:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-07-29 14:54:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2216d83e41c3021d46e7ad130e44b5b98d197b28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-04 16:24:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2eb5156a66c0f053eac34d48992ecae3ae15def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:14:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/218a8b330df74c8aab71b4d1c912901b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff94e13788b3456e9b57b0b14e0c35c8 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de75b3228ccf47aea0efc7be17d0049a : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bbc9d8e99574dfba3bdee7a1c526e75 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df64817ec198452bbc9055f7ba0ec1d7 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/386146920ca8407aa6ebeafb97d1b925 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e268613e424bc384412bb7b780dcef : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fc1d63326c64fdaaac779ff80108164 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f889d8b9ff3c45de8270bef7eeb1ce76 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/04d5bc05abff445aab539373475f9771 : SUCCESS in 12m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3cf034158d7c4f75adfed82064d71bf7 : SUCCESS in 15m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1d83a2151ef540ab9ac76f6a98332a34 : SUCCESS in 1h 14m 13s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7d50590f107649abb797452fc743452a : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b557f9572844eeaa1ab4371b2276375 : SUCCESS in 55m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/769b0dad4f92494188fc6f0a9818d5f6 : SUCCESS in 1h 10m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d45f0a8f32244906bf3f1b56adfa95f8 : SUCCESS in 23m 27s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/945d5aa15f2e4f0eaf08cb818f708595 : SUCCESS in 22m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0230b23356a4e0fbd2722778577b474 : SUCCESS in 18m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0a4729f887a145dea4ee172a3ae857aa : SUCCESS in 1h 47m 30s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/856402d328c1400aad1312d89357280f : SUCCESS in 1h 17m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/278a5228733940289327ec34f4cde90c : SUCCESS in 23m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/fcafef8e91264760a9460ddcea353d62 : SUCCESS in 20m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2c7991e26f24c399d6178c5c834401a : SUCCESS in 18m 29s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/89687d91b5cf49c9941b06a18ddd5fff : FAILURE in 38m 11s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bdbfee2166f047efbf7f67364feddcd9 : SUCCESS in 41m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12218f7c0cf045ae94689b9f58403bb8 : SUCCESS in 33m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/26517f41cbca4453a4386acc8af08b2a : SUCCESS in 18m 29s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4de3e2cd3adb487fb7101501ff4f2e8a : SUCCESS in 17m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0994bc092f614887b3ade1118317b2d1 : SUCCESS in 14m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a7e78388294f4c4eaef288bf82bd815c : SUCCESS in 20m 02s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6c708329319a4beba7bd1152eb967465 : SUCCESS in 19m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4c9b630de7e49228b6c71097f741d10 : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d31b1bbe5a35ba886dd68b34672d7497fecdaf84","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-04 20:34:54.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"f2eafbd5e4c3a35e706b06f21fc88ff71ea440f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-07-05 08:05:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e5a30665ebcf12d0b0e4df582498ec42a3e94562","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-19 08:57:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce710f4867cbb3a0f11c3a69cc8d10b27629a12f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-19 10:29:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"352b4353755508c6d88d810187e78c3a216bdd1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-20 12:28:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8f3afee39b9e093919318b0fd3ad8ce554976c7f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-20 12:29:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8393170da52d33b5c033d13f7502bf279c4abaa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 14:00:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31b72f6b7b794a8a94d4a33e7a300bd9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/335c0b94f3334df085c8ce9094a1941c : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a4106d7e854bd5af023bf40918cc4d : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c80cd7a4ffe54fee968783b457b1812d : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccc0c95a193c41429b45fd14d82103ad : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cd470cf237d4d6eba6fe92d9af0a857 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbebc67857743c3a67ba3516aeda0d9 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6da96977b7024300a85d2fb97a497d3f : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c6166c145234850989dfe8e69467125 : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f8871503773e4a0294b63f23723a2b7a : SUCCESS in 13m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f05fa242228e44a6b4717484d1d367d5 : SUCCESS in 19m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/da6ebc8135d6459bb0f09f4d2838243b : SUCCESS in 52m 05s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8397829b837540329bbaf328653b878a : SUCCESS in 1h 08m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2baafdcfd55c427095a8ba8a4ae10be1 : SUCCESS in 45m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a52a7ec08ab4df5a06c095979dcc814 : SUCCESS in 50m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9dd35d50d9d74b918d519ad9c0c23669 : SUCCESS in 25m 58s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b8400913ae934cf28eddb866e3cdc608 : SUCCESS in 23m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/879b9d0e5c334eeda605306eaeeb829a : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d61b2513a9364dc69ec5a871353b647b : SUCCESS in 1h 24m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1af37a09767f42849fe658f55316a927 : SUCCESS in 1h 17m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5eca2b8ccf12456c94b18a890c5b7367 : SUCCESS in 18m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/da10b8b739f2426f9a92c2f23aff57e7 : SUCCESS in 22m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddb7d58e4095465daa1ddb1bfee5cc08 : SUCCESS in 14m 51s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/8837b870cb774473815e38745d73aad4 : FAILURE in 25m 00s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f5c948a88c4a41e9a4b3d02c113a2446 : SUCCESS in 44m 59s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f9f2c8223f34a04bfcee27d3564a01f : SUCCESS in 24m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3f755e850ee7445baa73aaadef4938e6 : SUCCESS in 18m 15s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/436dc133e16646d184653c76ee2cb670 : SUCCESS in 18m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2197c5454684d2a98c2ffee33292223 : SUCCESS in 10m 52s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/62594e98ff41427b855d787a1e4d7d75 : SUCCESS in 17m 20s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3f9bf694a08046a4b443ebfb55015bb9 : SUCCESS in 21m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e06167a76f24cf5a3ac7986c79e6b1c : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":2},{"id":"91b04b700fa2201971daaefc4acc16fa6c3c94fc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-29 12:50:22.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"0b63e9242959356f4b9aa048cafa4b8c56dc0bc3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-29 14:54:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d0f806e669120997850c67a28a61347d0dbaec5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:54:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e4c51891a08b7d0e69adc09ea0b94ed357d8d69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 16:48:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42f30fe4a54542978978433870a3a5b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9562955f0c5243be9112ab22b34765e9 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/726e7c4e6e514942ba5080a0a914ca1f : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36cc05793a7a40e2b5cae57a095c03cc : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa8264267064b809629b364b31230cc : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c30bdefbf927455eabccf77b337c6904 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d65e11540c84e07a49fdef6f19f1d9b : SUCCESS in 7m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66e9ae1735a64d14853d08ef884d30a9 : SUCCESS in 14m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2a570ab4906473c86423757e0d0beb6 : SUCCESS in 19m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2efee615615546f6b75c5810cd88d6f1 : SUCCESS in 1h 07m 24s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/c8239fc1bf5c428ea2538f2493f1c9d4 : NODE_FAILURE in 0s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f98111585d364ddebf0905b91cb54db2 : SUCCESS in 54m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6ddf606ac77743939bbfa0a0b9d39cf2 : SUCCESS in 1h 10m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/41ae55cffac84e18b345e8c0c4edda4c : SUCCESS in 23m 06s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4f7bdd06e6ed4fd7b44a49c16779caaa : NODE_FAILURE in 0s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1450ba640a324679b869c32b36dee312 : SUCCESS in 16m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b973f33ec6e49a7836172e867100280 : SUCCESS in 1h 53m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a6029d5c891475e8ea63651f9553f1e : SUCCESS in 1h 18m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b4087350e91d44f29bea12a2d842e613 : SUCCESS in 21m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4f01409e6bbc4908a05268e6c76a7b8a : NODE_FAILURE in 0s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7933be717cb44e49b65fcbd55a2d47b2 : SUCCESS in 18m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0ee3009d37b04f2b99a102ba97bced89 : SUCCESS in 41m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d80dbe7fd18a45a3b5b2e045ec85bf3e : FAILURE in 30m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9b6a1713f4bd4aa183d54a5800bebc07 : SUCCESS in 19m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ea82c02b46ec40e2825d4eeb20ba00d1 : NODE_FAILURE in 0s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f3ad83bb0924aacaf9fca64c11f6292 : SUCCESS in 13m 35s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a89465e8f40d42fa8ba4f112315ad8dd : SUCCESS in 20m 24s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2d873a727b5f43568a0075e981f95bcd : NODE_FAILURE in 0s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31b81577ba4f451b97a601afa4db5ab7 : SUCCESS in 15m 14s","accounts_in_message":[],"_revision_number":2},{"id":"fa6d8702f98a07dabed0adedda9a72ca34f28360","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 11:09:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d2dc63e4ee07fc373bdfb12a9441d9c336c5060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 12:59:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d64af40f823544b992b3d9c47cb909fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44b4b8dad38443ff9fd5e1f8f4e3399b : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cfa5dae3b6a4254b48c006a862e1f93 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f540d153b3441f88272c0b822a924b1 : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eadf14cfb014856b67fcfb94bdbdd5a : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/218ab302fa704e1dbdd0492a08b955c7 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc40d8cd88247feb5cc1fce73557e88 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ab0013fd6264e91bec1d79c5cfc6ea4 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/514e29ed887b4347b664806cd38e0534 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd394e2fcf894189b3470418450d03d7 : SUCCESS in 13m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e161eeacec794845a6dd1301cdb75bbd : SUCCESS in 16m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f8b815ebebd347f8812855bebc9b4393 : SUCCESS in 1h 01m 44s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8f0356da12254570856b0b3ec3695bc1 : SUCCESS in 1h 17m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eeade7d6a95b4159aae5961ab06a2c44 : SUCCESS in 51m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e68b6d9d2e0845ba8ac47a5d4c91bfe7 : SUCCESS in 1h 10m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b0e96538e20d4623a913446528860295 : SUCCESS in 23m 22s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/159138bc1cd14a3485712260432239a6 : SUCCESS in 23m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92db677fd9fb4672be6393e5a3321e62 : SUCCESS in 18m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/50a4d912882d48aab9f8563d9361a369 : FAILURE in 1h 27m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ac639c75082441b817280ebd7578110 : SUCCESS in 1h 17m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d3d89d8b45634e44aafaa275fd894035 : SUCCESS in 18m 45s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a0b4e55ec0c14833b7c496af9410cfb9 : SUCCESS in 19m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f853856394684696a6b1f696df8d822f : SUCCESS in 18m 35s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/5f638e42b6714b7a844b4916b067c6d8 : FAILURE in 29m 33s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/891ca360002d49d58dcfdd7264082199 : SUCCESS in 42m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/650e98635dac4baba732f53f36d0032a : SUCCESS in 27m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0dd2a2d2e6874bf1b9ec32f2a4bbf919 : SUCCESS in 19m 45s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1537e3cf942348e997657e9d8a173746 : SUCCESS in 18m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/587ec9f310384cb18248411fb33dec98 : SUCCESS in 12m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ee7c732c00dd47598e2f958abf1de821 : SUCCESS in 22m 17s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/06632b09387e46ffae0bc864bf2eab55 : SUCCESS in 20m 24s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a29c7bf548244d4a938e5c7515a2327f : SUCCESS in 13m 12s","accounts_in_message":[],"_revision_number":2},{"id":"fc2e8f7cd873e49c510139c5076844d88dd9cecc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 13:13:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdbaf55932c5f2f27c001199f0c41c4550366caf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 15:22:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59a19b603f2d4ac1bca4161da0f770ce\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d0c70203834114aae904d036a159e5 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d592fcd74300462ebdb0f8a790d62c66 : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20b422aec5c146479292bb94e4eeb33c : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d785df37efe4309a10dc65dbcae0c0a : SUCCESS in 9m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/600d02a55cb744e3949a8771c532ea4d : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a1478d4775244a78a4087c686023d6d : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f2603d30f7c437da2bf9802421fb373 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/09cea47be23941cda39c8b838c0c9e21 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/782f6e44174e4504b70dd1d7348dd588 : SUCCESS in 12m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e2cb4dc26920458d83d73a4e343f7b67 : SUCCESS in 22m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/84fd70900c0246aa979197638374e91f : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/92f3ab95402a48828f021a8301c94b7c : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4191c7fad6524242afb1d597d3b0a4e0 : FAILURE in 47m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3ef06c2120ed44f9a8e003338d3d87cc : SUCCESS in 1h 11m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2ad8a363368547dcba1db6aa0eaf4efc : SUCCESS in 29m 34s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/321fdeca234f4d96a8a5ad8f22d7c5ba : SUCCESS in 22m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/733d33a01c494153bfcac015cc819f40 : SUCCESS in 19m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fbe0c42918444fc091d155d12c2d121a : SUCCESS in 1h 53m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9a2ca6c1a374ef7b016a6204168b074 : SUCCESS in 1h 07m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/12b9bc56d0764dd0b1126f5a8c8744b5 : SUCCESS in 22m 46s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b3f0566416f742fa9d784d6e04ce62b9 : SUCCESS in 21m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f63070bfc3e4e8bb5c4197a03b3fbea : SUCCESS in 14m 01s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/5c6a13b51363453f98599ec6e5d1599a : FAILURE in 29m 54s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/10bb9c6511b0438cb9208870ecf3174d : SUCCESS in 43m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6753996c5f740f8b6b023618d91f5ff : SUCCESS in 32m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0ab3498831844776afa872032a4a6ac2 : SUCCESS in 20m 09s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/06b9d22d4deb419c9f659f510f70601e : SUCCESS in 18m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf6e8e1dce6841d68118dbeebb2b0f5d : SUCCESS in 15m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fdaa71d43e6c4359a78c2e1eeae30d0d : SUCCESS in 21m 06s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e82a6929120f430497346c7dc301b8ba : SUCCESS in 20m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aa33d46a86740b48725bdb3dce85f8f : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c145347d9f6bcedcd9c5a212725151aeb7d4a112","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 15:27:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f1b05a66dcd6b2bfca37b9b5693901cf4c59a116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 17:18:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627520ad134f48cab36a59b8c9a9495b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/951cc2e1efac4c94bf2bc49c28315cd7 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308f4c0e7c734a27b90425bcc35e59f7 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91028da106c1411e8241eccdeaa31b31 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a933134889df4ed4b02c9b65eb9c9204 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30e4034348b7483a98672247a0bce78f : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce36372ba59443fb3241e6cb551742e : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/000dbff49fb64c5f8f1c18017f268381 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c1d38a65af548eabd772bc5a6d91228 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e4b7f0505ed475caff8d3ec43011e0a : SUCCESS in 14m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/87ec6bef043644c9a9f817149bf2c502 : SUCCESS in 16m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9fd1f15e72c04b33b26b4e79bbd92335 : SUCCESS in 1h 04m 57s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/c61ff6ead0d34173983695c8138e63ff : SUCCESS in 1h 12m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/930257d6088e4c45be6199e6197dc7b8 : SUCCESS in 49m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/edaec052d4a44550bbcd84e27d6c6bbb : SUCCESS in 1h 15m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d3085d7c1c344d4cb18603d8ea106a28 : SUCCESS in 20m 41s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/d8e694a403b9495face02bbef2976dd7 : SUCCESS in 21m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7a15a4146264fa49bc745d30f082361 : SUCCESS in 17m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/50f9fbfc9b5f4e609a8777c11a57c592 : SUCCESS in 1h 46m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96beafc771564808ae04a2869d4f315d : SUCCESS in 1h 24m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/797ab803d4e348128aa1316c98cdc45a : SUCCESS in 19m 07s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/efd4b473b21e4566a7a43ac694b9b34f : SUCCESS in 21m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b81856d70b242a0a58a3cdddeb6911d : SUCCESS in 16m 52s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/8e2eb210b8da44c0982191eca8bcb230 : FAILURE in 28m 30s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/af46aef9d2024824a5b394a85b6024bd : SUCCESS in 35m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a26109649aaf430d82fdd7eba36d516e : SUCCESS in 31m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2586be89cf2c4a1f90481f94dc5f7c02 : SUCCESS in 16m 23s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5bcbc53ed9ba40889cc304d0b7e4967b : SUCCESS in 16m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26f6cf79bbe5407a90c7e201e9f3e930 : SUCCESS in 13m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/928bde0521b1418a94bbca86d645d4cc : SUCCESS in 19m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4cb37be06f3a46f6aed80b9b8828bc08 : SUCCESS in 17m 07s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1729886b5dc419fb1edcf481eb7c11e : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":2},{"id":"61696e40967e66fc930d837824e4d7e0760e0123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 17:18:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf1930e24dc99c07873e42df7054db4d22f32f7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 19:15:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a3a123316124ab0b71a5f53820c3664\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0657afc1f1444cf694b8d516f2770c36 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db8966aa816a43b3a12f087ebbe43abd : SUCCESS in 9m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08627c4f40e145c48f03593cbf8dddde : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d9225c97ad471d9890e6ce0c1bd389 : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2d74a02cd9a432f9786f0c0621ade62 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ade164fe54cb436aabb5ddc9ea58fdee : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fba8f5998e174191b45f5ce9cc21a5e0 : SUCCESS in 14m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a8ca8001c1364eca8024fadda9cfe851 : SUCCESS in 16m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bd823f37e0cc40aa971859a4e741fadb : SUCCESS in 1h 15m 14s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/0cf870ad73be485b8696f8d88393f221 : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ebd9fdee33c49a3b080d70bd728ca9d : SUCCESS in 58m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/53957db5ca0643468841bd2207da613a : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/37302b47391a4b03ae1176870861976f : SUCCESS in 24m 33s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/29caa851fe0a4b1592207f292b4214dc : SUCCESS in 21m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e93565b68a84515bc3055a12bb72cc9 : SUCCESS in 20m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9ef94488dade43b08b96529ce87cfee2 : SUCCESS in 1h 57m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f2c812ca189467d9bbe4f42bcf5dda0 : SUCCESS in 1h 16m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3b0edd6c9626477db17b2e0c1d14bd5d : SUCCESS in 23m 11s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e39c7dc69d4b40aca947cb3f8834bcab : SUCCESS in 20m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97ef6733b63e44efbf207e91aee3ad42 : SUCCESS in 21m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/986e0ab429314a449ab7587b87aca84a : SUCCESS in 44m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47b17abc784443a3bce43725541e1285 : SUCCESS in 36m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ed1f5f3ad9e44039b0ac32f21b840337 : SUCCESS in 17m 43s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6a015b7a7bd4491b8c97646fe8c02a0e : SUCCESS in 18m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff343eca820b46e2b5b7db5fcd898678 : SUCCESS in 15m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3f05391e26bc49f3881f894856e9d71c : SUCCESS in 23m 04s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0e33296b3321497fba479538f580a0de : SUCCESS in 18m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a092d76f305b4a9b9a6c9a0d3ffe0f8f : SUCCESS in 17m 55s","accounts_in_message":[],"_revision_number":2},{"id":"ffde01583082a48b7dcddc0c9c4ddf09aa15b3f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 19:15:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b9882f17d7e0f6959ec0d80a2f08e78522120dd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 19:16:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b3a74be2024452b586ed7657f881f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6d2713c877488f9cb455ced1659d3a : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bc1cddca6be40bbb218067814c5e6e3 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0030c5f1ed5aeac7d991187b1761e0f664c3669a","revisions":{"614cbf97c738fa250dd52593e4c3825e7e8518ce":{"kind":"REWORK","_number":1,"created":"2022-07-04 16:24:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/848649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/848649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/848649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/848649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/848649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/848649/1"}}},"commit":{"parents":[{"commit":"836f394a6dc8a20197e13d9efce212d1efa93039","subject":"Run selinux playbook on seed hypervisor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/836f394a6dc8a20197e13d9efce212d1efa93039"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-04 16:24:37.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-04 16:24:37.000000000","tz":120},"subject":"Update to latest Rocky Linux cloud image","message":"Update to latest Rocky Linux cloud image\n\nThe 8.5 image has moved to the Rocky Linux Vault.\n\nChange-Id: I86c92a1b0b8cf5e4dc534e04709e38bd90f7660e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/614cbf97c738fa250dd52593e4c3825e7e8518ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/614cbf97c738fa250dd52593e4c3825e7e8518ce"}]},"branch":"refs/heads/master"},"0030c5f1ed5aeac7d991187b1761e0f664c3669a":{"kind":"REWORK","_number":2,"created":"2022-07-20 12:28:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/848649/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/848649/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/848649/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/848649/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/848649/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/848649/2"}}},"commit":{"parents":[{"commit":"836f394a6dc8a20197e13d9efce212d1efa93039","subject":"Run selinux playbook on seed hypervisor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/836f394a6dc8a20197e13d9efce212d1efa93039"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-04 16:24:37.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-20 12:28:46.000000000","tz":0},"subject":"Update to latest Rocky Linux cloud image","message":"Update to latest Rocky Linux cloud image\n\nThe 8.5 image has moved to the Rocky Linux Vault.\n\nChange-Id: I86c92a1b0b8cf5e4dc534e04709e38bd90f7660e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0030c5f1ed5aeac7d991187b1761e0f664c3669a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0030c5f1ed5aeac7d991187b1761e0f664c3669a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[]}
