)]}'
{"id":"openstack%2Fkayobe~850337","triplet_id":"openstack%2Fkayobe~master~I3a95382cf6e761c5c0e5c7a47d6477b3cdfa8c19","project":"openstack/kayobe","branch":"master","topic":"seed-containers-names","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-26 14:26:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I3a95382cf6e761c5c0e5c7a47d6477b3cdfa8c19","subject":"Remove unused parameter from seed container example","status":"MERGED","created":"2022-07-19 12:03:30.000000000","updated":"2022-08-26 16:20:49.000000000","submitted":"2022-08-26 16:19:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850337-seed-containers-names","meta_rev_id":"7343495a18c75dced74ab845413b0207a80db2c3","_number":850337,"virtual_id_number":850337,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 16:19:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-08-26 14:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-08-26 14:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-08-26 14:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 13:32:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 14:22:43.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-08-26 14:25:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"83ff26ede4ec95fc19ab73e44496a01865609b35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-19 12:03:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c4d5b1e070645e92b960af32e0c12e2d30e6d07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 13:32:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea5623ddaca9489cbb10d46d28e9ff90\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3588b50fa2b445ee80aee212e3eb2cda : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e33b859a894adea6246e2f0db794a8 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/423643744d0a4c3f99662581016da958 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14ebe5bf5e0c47aab9f86fdee6f64b7a : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7146817416e44179b6ee4732ab6de621 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/807e10b89f724fd58b8825ebb6b556f1 : SUCCESS in 13m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce1fab06d71741e0ab2bff5ec7e88710 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d007c2d25e9348cfb68ffa619077b347 : SUCCESS in 16m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6db676f691fd461986fdf4f10e8ee7d9 : SUCCESS in 15m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1a98075b9f1a4e908645e44d61d91795 : SUCCESS in 1h 10m 37s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a6379eb33ebb4782b81b480b8f403912 : SUCCESS in 1h 10m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9219d1ed0e9f4620a8997cee79bd4788 : SUCCESS in 45m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/367846693d214496885319c7f1755482 : SUCCESS in 1h 11m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/56269c94f6cd433989c8eb5a0e9e7e10 : SUCCESS in 24m 58s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a370e42f1bb04135952469dd7764710e : SUCCESS in 24m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77cfd4703798482cb362c9108cbe4f1a : SUCCESS in 16m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0db100ad60e84ea291faf4b56b3d9ce2 : SUCCESS in 1h 27m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/205474818a20406db49955efb385e082 : SUCCESS in 1h 18m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/82edc2459e7141979a489c3a8084ada9 : SUCCESS in 17m 14s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a37ff464db6a4194a374938efa0cd4f3 : SUCCESS in 23m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71caf60807654d429ef895bef009c3cd : SUCCESS in 15m 22s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1c10479461784bda8ebeb39a28015459 : FAILURE in 32m 14s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c4e7f4083c0d4979864c4a68e6569353 : SUCCESS in 33m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fa5a5a07c784b13b4a62d64c89a1dae : SUCCESS in 32m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/772d7fe672f34bac9db01bb0b222bdb4 : SUCCESS in 15m 49s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7e12e82306364792a6ce9b888f2c7733 : SUCCESS in 18m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4225f51beb654a28a0a9a8b12edd62fc : SUCCESS in 16m 09s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5bad757862ae43bb8ed25b3f02257726 : SUCCESS in 16m 21s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f9cbae98067944399b991b187d0265e0 : SUCCESS in 20m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cca8264fb6e04046a852c402669926ee : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":1},{"id":"62ea50a21e5f65ed8153ec164f9d7639a070f311","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-26 14:22:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"404e3be1a004be99e9e3bcae58e55846003c7e16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 14:23:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21ee4f93f8777e226da319bc1dc588404603e49d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-26 14:25:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cc4a4a73cec525a1c5376977be8bb71c73dd8ae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-26 14:26:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db461e528c92bfd89793ca65f338f6680451ac28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:19:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2abdcc7faaa04066b8a38d95ee626a63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6845029193a94b2ba5c59ff29827e3a0 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da596acf178440b6874a682b57832c7e : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90d6bad574464c0685daae7d22a20925 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42d225012684dc38aa0182c12fc49da : SUCCESS in 9m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4db9fc0e7369470bbb260c66530961ae : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2232607227a344519119a7ba04def274 : SUCCESS in 16m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c4277fa55f84f288ab06b160dcb42ac : SUCCESS in 16m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4d5bd2d4eedb4b2e940ff06354a67881 : SUCCESS in 59m 23s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/60c7550e44354de18dd692d619e27bbf : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b2d09960920460b8eaff692b98833c1 : SUCCESS in 56m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3b0b09a133874f5fa5a3c6b4d3c6b7e2 : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fba40a66e09a4674aed38127e5e86b24 : SUCCESS in 24m 33s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2e22a6473ae6498bbde57a182c607a17 : SUCCESS in 23m 51s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94fc45f5dc584cee9b094a8968ba8a39 : SUCCESS in 16m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9572224bdcdd4253ade740e243f5a6cb : SUCCESS in 1h 54m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2c7580401664ea1b8e470b6112500b8 : SUCCESS in 1h 08m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5d97d9f22c3c4768adea1e0bd401db5e : SUCCESS in 21m 53s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2b61b970d4a7494fac0edd6f0c4f440b : SUCCESS in 22m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/904856f0fdd145da9cb5425081d60bd8 : SUCCESS in 20m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2674ce8857a24c6e87f537f9f2c8c2fe : SUCCESS in 41m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1980b9bc6fb04f44809f22da1a46de27 : SUCCESS in 26m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/506adf1ba29f4c33a994ad93de55cd2d : SUCCESS in 17m 32s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/20c994e6c0334da0b745aeef670d8b92 : SUCCESS in 21m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b20b3fa1c5643a3bfd500901f1e88b0 : SUCCESS in 14m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bbcfa3481c664d649a1eed463e6234dc : SUCCESS in 17m 25s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0c2d0f67ea2145718fa408e57e67a1dc : SUCCESS in 19m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c34d2c23a02462087f083d96f1ee691 : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"043b0c044d39ff077ae2c21d510db1b59db94dba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:19:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7343495a18c75dced74ab845413b0207a80db2c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:20:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9300f682869546a7ba3c9a91ec2eb8d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a42d3be0391438291cf93e5b4975dbe : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9d00a2073cf2f29ecb7abff6ca1cc6a59947348","revisions":{"a9d00a2073cf2f29ecb7abff6ca1cc6a59947348":{"kind":"REWORK","_number":1,"created":"2022-07-19 12:03:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/37/850337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/850337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/850337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/850337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/850337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/850337/1"}}},"commit":{"parents":[{"commit":"17d49908af107f444cee1c1eeee3cca9ecf1825b","subject":"Add missing dot in hook name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17d49908af107f444cee1c1eeee3cca9ecf1825b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-19 12:02:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-19 12:02:56.000000000","tz":120},"subject":"Remove unused parameter from seed container example","message":"Remove unused parameter from seed container example\n\nThe keys in the seed_containers dictionary are used as container names.\n\nChange-Id: I3a95382cf6e761c5c0e5c7a47d6477b3cdfa8c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9d00a2073cf2f29ecb7abff6ca1cc6a59947348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9d00a2073cf2f29ecb7abff6ca1cc6a59947348"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
