)]}'
{"id":"openstack%2Fkayobe~850608","triplet_id":"openstack%2Fkayobe~master~I7e10b851016253bf85625e9daa89416c738242e8","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-08-05 14:01:19.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-08-05 12:24:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-07-21 08:35:43.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I7e10b851016253bf85625e9daa89416c738242e8","subject":"ovn: add network group to neutron-ovn-metadata-agent","status":"MERGED","created":"2022-07-21 06:18:48.000000000","updated":"2022-08-05 14:03:19.000000000","submitted":"2022-08-05 14:01:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850608","meta_rev_id":"a20eac1fb17832baac8dc54960eb2bdd21fad55f","_number":850608,"virtual_id_number":850608,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-05 14:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-08-05 12:24:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-21 08:35:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-08-05 12:24:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-21 08:35:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-21 06:19:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-07-21 06:19:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-07-21 08:13:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f2ef82bbc9c034077fc5f0b8f08b8ee2bee9ae9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-21 06:18:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18c83c46368489b0255b37c8c43da95c59a2c83c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:13:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9833dad669734cb4a187d3f5dca7768c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c54b6484465472dab542722dc93e392 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/045224e8060e42c7bc6d5c5884ca6464 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7229d266f5b6443d872dedb64c7ebeed : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82c46dfd75194a069554f5e78f272b5c : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b0c6a29d0134b639de7b3c969209cd7 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31a71578c3743ec80678f95341c4f49 : SUCCESS in 8m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d868a30a1326461da3f79e86a70333f9 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/183121dfcf1648a1b3cf07a0d60a2c70 : SUCCESS in 12m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/64474f3291d84f898b3496d91c8cf08b : SUCCESS in 15m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5ebf73e2868346a39e3e59777918f854 : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4237eff144324434959167f09b0a0538 : SUCCESS in 1h 10m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8072f2efe736469c8046edd685083acb : SUCCESS in 47m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b892c71728a1464ea2f932b88dbe646d : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/221507e65df14eab8b774f9c0b12b439 : SUCCESS in 21m 41s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/8250bd2a62aa418294c504dd1aeec73e : SUCCESS in 25m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d5286e43840441b9208709425b4ebe1 : SUCCESS in 16m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/713e0c56b3a54b13bce9ca4ed0905256 : SUCCESS in 1h 54m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f23addfb9874cb4bb5d1b7105dd67cf : SUCCESS in 1h 15m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4f6eec9a3cd54a47a0cda243b4619554 : SUCCESS in 24m 05s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e93484444cf1428298e53e36d7ee5e7d : SUCCESS in 21m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83bb3839edb443d396cf8c539f705579 : SUCCESS in 17m 12s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/c520e76127a0496db39bc31844b05a83 : FAILURE in 33m 29s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0d34b10a05874934bbaabc1fc90a96d5 : SUCCESS in 46m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df7231f45ce84018a4adb1b46d18ec2f : SUCCESS in 33m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dfb8489ae489405789f82b2fe8f5963b : SUCCESS in 21m 17s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7c2e3f44d8ac4ac7b91e6b9f0b334a17 : SUCCESS in 16m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7956e6225d734ca3b73cbba7b9b9db04 : SUCCESS in 15m 20s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ad122f40bb8d44bc978f2717cdda5be4 : SUCCESS in 24m 31s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b47e9297b6de4223a04095b67962ce16 : SUCCESS in 19m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d21230f29e2c4beea20babcb934bc65b : SUCCESS in 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d6a659f4ba6d0279c65bfb856dcb01e61782a699","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-21 08:35:43.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9c03f9a83c37e5565116b672c9cf802f0f8e18f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-05 12:24:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9f140fa33e59a2dae59ee85d64692c75c744373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 12:24:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9887b3b02a0b103c1c26f281ff0c8389000ee9b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 14:01:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c11cc6ae6a504ea3bc62786547817e63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f379c8aa2224bcd8ac65875ce97e15c : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a84cd200137f4b04b69391899f6baeef : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/806b8237373b42ae9c94e88388af37f0 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26825a6fad342dba53984a65ae94879 : SUCCESS in 8m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/39af9cb95ba14e4997dda505bed3b212 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/035679778fbb43ea98975b600e831780 : SUCCESS in 13m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0cdb5116d3f4f79bc7b829c6f2f9465 : SUCCESS in 15m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/10afee6cf69146629d653cb58d95d79e : SUCCESS in 59m 08s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/101215243acf4e9cb5e8edabaeed3ec9 : SUCCESS in 1h 11m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93197274b11844b88c9ceb71d45e8f94 : SUCCESS in 48m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c004425f244b443d90a22ce009b4009c : SUCCESS in 58m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/15b3fce9dfbc44d7a4d278e4e27eb683 : SUCCESS in 20m 03s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/fda4b831ca1c4925a75b2d04419dfd7e : SUCCESS in 20m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a88671a1d86b4e2c9a78551e85da0066 : SUCCESS in 12m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d235e3fa7a034807814ca6cf310430e3 : SUCCESS in 1h 32m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd2af334cb22496eb64be085416aa09c : SUCCESS in 1h 02m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/aaef0749d41c4e2b9988623cb8eb0bc1 : SUCCESS in 19m 54s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/28f7ebab6cc943a4bb88ce588036991a : SUCCESS in 18m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e36c28c3de9340c8a98a85e006ab7ddc : SUCCESS in 14m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc237f369d324015ad4d8e35d53f982e : SUCCESS in 35m 59s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/115ed85c1f564d69a0eaf9d5df3cf4a4 : SUCCESS in 25m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/81a13380c67a4c268d3ab0149535e3fa : SUCCESS in 14m 04s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0adee07380a14f24b1577f323886c568 : SUCCESS in 16m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66e57498ac254484a151021c78759232 : SUCCESS in 11m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3bdf0f4b6446456c9cbae9707f7df919 : SUCCESS in 15m 53s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/848749da66d24385840c162c506047f6 : SUCCESS in 18m 58s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5b293f7af0c4214b9273abd072523c0 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3300334d9474db66f1de3b58c04b8c8e242d7756","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 14:01:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a20eac1fb17832baac8dc54960eb2bdd21fad55f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 14:03:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/574b637c07c74755b3d977a36caf8f83\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a680a25a2f4675a1d33968af47d8b9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ee99c59481e4fb31119a418248fb4823d6a63ae","revisions":{"5ee99c59481e4fb31119a418248fb4823d6a63ae":{"kind":"REWORK","_number":1,"created":"2022-07-21 06:18:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/850608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/850608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/850608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/850608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/850608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/850608/1"}}},"commit":{"parents":[{"commit":"17d49908af107f444cee1c1eeee3cca9ecf1825b","subject":"Add missing dot in hook name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17d49908af107f444cee1c1eeee3cca9ecf1825b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-21 06:18:00.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-21 06:18:45.000000000","tz":120},"subject":"ovn: add network group to neutron-ovn-metadata-agent","message":"ovn: add network group to neutron-ovn-metadata-agent\n\nFollow up after I111df79f08eb76a735f33c6edf03b5a302fb6cc3 in kolla-ansible\n\nChange-Id: I7e10b851016253bf85625e9daa89416c738242e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee99c59481e4fb31119a418248fb4823d6a63ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee99c59481e4fb31119a418248fb4823d6a63ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
