)]}'
{"id":"openstack%2Fkayobe~851834","triplet_id":"openstack%2Fkayobe~master~I9c07e87ee43b3b56c3047ba983c42201dc659c82","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-17 13:59:23.000000000","reason":"Change was abandoned"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2025-11-17 13:59:23.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-17 13:59:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9c07e87ee43b3b56c3047ba983c42201dc659c82","subject":"Use Inspector database to store inspection data","status":"ABANDONED","created":"2022-08-02 09:41:54.000000000","updated":"2025-11-17 13:59:23.000000000","total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"8886a6ff082c12e8a9173a13cce841bff664f9e1","_number":851834,"virtual_id_number":851834,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-21 23:15:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2024-11-21 15:01:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 10:57:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 16:58:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-21 12:48:30.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-11-21 14:49:31.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"1d483c5fefd79faea42aeabb1d111ed728b0a781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-02 09:41:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ebebf504833d05e12e3a6ee8939d76ca8b1585c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-02 09:49:37.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"423de978a2eb2a89dad2c4b3d2bc40bebf46041a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 10:57:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c12209820b64e14b1fc5e9b09a00284\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4eda2850534462b8ebd14bc4fb2c13c : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e49c1bb0c97430fad757a9faf9aac00 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3ce7bfa4f5249108ce2e28bd1b30b76 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf00537b98b6440dbe80aecc7923a8dc : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5987e8e50144f849d58128ae9ffa96a : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a4861e2a30496f9971ca24386babeb : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf4f210cd10b4b048ac8f04a47c5dc60 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e7b2da0c365b42a2bbfb49d91a87ab23 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7648ad3dc75e4a1c8d2f9d7c1931a42b : SUCCESS in 13m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/59a17fff97b64b9c8d077240c90d3acf : FAILURE in 12m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ab8a684c6e824f778c05da929f08d7ae : FAILURE in 24m 46s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/edde9bf20fc5483294ccb6607afb2c40 : NODE_FAILURE in 0s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52d8dabf0a9d46ec8237eb9428354e7f : FAILURE in 18m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c914999aaeb1495f90a2ec093ff893aa : FAILURE in 24m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6dc262016f4e4cc6b02723c04b0aa168 : SUCCESS in 24m 19s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b31ddce7c91a4614b6a339941ba10f23 : NODE_FAILURE in 0s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45481a0c0a8845fd8ce7445b3aa6c91f : SUCCESS in 15m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/25164637f4f14311ad429e4f4cbd6454 : FAILURE in 1h 13m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/427a22e10d374f5487058b4f03fc247c : FAILURE in 55m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/79a23ea37f2640a5b6d8568e1f66feb0 : SUCCESS in 20m 29s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/273f1e961338466ca8078eec2ba66049 : NODE_FAILURE in 0s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fd0ab66675e430292651df3135ab1de : SUCCESS in 19m 53s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/8cb22c4629b543d09c3c972ed770839c : FAILURE in 31m 41s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2a946b630d804be3911dfe3130f1335f : SUCCESS in 43m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c628f4befbd4746b64e307aa0ebb839 : SUCCESS in 31m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/426dc4f9ff2149bc80a2e0d44cba0ecc : SUCCESS in 18m 21s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/44eb7c3ef44f4e9f9e3735e9673f12b2 : NODE_FAILURE in 0s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cc11b6ceaa24738806717374e33c20c : SUCCESS in 13m 55s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e9bc8b5c7d86404ab9dc3ac94b1f3d45 : SUCCESS in 21m 30s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f9cba01fef9b4c5186d07508a8f51121 : NODE_FAILURE in 0s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7ba65600cec40c7899dee8def7abb26 : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"96cc5a7b4755b0be56b2ffa740c4031e2af14efd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-02 11:05:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"908b23d93e91b57d791faa36c80a07c648b7585a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 12:53:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/560b9055d7c84a98af4ed008cc3747cc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/655c9e7e2099485d911cb66c5e56360a : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c84bfa0a7c44b78ee45a4d33138f28 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06698b307f7c4b1c989c43d30079d688 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65258a73d54644b19abeeb2ec0e54ea9 : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db0482f45d7243acb22b38f1438e073f : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2098ed893eb4e09896ca6c19c9d41a5 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34c4d1b598cc49679b897b9d346be07c : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ce2f1c33aad4f04b8dc4668ea8fae48 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4315c42ab089429a881860331b6a8c70 : SUCCESS in 13m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/85695e5a47684692b3368bab56e858e8 : SUCCESS in 19m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6c8c704af095400b8407a91e12122fbd : SUCCESS in 1h 08m 42s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7be8a8850d96413194c4c108619902e1 : NODE_FAILURE in 0s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/064b354ba5894caca9a3f85643340ed5 : SUCCESS in 48m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c88efad54f8740a28dc26efa7dd3eeac : SUCCESS in 56m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cdd6320606914e698b5607419547a34c : SUCCESS in 23m 53s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a12d971e4bc84720b3f9f0aefa89a189 : NODE_FAILURE in 0s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99c90918db28456abb95f39c144f9f09 : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5788ed287a4e45ebb0c1c90534d0e98c : SUCCESS in 1h 42m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b024ecbe7ee8468d8d26e4a81bff87fc : SUCCESS in 1h 20m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9793753c3b5744ae90d658b1ef77f8b6 : SUCCESS in 22m 27s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f60ccef450e04bb48c5dd3225b9f38b5 : NODE_FAILURE in 0s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b5b69981fa147ad893513b41820648a : SUCCESS in 14m 55s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/98a71db75441495bbb40e576cfcf24a8 : FAILURE in 30m 15s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/12d8f6c5a41a4a6b923aad64963a2388 : SUCCESS in 36m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92b4a01f33e6421d9da5e8de041f1913 : SUCCESS in 32m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6381536015ed416f8f76835538faee50 : SUCCESS in 14m 25s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9f02271902194be7b5dff90d4c3f02af : NODE_FAILURE in 0s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6ee6466d52e4ebca794930e9741eb21 : SUCCESS in 11m 15s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b125fd8bffd44e8290b623410645f10f : SUCCESS in 15m 57s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/147651d74f4b4ef68e8d54b68beb2a99 : NODE_FAILURE in 0s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e08da9ad3e2443f580e42de197925ce0 : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b3f4350836ce5af26734fa2ed0141f894f37f1df","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-20 16:58:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e7799ad20b6697ec565657cc5578938675577087","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-21 20:39:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0b27a959a2dc9c77571f9ddd801b19cf9bd544de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 23:15:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e536a514056a42799759ebbe8c08a707\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/428ff95a39e44ebea4b13c04fa7e91c4 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce2b484c10c48a8a3ed81036899d831 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a493c0e701942b2864afbc49dcb0550 : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cca6499dfc024a2ea25391fa85372fca : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/54200104be094a4da59c84f20518f793 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1ac6afb5f6b845d19c6f96132af2cae8 : SUCCESS in 8m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/70c63ec35cf54ce2b9679a88a1305221 : SUCCESS in 1h 06m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3448832444764b549e3454830b58958b : SUCCESS in 51m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ad019ca0a22453fbf67beee44ff5b09 : SUCCESS in 1h 02m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/94f4a05c445946609cc500895b0df28f : SUCCESS in 1h 26m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b2b64f5d3f5b47cbb3ea2b71ead9512e : SUCCESS in 29m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/12627f1ae0d5499caa3b19a5433355ee : SUCCESS in 16m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25bd4f539c9942f0992fb80dd0bb87d8 : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0d4bd4a042ec46ae84ee5a2555edd5bb : SUCCESS in 2h 27m 59s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/859fcd1a50d14a7bb1cf440d4f1e5144 : SUCCESS in 2h 28m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c2fd5b2c9a1c4d64b32395e8576fc83f : SUCCESS in 17m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e0cb31b767143fd845ab9e4a99c689e : SUCCESS in 14m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8d3ea1dd066b498582dc40f483f40571 : FAILURE in 27m 20s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/90e6864c85994cbb809e3adfa6eac7b5 : SUCCESS in 26m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/622116d9879e4871829179a76d528c89 : SUCCESS in 24m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c0878f29041c4da8a31d410eb91d357e : SUCCESS in 10m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66fdce17413747ee8d21e63be06939eb : SUCCESS in 9m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eb74f0a63c8f4d7aacdeb9a00bf54b27 : SUCCESS in 15m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44421d028b634b2ea1cc51acdcc7761d : SUCCESS in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0445c85fb2c24155bd1300a8f3f0a9de : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11a60c5c5a6744078b142f91aaa45201 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a496952a23f24b3c85ac1c65a7298c7c : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":3},{"id":"16c0e19527ddc4d38a3ee43f58bb08899443b6b1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-03 13:13:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32c2e92300f3816000942fba9338a0462277d370","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-20 15:41:09.000000000","message":"Abandoned\n\nInspector has been merged into ironic","accounts_in_message":[],"_revision_number":3},{"id":"13712b293cea189f680faadcd39f198b6542d0a7","tag":"autogenerated:gerrit:restore","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-21 14:46:50.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"723690666f026053100be8462f72e422fd9fae89","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-21 14:49:31.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"cdb9a3f956ef1b56eaa2142912efdc4febc66357","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-21 14:49:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"51bee4587380b19b1a5bcd83aa95524ca146d965","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-21 14:55:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6be588ebb7cc018e2c91509008c658a17cc98ebc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-21 15:01:47.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9616134702d29c58dcf07ff859c7b98f7b3aea93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 17:01:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de5bc86d6724aa1b85234f36d050bf2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/406f2249795d483fb0e78f35114ec48e : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915941ed9d48422eb45dbdea0a43b6a1 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34237b0d2c314477a214ba158a493681 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d05785d7533479d8c4fa611f94e680b : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1368fa6c80954d5a979e63f20c766593 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/27616e26840f4a9dad017f9369e995d7 : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b33ea1f988724b709ea187f7f2cd4879 : SUCCESS in 12m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3657c847536f4d54ba1ceceeeffa2866 : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3d5245fa1ebd4aa1b39ba5df43ec0fcc : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1d7b2c9eb65b4cc79c559726a892d124 : SUCCESS in 1h 01m 22s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/293d19757d80481a97fdc3c1b931ef3c : SUCCESS in 55m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b53111753a524350821456961c4a2885 : SUCCESS in 1h 12m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/022cdde3f1e44ef996fac910fe6bece3 : SUCCESS in 25m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/40808b6e8e1e4599a40c47c03085ae6d : SUCCESS in 17m 28s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/131eb1ac47c3416baf8282b6fdcc42b9 : SUCCESS in 14m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c61798709c824b9fbbca9c074a403cd8 : SUCCESS in 2h 06m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfccf3db09924237b72e426ce8b374f1 : SUCCESS in 1h 40m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/94dd9b5ef22549e89e330c12dff09792 : SUCCESS in 12m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42c3a335c713424a9528fae9b17c8e8a : SUCCESS in 19m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1025c4f9fbad45a0b7754e5383bd8645 : SUCCESS in 50m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1c8197f2f0e9432598b3f420403b10b8 : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9786ae77215c4fafa8d85b00657fb9be : SUCCESS in 30m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/05ad1c82d0b94a2c8910eee5084f5eed : SUCCESS in 12m 27s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1af5ca2e10b2424bafb8aed0d5affa2c : SUCCESS in 18m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0dd189f253540e594d92412db6157e4 : SUCCESS in 13m 03s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/96d40d23b6f44a2a8601df2497671de1 : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8d4b59e2124e4abcae65ffee6b255358 : SUCCESS in 22m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e0fb5cfa0574dc6a270ab1406e88c1f : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":3},{"id":"6b7d0a19754a118fc430281e8cda8cb92c1bed54","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-22 17:44:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8886a6ff082c12e8a9173a13cce841bff664f9e1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-17 13:59:23.000000000","message":"Abandoned\n\nDone in https://review.opendev.org/c/openstack/kayobe/+/959849","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"413c7ba49e421b495c38de677a4689161014adae","revisions":{"235c2e757cb3a994de66b36960e5dd1b2e0caf88":{"kind":"REWORK","_number":1,"created":"2022-08-02 09:41:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/851834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/851834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/851834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/851834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/851834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/851834/1"}}},"commit":{"parents":[{"commit":"17d49908af107f444cee1c1eeee3cca9ecf1825b","subject":"Add missing dot in hook name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17d49908af107f444cee1c1eeee3cca9ecf1825b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-02 09:19:37.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-02 09:41:35.000000000","tz":60},"subject":"Use Inspector database to store inspection data","message":"Use Inspector database to store inspection data\n\nIronic inspector supports different backends for storing introspection\ndata. Originally, the database was not an option, so Kayobe provided an\ninspection_store nginx container that poses as a Swift object store.\nIt is now possible to store the data in the Inspector database, and\nthere is even a tool to migrate the data between stores. The database\noption is simpler, and avoids running the nginx container.\n\nThis change switches to use the database store by default. The release\nnote provides information on how to migrate.\n\nStory: 2010205\nTask: 45929\n\nChange-Id: I9c07e87ee43b3b56c3047ba983c42201dc659c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/235c2e757cb3a994de66b36960e5dd1b2e0caf88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/235c2e757cb3a994de66b36960e5dd1b2e0caf88"}]},"branch":"refs/heads/master"},"4acb2a59714e7367dd79d00e4b09378629930d54":{"kind":"REWORK","_number":2,"created":"2022-08-02 11:05:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/851834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/851834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/851834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/851834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/851834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/851834/2"}}},"commit":{"parents":[{"commit":"17d49908af107f444cee1c1eeee3cca9ecf1825b","subject":"Add missing dot in hook name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17d49908af107f444cee1c1eeee3cca9ecf1825b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-02 09:19:37.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-02 11:05:01.000000000","tz":0},"subject":"Use Inspector database to store inspection data","message":"Use Inspector database to store inspection data\n\nIronic inspector supports different backends for storing introspection\ndata. Originally, the database was not an option, so Kayobe provided an\ninspection_store nginx container that poses as a Swift object store.\nIt is now possible to store the data in the Inspector database, and\nthere is even a tool to migrate the data between stores. The database\noption is simpler, and avoids running the nginx container.\n\nThis change switches to use the database store by default. The release\nnote provides information on how to migrate.\n\nStory: 2010205\nTask: 45929\n\nChange-Id: I9c07e87ee43b3b56c3047ba983c42201dc659c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4acb2a59714e7367dd79d00e4b09378629930d54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4acb2a59714e7367dd79d00e4b09378629930d54"}]},"branch":"refs/heads/master"},"413c7ba49e421b495c38de677a4689161014adae":{"kind":"REWORK","_number":3,"created":"2023-11-21 20:39:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/851834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/851834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/851834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/851834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/851834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/851834/3"}}},"commit":{"parents":[{"commit":"0cb0e9feef96a8466a25302bee31a5577a531022","subject":"Merge \"provision-net: Remove transitional unshare code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cb0e9feef96a8466a25302bee31a5577a531022"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-02 09:19:37.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-21 20:39:01.000000000","tz":0},"subject":"Use Inspector database to store inspection data","message":"Use Inspector database to store inspection data\n\nIronic inspector supports different backends for storing introspection\ndata. Originally, the database was not an option, so Kayobe provided an\ninspection_store nginx container that poses as a Swift object store.\nIt is now possible to store the data in the Inspector database, and\nthere is even a tool to migrate the data between stores. The database\noption is simpler, and avoids running the nginx container.\n\nThis change switches to use the database store by default. The release\nnote provides information on how to migrate.\n\nStory: 2010205\nTask: 45929\n\nChange-Id: I9c07e87ee43b3b56c3047ba983c42201dc659c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/413c7ba49e421b495c38de677a4689161014adae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/413c7ba49e421b495c38de677a4689161014adae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
