)]}'
{"id":"openstack%2Fkayobe~852101","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I7e10b851016253bf85625e9daa89416c738242e8","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-08-08 09:38:56.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-08 07:59:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I7e10b851016253bf85625e9daa89416c738242e8","subject":"ovn: add network group to neutron-ovn-metadata-agent","status":"MERGED","created":"2022-08-05 15:14:27.000000000","updated":"2022-08-08 09:40:14.000000000","submitted":"2022-08-08 09:38:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852101","cherry_pick_of_change":850608,"cherry_pick_of_patch_set":1,"meta_rev_id":"024591867c95efeaeea66537227dc6bc7d04cd28","_number":852101,"virtual_id_number":852101,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 09:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-08-08 07:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-08 07:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-08 07:59:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-05 15:14:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-08-05 17:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 07:40:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-08-08 07:40:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"d650f15172aac8c4759e5d956ed99eb914dcfbd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 15:14:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9733c03763112a74a42e6a4d87e022d0be0804e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 17:00:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c9cd787623c47648a9d4d91b82e2f63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d701f1ce9964b2db23a4444e0f274d6 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec1728e1c264cbab4cb7b9feb41a0db : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25f35c3fb6ea43459f8d70f8c0787521 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b4d1336abb2495682928bfaf995342d : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe161f102ea14db5afe8a7fb2257c701 : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/338e7c4861f742b78004708733c2eb8b : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73fd628e48a94973af9898288f02ebd3 : SUCCESS in 15m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/46c675778a2242fbbebe656fb4200052 : SUCCESS in 42m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a541c6bfcd044a34b1f3af1ea9b5a125 : SUCCESS in 53m 54s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7833d1848cbd4cebab0a45facfc566ce : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/820e492c1fa4400e83f750b550fbf5bb : SUCCESS in 47m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bfb8e6e209fa447692588596e19dd71a : SUCCESS in 59m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d90096fe12bf44d7b4cc172136694251 : SUCCESS in 20m 12s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/38535ecdecac4ca398ff1135d6d9f084 : SUCCESS in 22m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a1a2563f0c74b85801917b00fb16c40 : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e341a98f5c9d4c6ca74e51f94748b9c4 : SUCCESS in 1h 33m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c504d9aff93448699e5d90ab3310be71 : SUCCESS in 1h 03m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ce7b5fd40bf64744a2057b89294c6960 : SUCCESS in 22m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3647e45665b846618eb509af72010e74 : SUCCESS in 21m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfe6fbcb3fcb48a4be5f33e2ef1c26ac : SUCCESS in 14m 58s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/a4bbf01edff54024aa8be1d64f9b1442 : SUCCESS in 43m 38s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d6515936ea0541a1b74ffa4cb9a66d6c : SUCCESS in 42m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ac4a08d7b7a4f8ab8e71724155238d4 : SUCCESS in 30m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/33fd1edc514e4fdab5fa68a251bba16e : SUCCESS in 16m 03s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a2a0cada0f1543b7b9befa328fdd2623 : SUCCESS in 18m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41f7265b293b4ea8a2b4cc7041847620 : SUCCESS in 11m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d77524d20c2b477b8e1b7a4666ee0ec8 : SUCCESS in 20m 44s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bf0227b8dd5541dfa51ffb6d5b0f8806 : SUCCESS in 19m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dccc83e7895f4def9f7894e4a60dd070 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2d91d1787b97ef541608bcf9d3b519f4dc8ba6db","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-08 07:40:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93b3bee49ab9ac15273061909fee037c1a5b4753","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-08 07:59:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfe8cf4870f41d0693a680d13a0ef587f5a9aebe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 08:00:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7694d711411f29b49ccd2d7664d5debfe270d236","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 09:38:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ab53c567904220997e15bb7f9f6a69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a394d8dc5c149daa0ca6d9261740091 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94bd1b892f7440ef9b76901b253a889e : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3756a203ae1048af9cd0d02e8c421cd7 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a85c28c055a435db1822e4158a87e24 : SUCCESS in 8m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8e7d9ef6d2ec47ae9aa2b34c1b2168dc : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/369c600e41ac428da011eae3d51a852b : SUCCESS in 13m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1410575275d24d77ab574fc8b0e77047 : SUCCESS in 15m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/274b4b2faced4b7691070eef8c6c1ac0 : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e0e183595216432087be3b0a277878d5 : SUCCESS in 1h 01m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cc9f369abde4a9c9b685b11aab22c30 : SUCCESS in 51m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/132d358ab5e14466bc159927a315fa0b : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f1d218949a9144baa3cf889d52209a32 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/ed3199694a004566b191e2a4cbe99def : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f964206c6a1468599743843196d8350 : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5db469a4f4134c03a22b3e576d61b2ee : SUCCESS in 1h 37m 47s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1219a8f537f84cfc8b1e3b0965e2a8ae : SUCCESS in 1h 15m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bed3716bc6b14b0ba9a796dc50d84fae : SUCCESS in 22m 30s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f6ce149365a84bd4a7c39174e9d51ba2 : SUCCESS in 19m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c3e46e3281a4988aaa06fa8982b24f2 : SUCCESS in 17m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d34b874f0d884f2d97835826d8ded85c : SUCCESS in 36m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/491f1e9b10234ca8b531c98f183e48b8 : SUCCESS in 28m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/64cee81a0b6a48ddabc85864e295991f : SUCCESS in 17m 27s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/744b785037224eb4bdacd738ecca9632 : SUCCESS in 16m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a4e573d667a4449b335e0245f9c1109 : SUCCESS in 13m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2722343bae4d4ce899442fbea4bf78da : SUCCESS in 19m 56s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/03fe70302ddb4c30a05184f1383ac9a0 : SUCCESS in 19m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8edee6132174475aa794c290a3ecdc6c : SUCCESS in 15m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2a642cf6fd9e185098549e36be026baf1b550bcf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 09:38:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"024591867c95efeaeea66537227dc6bc7d04cd28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 09:40:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd091ab1585d49889713591211202002\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4695408598149c8aa7dbbf2c8ec88d6 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1a8bbf06372d65aff8afdbb2c73ea2e6af6cbf4","revisions":{"a1a8bbf06372d65aff8afdbb2c73ea2e6af6cbf4":{"kind":"REWORK","_number":1,"created":"2022-08-05 15:14:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/852101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/852101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/852101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/852101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/852101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/852101/1"}}},"commit":{"parents":[{"commit":"18b0d0ceed15ad4f74b2c074ce45190e629bc621","subject":"Add missing declaration of kolla_enable_hacluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18b0d0ceed15ad4f74b2c074ce45190e629bc621"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-21 06:18:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-08-05 15:14:27.000000000","tz":0},"subject":"ovn: add network group to neutron-ovn-metadata-agent","message":"ovn: add network group to neutron-ovn-metadata-agent\n\nFollow up after I111df79f08eb76a735f33c6edf03b5a302fb6cc3 in kolla-ansible\n\nChange-Id: I7e10b851016253bf85625e9daa89416c738242e8\n(cherry picked from commit 5ee99c59481e4fb31119a418248fb4823d6a63ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1a8bbf06372d65aff8afdbb2c73ea2e6af6cbf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1a8bbf06372d65aff8afdbb2c73ea2e6af6cbf4"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
