)]}'
{"id":"openstack%2Fkayobe~852102","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I7e10b851016253bf85625e9daa89416c738242e8","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-08-08 10:43:16.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-08 07:59:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I7e10b851016253bf85625e9daa89416c738242e8","subject":"ovn: add network group to neutron-ovn-metadata-agent","status":"MERGED","created":"2022-08-05 15:14:37.000000000","updated":"2022-08-08 10:44:23.000000000","submitted":"2022-08-08 10:43:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852102","cherry_pick_of_change":850608,"cherry_pick_of_patch_set":1,"meta_rev_id":"21bcf8dbd0cd0198d3ee5fefd1a0c62ea6e934dc","_number":852102,"virtual_id_number":852102,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 10:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-08-08 07:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-08 07:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-08 07:59:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-05 15:14:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-08-05 16:52:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 07:40:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-08-08 07:40:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"82022743562f8d63fedc5c35c9a7a3711b998a16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 15:14:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7681401a598460deabb5861410bec98cc6ff2788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 16:52:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ddf1ee6bf884400a5967b0ab1be9470\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6103a1d4f1b4d1b9a2d4d3b72fbf78d : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d0c8aa2d2954e7087af08cd680e1329 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ece340a41ce84bb9aca6793ac70aafe5 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6a645508de94b4eb6d7ac7d035873c7 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2d49cd11ab044fbb943d6d8db2249f8 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7617bfe43e4f79b1ad832f6c40158b : SUCCESS in 8m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e3c2a475aaf4517a09c5f8c3f3b14e8 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c4be51fe27ad40c4b116540570d63b9f : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/771305fa2f714bcbaeec6ed82376be55 : SUCCESS in 16m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6c223dbeb4c841e7b982018bc8874a40 : SUCCESS in 1h 01m 44s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f7b799799cfe42be8b6a0aec3c68ccde : SUCCESS in 1h 11m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a885876eae7f4c7b9d7ed4cebf28c005 : SUCCESS in 50m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3d2b427dcb154aa18077708861b89331 : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7ab6a3e7171b414d99cf68c1ff0a048a : SUCCESS in 29m 51s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b9ff77c9758845d3b9004aff88b63e2a : SUCCESS in 25m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e506465fd354274b061040792cc9901 : SUCCESS in 17m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d47161bf4b62429fbf9fa0ffaa97a9ac : SUCCESS in 1h 28m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c360b4e220ad4a90a702abc08db5fbef : SUCCESS in 1h 01m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/92784e1de25248c786f4ce2a66dc3b41 : SUCCESS in 21m 30s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/02fe31a989da45c1890c127d9808cc22 : SUCCESS in 21m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45b978549f58467480c1ef2fdc838447 : SUCCESS in 14m 13s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/8c0c9660ebf84014addf5b61ae30b54b : POST_FAILURE in 1h 21m 17s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c43eff708bde4132b6038db98c75b6b2 : SUCCESS in 25m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b463a978f3c4d3bb358b4e38138d840 : SUCCESS in 25m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d2f4bf253d484407817f5b4006b60935 : SUCCESS in 15m 44s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2ca9d54e30714ce49c087225587a7c47 : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5da809645f44134b1f033b6d336ae57 : SUCCESS in 15m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4d4872e582a14c1ea642a674d1e403a6 : SUCCESS in 14m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/596ec27093fa4ee483503d8b95a55e57 : SUCCESS in 20m 03s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0afb356bd040413dad16100370fe6c30 : SUCCESS in 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0baa9e20b7c0d230da485333c91338991984a2b1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-08 07:40:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32cb644953f1f8877862cbb5abb3c696a1a9724f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-08 07:59:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"583988efca31b0273982eb15fd12524a18bd2d70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 08:00:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b97944a2eea38184333c81a6baf5a147d39313cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:43:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/720afbe7e86148109606e5108d058e52\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c298fee79a94e15909ccb5383a6d5cf : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/075ca7667f714eb9bb37aa0968374601 : SUCCESS in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/259f324ed5aa4c698d82cb64afcf51bc : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dfe6bc5404c404d8b0ec4bfc9b091eb : SUCCESS in 8m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a53e6fd1c3f04864845891f80c8306fa : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5daf941f3484667b4f1996dd0855af5 : SUCCESS in 16m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/66f22ed28c214c3699b0315579253303 : SUCCESS in 15m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3e8f890290af476a90c6252fd85b1594 : SUCCESS in 1h 15m 53s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/462bd112686e4d69b6d04480dd5dc2c6 : SUCCESS in 1h 10m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff6fcde2959841b5b1bf1ecd1d842907 : SUCCESS in 55m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/843ee9b9c96749a5830d650cbc40e33d : SUCCESS in 57m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/37061450342642bca4fe51642b106bbd : SUCCESS in 25m 55s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a97612ed756d40cbbcdee44d7fe4576f : SUCCESS in 23m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bdbf774da1a4988be7c9ad48137bd45 : SUCCESS in 12m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aff8266b71184947af75630fd3af9131 : SUCCESS in 2h 08m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e214631eded94f8d83ed4604ba66251a : SUCCESS in 1h 08m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9edc1d1388af47ce85431857edf92a57 : SUCCESS in 22m 53s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6f5737d62d164a82a5c1b089d75d692e : SUCCESS in 24m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/569b50809bfd49ce85ea4a8b3345ae6b : SUCCESS in 22m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b954062f8cc34586acde7334271a96ca : SUCCESS in 28m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebc737075c0a4f858b378cf0b71c743c : SUCCESS in 27m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/98566a5bdad24adfa55e15eeadbed824 : SUCCESS in 18m 12s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/907e86f33c4045a79be7a801825b3c3c : SUCCESS in 17m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f71ad22efce48b08ff54b71d2694009 : SUCCESS in 10m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7cf64c53595942469fde334ff5a3bc97 : SUCCESS in 22m 35s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c7c692c8bee24837b3a909dffd6e885c : SUCCESS in 19m 34s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a263129b537847529bcfeb59c5e3418f : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"203f528c41d23271c9d661be9c6e68f27931c2dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:43:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"21bcf8dbd0cd0198d3ee5fefd1a0c62ea6e934dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:44:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76a6b1934c604ae29b345d72ea089c22\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55953265096d442aba28164a7c7340e7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac89fd0500d0c57bf1908d91aed845926e4e2658","revisions":{"ac89fd0500d0c57bf1908d91aed845926e4e2658":{"kind":"REWORK","_number":1,"created":"2022-08-05 15:14:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/852102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/852102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/852102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/852102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/852102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/852102/1"}}},"commit":{"parents":[{"commit":"582f05020b382cab693d4eff58ecfdeeeb104397","subject":"Add missing declaration of kolla_enable_hacluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/582f05020b382cab693d4eff58ecfdeeeb104397"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-21 06:18:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-08-05 15:14:37.000000000","tz":0},"subject":"ovn: add network group to neutron-ovn-metadata-agent","message":"ovn: add network group to neutron-ovn-metadata-agent\n\nFollow up after I111df79f08eb76a735f33c6edf03b5a302fb6cc3 in kolla-ansible\n\nChange-Id: I7e10b851016253bf85625e9daa89416c738242e8\n(cherry picked from commit 5ee99c59481e4fb31119a418248fb4823d6a63ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac89fd0500d0c57bf1908d91aed845926e4e2658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac89fd0500d0c57bf1908d91aed845926e4e2658"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
